)]}'
{"id":"openstack%2Fswift~934154","triplet_id":"openstack%2Fswift~master~Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37","project":"openstack/swift","branch":"master","topic":"part-num-errors","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-03 23:05:28.000000000","reason":"\u003cGERRIT_ACCOUNT_34892\u003e replied on the change","reason_account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-03 23:05:28.000000000","reason":"\u003cGERRIT_ACCOUNT_34892\u003e replied on the change","reason_account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}}},"removed_from_attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-11-22 17:07:57.000000000","reason":"\u003cGERRIT_ACCOUNT_34892\u003e replied on the change","reason_account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}}},"hashtags":[],"change_id":"Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37","subject":"s3api: error handling of complete mpu and part-num","status":"NEW","created":"2024-11-05 17:13:56.000000000","updated":"2024-12-04 00:17:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":10,"unresolved_comment_count":2,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"35ad223f0ea0d5ecf0b2432463e7a3f3c7581444","_number":934154,"virtual_id_number":934154,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-04 00:17:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-12-03 23:49:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-12-03 23:49:17.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"reviewer_updates":[{"updated":"2024-11-05 17:18:34.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-11-05 17:18:34.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-11-05 17:18:34.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-11-05 18:13:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-13 19:03:55.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-11-13 21:24:41.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"27c6ee565e8138316bcd2d2df3d4e62547f8ff52","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-05 17:13:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e3b7cfad6dd799d4ed5655e025856ad561c5295","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-05 17:18:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b897fea3f593e8a2efc82d521b09f763305b7d00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 18:13:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bd270c502d44607b4e7f534c9236f78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c55eb989aa224e2aba0d90f7418ce8e2 : SUCCESS in 7m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/e0b00fdea23a4e23a8d0e1c7cc9686c3 : SUCCESS in 45m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40390ccbd76946f19dbe276143521552 : SUCCESS in 47m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dc956103825b4b94b06d308dc26bacf7 : SUCCESS in 39m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83e0bc63332b4cf2af7f7566a7d76010 : FAILURE in 36m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17c3dafe209e4abd9125b0743b73c9ee : SUCCESS in 12m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56e0ff9b10844d1293910c11b779769f : SUCCESS in 12m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf471f5938ed400a8e1ca21daddf734a : SUCCESS in 20m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4c1cdb06dfef4c2ab7decbfc74c6b2dc : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a3b72c98db32460eaf43dd7b491b4502 : FAILURE in 16m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f620834e8b2c482ea6861cc88163a615 : FAILURE in 20m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6d3bcdf596454fabb54e58e979b253dd : FAILURE in 11m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/26e90f1186d14e2383b37d15f139fa1c : FAILURE in 19m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e20a9e37d3df4422a2c970cd13138ede : SUCCESS in 13m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7481802a45374415a424a919e7f9ae65 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/27c4d4ae2b2e42dbbfb67572a5d38fd8 : SUCCESS in 10m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5ac7bf30b39a493a812ff5960df940d5 : SUCCESS in 11m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/11f005cd682d41a78879b1fce7b7cdd6 : SUCCESS in 12m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4c6699a27faa40c3bf283a41386ed8d9 : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/435143e2f5e44e0fbdc1e2c561d42b83 : SUCCESS in 3m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/99e96211085a492b894746feb2bf231a : SUCCESS in 15m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/45f706629e81459e8f5451d3126fd2e5 : FAILURE in 4m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a19740924478443ca3a8eb87a3c7fe73 : SUCCESS in 48m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff696c3b9ac4478d9cfc186b05e44e2a : SUCCESS in 32m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/42a442cb3e534551b3dbe4f5077d16b0 : SUCCESS in 35m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d10beb1a155544308063b9871124ce23 : FAILURE in 14m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05bd3c9624304660812c2e5ee6dd11a2 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/955e50d518a84a2992f386cc32b4fe90 : SUCCESS in 15m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82e1ca1ec71f40c59705132f5c3fa599 : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":2},{"id":"c7568ab25f44a731e6b3f564feb628b28d6c2ec0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 19:17:46.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/748e58291f3147feace910be5beae526\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee5f3ec1d94944c3a1e74c0403c1910c : FAILURE in 1h 57m 18s","accounts_in_message":[],"_revision_number":2},{"id":"8e75589ad1b79c4147d964e1433d86a26dbbc88e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-06 18:48:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f16265200d100a4a2006dd5e5a71b01eba164b73","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-06 19:06:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8c3e9a3a447f2ddbd534cdb530221b72385f3f0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 20:04:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cd2234fe430491ea2ef4760afceb3c4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b5e2c8196c2e44219319b608c0db2446 : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":4},{"id":"5608f4ece31de962e7e2a9dd478e58881c8ac405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 21:09:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e295246b7b24380b0b55d3f0b00535b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eafa87180b8540a9a77112ee7b4657b6 : SUCCESS in 7m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/d0c6d71b565c450d9c5f7c6774067c9d : SUCCESS in 1h 02m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a5e5fda0cfed43f29e6ef4bfbf50752e : SUCCESS in 40m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c71797425322492385da128c5aca6595 : SUCCESS in 52m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3eac294561f84b15b24b93c782b51135 : SUCCESS in 45m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92200cabda534aa381b748a88d9f307c : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/894de7b301314ed2be80e06143b3dca7 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da9cf5ff665947ddb70353406f8c432a : SUCCESS in 20m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e46d6212b5d244dca63c834d8c0079f5 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ccce7f20d35454599c5a4233be78a38 : SUCCESS in 9m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5c4041d5c0f4993bc009d0a1fbf0642 : SUCCESS in 18m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ac8a867770754ec7bbd612586c76b82b : SUCCESS in 22m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/45c50ae1e52843fca082cd19ae9f0958 : SUCCESS in 30m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/300bb60983a74345801f18a397cd4505 : SUCCESS in 9m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3eb25822cf8f410388de3fa6d05ed4ff : SUCCESS in 19m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/694f755795b54d4594aea60aa780738b : SUCCESS in 19m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/860beb6ac2424f9c94a131920073bf2d : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1f3608f0209c41e0b4509e7a238a5df9 : SUCCESS in 13m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f996a1d5666c40f8bb04391c9e47eba6 : SUCCESS in 16m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9f0fedcaf8343eda254621444e5709f : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eed5e367ef414ae9b728d14b7101b98f : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c6c4439f69a64e46b1c46df32fa5ea7d : SUCCESS in 6m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2e3a6df824c4d1a8dbc57ba6825d36a : TIMED_OUT in 2h 01m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f85bfad777da4e228e088d6815cfffc0 : SUCCESS in 51m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/57d98fb4636048e5b9d6e8fa2eeeaf38 : SUCCESS in 58m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a8a44f73dfd487c9fe3679139ceddc9 : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21dad3881a846a2a5b7a4107fef5d77 : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d2ae637ee1244f359a37c876057ba760 : SUCCESS in 14m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a44d0460d9ea4aa79ea4a05c681ebd70 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":4},{"id":"c6a63349dc36b1cd08621fb4a7193962cf8d5073","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-06 21:28:14.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"6d6b49a8406fba862913b563b4cef2f2e188c98a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-06 21:43:33.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f1114f6c75649453e3610dc4499b96a34326c96d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 22:43:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbc92f8da57a4a24b429ff2790d7a162\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878cb9f47e2349d1a530c2015a62102a : SUCCESS in 5m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/aac837fafd334df5bf7dce6ca81b0cc4 : SUCCESS in 47m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c7e89586083e4937a2aaab66f53aa217 : SUCCESS in 37m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e02d3e5fe6742aaaf320715408155df : SUCCESS in 43m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3300a66c95f7484d9ddbd0a147d85e57 : SUCCESS in 42m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/115a0c541d8a40ecae42319a20ac8d99 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8490e02df9c409a8ffe2a7c4d44681e : SUCCESS in 13m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20566dcdc652465d9fb56ae85db1638b : SUCCESS in 20m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a704edded5246ebbde2ed3f811ad9a6 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/26231aadc67644b485d173e872275d63 : SUCCESS in 19m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0547c03811f34943bfecffd29511f751 : SUCCESS in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1008cdbb8d3540a39d773a9d93b6bfe1 : SUCCESS in 18m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e02f0300919747e4a85f04fec69956ed : SUCCESS in 20m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5d7231c0a4d54b4287c80f152624ac65 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/81e1bea6e6ae4f40a32c9f6e1988e39a : SUCCESS in 13m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/37968844e18c4df09df0c827fdaaf077 : SUCCESS in 18m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5efc65ceb84a4efa8ff2b3d33d39634c : SUCCESS in 11m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c97405144a844e02b77556ff3aaa89a9 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0d49c5d3d2e743898e220a0bad72bf93 : SUCCESS in 9m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3315daabee3a4b95b83db4e7623e6126 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/81df585a2fc84df4a2b66520de847dd0 : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/427fb93051504394bac39c214babe3e8 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a05603993e748a2b905469a2928b87e : SUCCESS in 54m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52157527d8c345258b9a4274007c56a8 : SUCCESS in 35m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/419dff6bfabd4331a6e42c64bf3e5389 : SUCCESS in 53m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fa48da291a044089ddfa50c32951bc5 : SUCCESS in 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea6f077d3b1846e2aa40e9103ed21876 : SUCCESS in 3m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1461baab82454d36b0c9d9681fc38fb8 : SUCCESS in 20m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd5757864a624385ac954a92417a980b : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":5},{"id":"3fdd8a5ea915dd0c3450b53e6460bf4facd2278d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 23:10:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa35cdc4be7848e496c58f2e796ee99a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0044df0d493d4e01830a516bb7880fd9 : SUCCESS in 1h 23m 36s","accounts_in_message":[],"_revision_number":5},{"id":"4b77f6b8efb343870a11d12069e9d13b3f198da1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-06 23:15:55.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fc0a6f9f423244214d47e3fe6f66f2c029aed0e6","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-06 23:16:30.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4b9ada14f3d719b311e6aa0262dd313f34bc0cd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-06 23:19:06.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"2a5215062c157fd09e996849e514fe373ebd8935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 00:19:50.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54bd01345dc44859a011e2e774cf027e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e97fce726fc46328bd6f630e76d3761 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/c48b8f1be0874c39bc5718917e9d9649 : SUCCESS in 49m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a53b03e341d4b86b6c53eaa3c896f62 : SUCCESS in 46m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aa4fd02fbb05449a9b77f211a08d8918 : SUCCESS in 45m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7959cf1ec4742039f8c84e6e8341438 : SUCCESS in 53m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8eefb5d7ef6a4e4fbc02d949d7115236 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84c15464988f432a9a25ceafeb7ac8de : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20519af06833426ea8b302b23709e34a : SUCCESS in 12m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1607d74de07d4add8cc5ea0cc1e88412 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/46f048ac57214234883d1f34cbc725e3 : SUCCESS in 10m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/777be668514b4fba915f9ed5fc08b634 : SUCCESS in 9m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb72e4d698ac4764a54fb4657e37a5b4 : SUCCESS in 10m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b8a953ee18d54118ac8248c5ea5de9e3 : SUCCESS in 14m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2a9a2929eb9b4b4cae1d9f02843f0502 : SUCCESS in 10m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac3e2c8d5acb4a7d81fd487900824af4 : SUCCESS in 11m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7ce728f628e84fa69abf6f1829e8542d : SUCCESS in 17m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/34fb6d42b8bd46b6966fc9427efe36e4 : SUCCESS in 13m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/99165b0b02c04cee9b88a57ff9980ec5 : SUCCESS in 15m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4e8bfb562f7e4fb88309c86d3ccdd5fd : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/37dda66827ef43acb4b3fbdb404255d6 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fb3379ba4d804c6cbad23cf9bac29087 : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ba4c2fec50d144da932b7060ecf0a095 : SUCCESS in 5m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/210c070312fe4f8e940c8d352a985e5b : SUCCESS in 50m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/65bea124d2844240a16f0b75e657f84a : SUCCESS in 39m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c00ee377035348f6bb7100a9497166df : SUCCESS in 48m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb3c0b9ac3d44e0f803b53233bb993cf : SUCCESS in 20m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6101bea33b0a4359be4a4e38bafd3849 : FAILURE in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11ede868d1dc4be68800a9788acd2aea : SUCCESS in 17m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7e583cf726f4b5e8ded90fa11770deb : SUCCESS in 46m 16s","accounts_in_message":[],"_revision_number":7},{"id":"d051d1f7733f7eed7b235716063bf6cc4c498434","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 00:41:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd03f2259ce14ec3baf789b51c0aec07\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/391411a444d24b80b69b7588b080371d : SUCCESS in 1h 15m 17s","accounts_in_message":[],"_revision_number":7},{"id":"a99de838ea71b2eb010328ef844be3c0cb14fbad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-08 01:05:32.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"253c4ae8e64879c5c09e9ce03547ff2c7dc8147e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 02:15:02.000000000","message":"Patch Set 8: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9717c6ffbfc048c9a6a73a69b5efc586\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee0813e1a975464884b93c8235ff1cfb : SUCCESS in 7m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/b01d874f4067478fa9d094a2d577ae0b : SUCCESS in 50m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b8685dbc26a045f7beb8c23f0b9d8d24 : SUCCESS in 48m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/006e8b668bb341fc93814020a6bdd49c : SUCCESS in 37m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e632b0c7f29437ab724ffd3128436f2 : FAILURE in 37m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0902de2baa0b4129b6ca4d9a3817f1ea : FAILURE in 20m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf7087d0705e44b6af2eb0d3bbc48638 : FAILURE in 21m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5be902355ae94d81b96da79d3225a219 : FAILURE in 12m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c7518f475e040d9b34083f5028b117c : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4fa812c4d3a64c09ad1d604e56757f2d : SUCCESS in 21m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dab47be2f8a9478fa475658f869de9d4 : SUCCESS in 11m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b0eeb2be885406ab604b291ed8c8264 : SUCCESS in 20m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/134c3e9ba0d04220af26a384e26a1dfa : SUCCESS in 30m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dd783ff192574996935b45ad82d02b66 : FAILURE in 10m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/da6902d9fdee4c268ba38aebd9769863 : FAILURE in 17m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/30734491011a444e822a4cd014c1e3f7 : FAILURE in 10m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b5b2d25867164ba8a83b970273724a8c : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f7516840bffd4694b864c1f6fe6477ad : SUCCESS in 15m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/01a864cdeea2422b82eb40e62e513bc8 : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86f84e99f0ce49df8b897e234db97d1c : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e8d9cc43be05404da8957e77cbe8c9ea : FAILURE in 9m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8d412a3856594b6593672c3342beecca : SUCCESS in 7m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa61b18547294505b648cfe17f386e39 : SUCCESS in 53m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25431a97480e431884a4acbd4610a81f : FAILURE in 25m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7f4e6df9e1e147808fb0feab0a018fd6 : FAILURE in 27m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fbb9b57047b4d759258ffa326bf191f : SUCCESS in 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b22d910e58714135b93ecce219b9a265 : FAILURE in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6622a6c1ced94ad0a10284da374c84bb : SUCCESS in 16m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd410aee53794e7c8f274f4f1d88e9a3 : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":8},{"id":"6428d3927074c1d08242f3ed1f630b5bd02ab34d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 02:28:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d2c4435b3640fd96c64b15bd46ec0b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a4f132fe19ee4464a7be05f253eea947 : SUCCESS in 1h 20m 25s","accounts_in_message":[],"_revision_number":8},{"id":"7af855ca229d80f9ec7844bdcc5405be4a263398","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-08 17:15:59.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"6b8e89e5b5d204184ee198f3398df5e915b290d9","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-08 17:16:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"14d39572256605ff90fd26a0af642bc51503f613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 18:28:51.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1656747ef054b2b9cf3c59a2917689f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277b71fa8c5943938a7a88fe9f8bc815 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/c0ef7e1ba11d419aa85ffe70133c2626 : SUCCESS in 1h 00m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a23b058e0344d20b7a429857a972a1c : SUCCESS in 56m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e1b5e0e1be1149a6889afb751aa29b9c : SUCCESS in 51m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b38e715d5288499292f01519c02c68ab : SUCCESS in 1h 06m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5dc33b298cbd4c94a9031a9dcd81ef3b : FAILURE in 19m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be3cc864d2a74356842020e56e27e4d5 : FAILURE in 22m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb882fb2243f434d919856e76d8baaef : FAILURE in 18m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7e3af02013fe46c4b6c37254451bcfbb : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0d34b4a251c04d93b1cb09958904f2ee : SUCCESS in 19m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/524fdf8e1eed47bfb43be70abe7f23d5 : SUCCESS in 15m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/da0724c77cdc4a83a2d393303ee74d4a : SUCCESS in 17m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d3efd7bfadf740b9b4ca703da8a432ca : SUCCESS in 26m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1234fc89077f496a8356425109b7cc16 : FAILURE in 16m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e70d9a344867445cb9896b2c390597c7 : FAILURE in 14m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/249d639a9d5944a596b0b0e396433c40 : FAILURE in 15m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/92312ccd46874988aa80ab960f0c1bf4 : SUCCESS in 18m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7f8545b3c0424fb38619c333ab3d5885 : SUCCESS in 18m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b6fcf3af58b44e98bd2692f399600217 : SUCCESS in 19m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7960df86fc064a4fb51d031a599ff52d : SUCCESS in 5m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/983e4ac52d224867912dcbb09dd8d53f : FAILURE in 17m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6eb66d48a5604cb2a9b77b0e1b5e53d1 : SUCCESS in 5m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3abed3d07466425e9a9e79fc252fcda5 : SUCCESS in 40m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58e78ee577204bd8932ad7be2239713c : FAILURE in 36m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34c6db4d3f8f40b18a43e612a48c4d92 : FAILURE in 36m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/056c9eae96244e2494d61fef9c2d4c8e : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98a27925edaf42a781cdbbf2dae83756 : FAILURE in 3m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f7055b944ff4ed0b71b05ae40daa97d : SUCCESS in 20m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8d033885ded415abcd291b5fd045581 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":9},{"id":"d6b162a80abf68848810ff21f3322506474c2d55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 21:46:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bffc88e0d094b1e876394ce6aecc234\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce0c8abb699542ecacccbf1e8c32b738 : SUCCESS in 2h 05m 05s","accounts_in_message":[],"_revision_number":9},{"id":"21d6a3a1c2c4eaf69742433905b0eb2b811c9bfa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-13 19:03:55.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"25b73d9ebfedf34d1d5821c975f822af20c3536e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-13 21:24:41.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"75b5151e97ce5c0a215acacd318efb636967affd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-21 20:14:20.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"371dfe01d6f889eb1824c914331da4ee11bbd79e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 21:27:26.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/523da6425d7244078aa65156dc7700ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a55d539f3e456b815607aab22884cd : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/b6b46e4206204a39a6fbe12b7d1ce539 : SUCCESS in 1h 03m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb31cc07fa784162910495988365abd7 : SUCCESS in 40m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0948e9a63a1547a689d3a94d5528404b : SUCCESS in 57m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd9b0418916f4a8f9b9c352574e59475 : SUCCESS in 54m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73f6a430b7044b5fa6617421efe2f439 : FAILURE in 18m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7fb558d340943fcbe2e8b3465337271 : FAILURE in 17m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9b2370239ee44bd8c9640d7323e7dca : FAILURE in 19m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8e33e8a6a434265887ea92874cf05ab : FAILURE in 6m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/244eeaa001e941549ff9bee01ce2707a : SUCCESS in 14m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/232285ed30624807a6cdf25fac3f2c21 : SUCCESS in 17m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/984bc2a63328427f8e9bd238511925cf : SUCCESS in 19m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1d5215215c504ba59bc252f26102e81c : SUCCESS in 21m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9547048e06bc40bc904939fb827cfb9c : FAILURE in 15m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/301274db6ad24753a6e287ca1c517215 : FAILURE in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b990410c1b89485baeba213e39b54747 : FAILURE in 14m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a1aa73a711ec4ccf83f4c59787b18489 : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/35ab68c0087143eca468716ed2c1b6be : SUCCESS in 13m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/55ac1350b4314cafa41be7e0d729b8aa : SUCCESS in 11m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b330bed9561c4738a016fbefafd3ea86 : FAILURE in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c3df5578fbdd4c62a2e8ba1c97f4ef5b : FAILURE in 11m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/906c5dd2fd224ac98a3a5abe5ddfd8b5 : SUCCESS in 5m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c16c66e33bd4328b44406e85f3e2982 : SUCCESS in 50m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6d3c8002111e4092b8c8bb1f4030ad4e : FAILURE in 23m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/01750592be97457eb16b259eb8fed828 : FAILURE in 34m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc35db93c6954f0da0736c75b53ef93a : SUCCESS in 20m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4674aedb96b4afc95df9f8ebf4a01ff : FAILURE in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3a79757bd87c4d9290d02ff3d4f0a44b : SUCCESS in 15m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c648439d7e14134b042d07249e49d2c : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":10},{"id":"1b585158aa3f3eb7aede8988c6b57df5c0a05df4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:06:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee750b24514d47fcade42d7bbacbd76f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/222c5efa2e8548fa998b2e226d5c89ff : SUCCESS in 2h 07m 36s","accounts_in_message":[],"_revision_number":10},{"id":"6a2219a4ac6f2e0b7518c3abe763594b95557ce3","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-22 17:07:57.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"1a9dadd82249c9b53b7e8d0ab754a9afbc50ce83","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-03 23:05:04.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":10},{"id":"72f96487918ab2ea95d830a648ede9b78bb9ea16","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-03 23:05:28.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"02e9d34cac90cc60fd2685c26e1c14f8f4aa5343","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-03 23:05:55.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nThe following files contain Git conflicts:\n* test/s3api/test_mpu.py\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"7b921969348f23500e7e3f4fc8d6ec09199d5414","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-03 23:08:21.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ac424a61f2351c2f01d287884e0653cf61efb5ce","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-03 23:49:17.000000000","message":"Patch Set 12: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"509386e4eb7250507862b08358d7c8af65aee790","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 00:03:39.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb7e6c84b08e4157b4f306a34e43a4ac\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b26a390a7f5438fa91748035e4f6baa : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":12},{"id":"35ad223f0ea0d5ecf0b2432463e7a3f3c7581444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 00:17:36.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/479e89fb82394d3bb3a8f7472d8ecf70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73182a9445145b7afadc118ac355987 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/e9af7b89831f43d48f47ac8dae67ab19 : SUCCESS in 46m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f562af567067476dbdeb0def228d5693 : SUCCESS in 54m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ddc23333e72540a981bc030a0113168c : SUCCESS in 44m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e356f9b335c4d0c866ae3d16b8fba35 : SUCCESS in 1h 03m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/90d6dfac031645c9bab1627e52b89986 : FAILURE in 11m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7619f87144b487fb47ad103c7c17350 : FAILURE in 18m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37b348089a5e4a1d917046dff98184fa : FAILURE in 26m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ff6a5a82298548fc833d3bd69fdbcf1a : SUCCESS in 4m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9cb9e64c8dfb4b8184ca1682cccdd26f : SUCCESS in 23m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e8a6ddf4544c4cbab6b51676b6da4493 : SUCCESS in 13m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bdbd6a3f7d4943e185bc72be2b81c172 : SUCCESS in 9m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4fab5d007666480484643a2586b3d641 : SUCCESS in 29m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4cf4c1996b32452dac92e982258e2699 : FAILURE in 19m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7c0a3edfadf3487aa92bbc367ab26b10 : FAILURE in 9m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/79f880b4f29b4e70a7dc31ae5768e375 : FAILURE in 13m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bd5dc667a0ec4938bec9e36dd3d935ec : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1b965bdcde244c82ac96b91ddc40d607 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/15c40c3e97fc480582d7e473b09c3043 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5773c323011144f692a2e2b206e5f24d : SUCCESS in 7m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b9f1d43b6910459fa5d5ad20aa11e3bb : FAILURE in 16m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a803e624097d40099df7c554bfac8ac3 : FAILURE in 8m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/572d7e832ab54f01a926db1dcb2cdd98 : SUCCESS in 54m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13c40547e0d34d1e84f94a2479ee68c2 : SUCCESS in 45m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/451ea7bd50d540bdb6a64f5844ab8ffa : SUCCESS in 51m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad091cdd2e8b4c968654bc2b7e8d721d : SUCCESS in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b25e834f88a4321b79e281e0e5f5cb8 : FAILURE in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8cf51bec4c6043e98ff7eb6cd33a18fb : SUCCESS in 17m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1eaab62417a9446cba8c564b80850197 : SUCCESS in 54m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0e1fa2e3072451e8b5601056dcfc582 : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"9eed80bb17ec8650c1b998410ad97749edf28c31","revisions":{"7c4b6e65d7eae9216e5e635f2d18d7eeaa94c353":{"kind":"REWORK","_number":1,"created":"2024-11-05 17:13:56.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/1"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 17:13:50.000000000","tz":-480},"subject":"s3api: fix error handling for part-num w/t range","message":"s3api: fix error handling for part-num w/t range\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c4b6e65d7eae9216e5e635f2d18d7eeaa94c353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c4b6e65d7eae9216e5e635f2d18d7eeaa94c353"}]},"branch":"refs/heads/master"},"444000f57b2eb9cf7161178e38216cb831b0e8f9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-05 17:18:18.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/2"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-11-05 17:18:18.000000000","tz":0},"subject":"s3api: fix error handling for part-num w/t range","message":"s3api: fix error handling for part-num w/t range\n\nDrive-By: Raise 400 on SLOs as well\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/444000f57b2eb9cf7161178e38216cb831b0e8f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/444000f57b2eb9cf7161178e38216cb831b0e8f9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2e1efcb90dc499aa24b6e47f76b255ad32ed3b67":{"kind":"REWORK","_number":3,"created":"2024-11-06 18:48:12.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/3"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-06 18:48:00.000000000","tz":-480},"subject":"s3api: improve error handling for part-num req\u0027s","message":"s3api: improve error handling for part-num req\u0027s\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e1efcb90dc499aa24b6e47f76b255ad32ed3b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e1efcb90dc499aa24b6e47f76b255ad32ed3b67"}]},"branch":"refs/heads/master"},"8b97de4dceae41f288d35e4da48c4434bd9ad6af":{"kind":"REWORK","_number":4,"created":"2024-11-06 19:06:03.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/4"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-06 19:05:55.000000000","tz":-480},"subject":"s3api: improve error handling for part-num req\u0027s","message":"s3api: improve error handling for part-num req\u0027s\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b97de4dceae41f288d35e4da48c4434bd9ad6af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b97de4dceae41f288d35e4da48c4434bd9ad6af"}]},"branch":"refs/heads/master"},"e3fa9cbf014f881e82709e060277808c96d157c2":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-11-06 21:43:33.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/5"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-11-06 21:43:33.000000000","tz":0},"subject":"s3api: improve error handling for part-num requests","message":"s3api: improve error handling for part-num requests\n\nAdd BadRequest response in the slo layer, add relevant\nerrors to s3response.py. Improve part order validation\nwhen completing multi-part uploads and enhance test\ncoverage in both unit-tests and cross-compat tests.\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3fa9cbf014f881e82709e060277808c96d157c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3fa9cbf014f881e82709e060277808c96d157c2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c69aef72ba1243240e5978b51daa0b98a45bc326":{"kind":"REWORK","_number":6,"created":"2024-11-06 23:15:55.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/6"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-06 23:10:44.000000000","tz":-480},"subject":"s3api: improve error handling for part-num requests","message":"s3api: improve error handling for part-num requests\n\nAdd BadRequest response in the slo layer, add relevant\nerrors to s3response.py. Improve part order validation\nwhen completing multi-part uploads and enhance test\ncoverage in both unit-tests and cross-compat tests.\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c69aef72ba1243240e5978b51daa0b98a45bc326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c69aef72ba1243240e5978b51daa0b98a45bc326"}]},"branch":"refs/heads/master"},"24046e5e74953ad32b9a5c94a41d7428b3233e04":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-11-06 23:19:06.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/7"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-11-06 23:19:06.000000000","tz":0},"subject":"s3api: error handling of complete mpu and part-num","message":"s3api: error handling of complete mpu and part-num\n\nAdd BadRequest response in the slo layer, add relevant\nerrors to s3response.py. Improve part order validation\nwhen completing multi-part uploads and enhance test\ncoverage in both unit-tests and cross-compat tests.\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24046e5e74953ad32b9a5c94a41d7428b3233e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24046e5e74953ad32b9a5c94a41d7428b3233e04"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d2fe77f102f68e722003d6c9f3a073e1a6f9ed46":{"kind":"REWORK","_number":8,"created":"2024-11-08 01:05:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/8"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-08 01:05:24.000000000","tz":-480},"subject":"s3api: error handling of complete mpu and part-num","message":"s3api: error handling of complete mpu and part-num\n\nAdd BadRequest response in the slo layer, add relevant\nerrors to s3response.py. Improve part order validation\nwhen completing multi-part uploads and enhance test\ncoverage in both unit-tests and cross-compat tests.\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2fe77f102f68e722003d6c9f3a073e1a6f9ed46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2fe77f102f68e722003d6c9f3a073e1a6f9ed46"}]},"branch":"refs/heads/master"},"38a9926c5ba68ff8e5d202ce5d177fbeb6a5da52":{"kind":"REWORK","_number":9,"created":"2024-11-08 17:15:59.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/9"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-08 17:15:43.000000000","tz":-480},"subject":"s3api: error handling of complete mpu and part-num","message":"s3api: error handling of complete mpu and part-num\n\nAdd BadRequest response in the slo layer, add relevant\nerrors to s3response.py. Improve part order validation\nwhen completing multi-part uploads and enhance test\ncoverage in both unit-tests and cross-compat tests.\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a9926c5ba68ff8e5d202ce5d177fbeb6a5da52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a9926c5ba68ff8e5d202ce5d177fbeb6a5da52"}]},"branch":"refs/heads/master"},"398236166a1efdf0343fa1a037835e4af23b4834":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-11-21 20:14:20.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/10"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-11-21 20:14:20.000000000","tz":0},"subject":"s3api: error handling of complete mpu and part-num","message":"s3api: error handling of complete mpu and part-num\n\nAdd BadRequest response in the slo layer, add relevant\nerrors to s3response.py. Improve part order validation\nwhen completing multi-part uploads and enhance test\ncoverage in both unit-tests and cross-compat tests.\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/398236166a1efdf0343fa1a037835e4af23b4834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/398236166a1efdf0343fa1a037835e4af23b4834"}]},"branch":"refs/heads/master","description":"Rebase"},"2bda133fda4bec95b2ddbf69e22b4da4c4e141eb":{"kind":"REWORK","_number":11,"created":"2024-12-03 23:05:55.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/11"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-12-03 23:05:55.000000000","tz":0},"subject":"s3api: error handling of complete mpu and part-num","message":"s3api: error handling of complete mpu and part-num\n\nAdd BadRequest response in the slo layer, add relevant\nerrors to s3response.py. Improve part order validation\nwhen completing multi-part uploads and enhance test\ncoverage in both unit-tests and cross-compat tests.\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bda133fda4bec95b2ddbf69e22b4da4c4e141eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bda133fda4bec95b2ddbf69e22b4da4c4e141eb"}]},"branch":"refs/heads/master","description":"Rebase"},"9eed80bb17ec8650c1b998410ad97749edf28c31":{"kind":"REWORK","_number":12,"created":"2024-12-03 23:08:21.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/54/934154/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/934154/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/934154/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/934154/12"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-05 13:44:49.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-03 23:08:09.000000000","tz":-480},"subject":"s3api: error handling of complete mpu and part-num","message":"s3api: error handling of complete mpu and part-num\n\nAdd BadRequest response in the slo layer, add relevant\nerrors to s3response.py. Improve part order validation\nwhen completing multi-part uploads and enhance test\ncoverage in both unit-tests and cross-compat tests.\n\nChange-Id: Ief41a8e02743f79be2e1e2a57e2a4f5ca140fa37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eed80bb17ec8650c1b998410ad97749edf28c31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eed80bb17ec8650c1b998410ad97749edf28c31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
