)]}'
{"id":"openstack%2Fswift~934170","triplet_id":"openstack%2Fswift~master~Iebd97b7eea1424058d9d77ee7d78f4adcc0561bf","project":"openstack/swift","branch":"master","topic":"removeLoggerAdapters","hashtags":[],"change_id":"Iebd97b7eea1424058d9d77ee7d78f4adcc0561bf","subject":"fix \u0027n sq? test ideas","status":"NEW","created":"2024-11-05 18:21:59.000000000","updated":"2024-11-05 20:49:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f9702768cae3d69f44abd0ee328b18909225e53","_number":934170,"virtual_id_number":934170,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-05 19:35:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-05 19:35:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfd3f45e59fb3c38e1dbe9eb13f4ac35d6189e16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-05 18:21:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e8457d2722b4d68047d6c2d6476b5df145ff70f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 19:35:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5063d9ebb2df431e8ffabbabdfe7ad62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9350bb47be4b4d2ba92451655ec2b7fb : SUCCESS in 8m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e1bfbf5e509e429fb2bef0f050ac6890 : SUCCESS in 1h 05m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef74fad2a102425ead077e1dfd805b34 : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d23c4fb13ae45cbac055d8851426aa6 : SUCCESS in 26m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87a6c32be3d94034ae7dda53a695dce3 : SUCCESS in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e534c0534129467ea30470d85d824e1f : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ba7f5f749bfc4a788bbf15488431adcc : SUCCESS in 10m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2ccfd573eac34f1a82b82cb5a5835f25 : SUCCESS in 18m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a9770a16f918490e8adae6d19a89117e : SUCCESS in 18m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/af90b3d0ec564c8faa71282994bc9132 : SUCCESS in 23m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/81df64e863cb4c0dbf3266b205e6e3f9 : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f16694e6c8f744faacb6229f8e26e0b0 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7b003d5fe0914770b271c804d3823b9c : SUCCESS in 11m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/208f007bbf93444f9b21883f4339e9a1 : SUCCESS in 9m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a84b1dd56a234f4986e8fb3b64047e69 : SUCCESS in 9m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1dd5ff44cbc9481388aab33c52bd6e1b : SUCCESS in 9m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50564db295874da9863053bbb3e3cd53 : SUCCESS in 29m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a010426d7e74420ba81ccbb87e6ca20 : SUCCESS in 49m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8efe9b4d0bf4a2fbdad20e1fc254d46 : SUCCESS in 19m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b22e5c515ba44705b77f640f3aa7c2bc : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8fa16e86fced4738a96dae7883cfbd53 : SUCCESS in 17m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7f9702768cae3d69f44abd0ee328b18909225e53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 20:49:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19a4b60a383f48a291d905df3182308d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5bf46c053ed44bcba2dafcf32bc0f159 : SUCCESS in 1h 42m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b7ab30ced4b1f19680e9c3b410c230f6d5211eb","revisions":{"8b7ab30ced4b1f19680e9c3b410c230f6d5211eb":{"kind":"REWORK","_number":1,"created":"2024-11-05 18:21:59.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/70/934170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/934170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/934170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/934170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/934170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/934170/1"}}},"commit":{"parents":[{"commit":"ec8b868bb1368cfaaaaa8d0bf8699161e26738eb","subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8b868bb1368cfaaaaa8d0bf8699161e26738eb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-05 18:21:40.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-05 18:21:40.000000000","tz":-360},"subject":"fix \u0027n sq? test ideas","message":"fix \u0027n sq? test ideas\n\nChange-Id: Iebd97b7eea1424058d9d77ee7d78f4adcc0561bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b7ab30ced4b1f19680e9c3b410c230f6d5211eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b7ab30ced4b1f19680e9c3b410c230f6d5211eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
