)]}'
{"id":"openstack%2Fswift~934200","triplet_id":"openstack%2Fswift~master~I1c0aba9830433f093d024b4c39cd3a3b2f0d69f1","project":"openstack/swift","branch":"master","topic":"object_write_yield","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-11-29 16:32:45.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-29 13:59:48.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I1c0aba9830433f093d024b4c39cd3a3b2f0d69f1","subject":"Object-server: add periodic greenthread yielding during file write","status":"MERGED","created":"2024-11-06 06:49:56.000000000","updated":"2024-11-29 16:42:10.000000000","submitted":"2024-11-29 16:32:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"934200-object_write_yield","meta_rev_id":"c67e43a6c90be024e13c3d1fb8ff8a351dd44c8d","_number":934200,"virtual_id_number":934200,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-29 16:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-29 13:59:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-29 13:59:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-06 08:10:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-29 13:58:47.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"46939f01af0ad1845edca6a326841304bdc62bbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-06 06:49:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb64a96432f56ad53750b8fd268cc1ae5725d9a6","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-06 06:50:14.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ce05b6dd49c3e3bd13e89c9c3cf44e3ae77b66e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 08:10:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf11cb0cdee0478dbf8b87c775c03507\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c24928877204c9ca9067d9f1fab0c61 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/800b3688e95244439f1eb54f2887d3dc : SUCCESS in 54m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/512820c34c124129b6f28ccbdcc5b98b : SUCCESS in 1h 06m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bda4cedbb79b49b7979148b2f7f4619b : SUCCESS in 40m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/741484f97b7248ac95d5bd3185a4fb3e : SUCCESS in 59m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43c34e0b172743059288c820ccb2e0cd : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c362cd71bc1c4a59ba208b582370c041 : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ed4a10a40564f468743284de173e85b : SUCCESS in 21m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3930e8e897f4690bfc4e19c9bc5289b : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b1bcaa0b14dc4ca49e15bcc0331b95cd : SUCCESS in 20m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de0b77e9026a4740a227d1b816e2f9e8 : SUCCESS in 17m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2a778a9b1ff942159a1c2a2bd1b48bc3 : SUCCESS in 19m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/60d54030ba8140c18e7c1e1673cd2699 : SUCCESS in 18m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/90839e36f1a8438997ae30abb953fe02 : SUCCESS in 18m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f31230332554465bb87f7db35642854c : SUCCESS in 10m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/228e8e36197249639f8e2597913f9658 : SUCCESS in 16m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/76aa03e2fc3b48ba99ea4a7258e4360b : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0e39c5d3963e4190b3aaed6f837662c3 : SUCCESS in 14m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/041fcf5acdf44ca7aee09bdb56a961ab : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2285f91adec44843828193c23b73cd7b : SUCCESS in 5m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8477d2b0e8b247398667d550dc2276e1 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/786ba8084b4a4f0e820b8d5e44b25702 : SUCCESS in 6m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbd27ba002fe40c2a2bb77009e7d684d : SUCCESS in 53m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e3e28400d824f6ca519fd25cea54164 : SUCCESS in 37m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/23ad7742d68c4980af31efcda89bed21 : SUCCESS in 56m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a5b0e6bc6404892bc4c2b85171a828e : SUCCESS in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b742ba05ced46b6b82fdc838cf5ab69 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f95d50dc50e4acab0df5e1fd7e8c100 : SUCCESS in 18m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f673d26f9f714efea06dbfb3a4431c9a : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":1},{"id":"44f1e330d4e0f0b2be3766da0345be0bed6c69ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 08:35:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eb7a45da20549f6996541e31b4f8da7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8838b58fe6a1411f8f3f9d5ab5754d8c : SUCCESS in 1h 43m 45s","accounts_in_message":[],"_revision_number":1},{"id":"372e9c656f93749ccd392dff91e2a193539937b2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-08 06:02:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"feebe02a7fe79688b5e68cdea7cb78d84704d446","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-08 06:02:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"de4402410d77eecc95d9d43c1c3bceaf3ba235dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 07:12:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/628efd2f185c4a668e8214084943512a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a805208a77e84471aa32cb9a46dff61e : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/4d354769a7894a1589f2f7b0077b1368 : SUCCESS in 1h 04m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b77635bceaa44963885073e389ae67a6 : SUCCESS in 59m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/66dddcbad85e41f3a7a1e0d2fd62e15b : SUCCESS in 49m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/617b2ac2dfd841c6b32c0a4a1d667a86 : SUCCESS in 1h 03m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cb53092f822439098acf9a9fc01b87d : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00598121adad41609be0ea788baca797 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17505c1bab7f43cf93ec4f48cbc8fc81 : SUCCESS in 18m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77bd7e75bb0140acbaa146aae7f2e979 : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f35be1910c24ed5a9966e079c400534 : SUCCESS in 18m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d3a4ff171f814323a5b89fd44eb8f239 : SUCCESS in 15m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8e60f66530e14790bdbc1be91fcc7798 : SUCCESS in 17m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/df254f5df3c64fe899aeec267ebefca6 : SUCCESS in 21m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/86dc40e554a24ca3bb423a21db4b4ad5 : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e3acdd50c0514cefabbaebf0e3d79248 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bcbf069bda024940b205cf7887e6400b : SUCCESS in 12m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a32fce76031544ce9e29dd053fdb5169 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5c5989925dff4b269386c344ae1730bc : SUCCESS in 10m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/166f41dcd3eb4588869694180f4565b2 : SUCCESS in 14m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a08deafa2662457a9cfa114adadd39c6 : SUCCESS in 4m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/08cbc0c2f48640a1bd37a53589ee639e : SUCCESS in 12m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3fba4e69be0545dfa0fc45476fe5134a : SUCCESS in 5m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/998b3816c4e645cf90899e9aa36cb7f2 : SUCCESS in 42m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d4f2742d8f5476ca1021181261d1535 : SUCCESS in 33m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6266df4a158444d9bd0eda207bc6dca : SUCCESS in 33m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a20644c7d58d4569b35f4e4793d19f97 : SUCCESS in 19m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ac2d94e2c5f4d9982f0614063a6ee1e : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ae16185ec39849ee8e1759e8ecb973ac : SUCCESS in 17m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f46746ca60f44399995560e393f89121 : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":2},{"id":"0ab2bf578ec57b2abe3c245d8f0d25744d4e1a82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 07:27:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20be1a4825384ed1bb55202cee752ea9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2db0ff4825f1468a9d069abf81cec27f : SUCCESS in 1h 23m 19s","accounts_in_message":[],"_revision_number":2},{"id":"8706bbab37314b39f42af882bff417b0d65dc45a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-29 13:58:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"75381c959ee5b9791a1c73953d93ff3453f9ed24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-29 13:59:31.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"71bc6e83881127d72fa6dbb0c2d9f4084fbe239f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-29 13:59:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6466e52bb38e3693320d224a036dcba88166cd40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 15:16:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b270ac9dd6dc4107bcd8dbca3a7125fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dda0f28b07b4e85a81dac609bc51f71 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/f5af3b5eb3dd4a3190f405da82d85eff : SUCCESS in 1h 09m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d3b396378a54123b84c3c551ede920f : SUCCESS in 1h 06m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3acc2482efba44e4993569406255aabe : SUCCESS in 57m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/466275f430864e148f8fac2120df1481 : SUCCESS in 37m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e23099e9ce44a8f9c07c926cca1e6f7 : SUCCESS in 13m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1cc2bfc8a794cd0b7ebec8251c496c8 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f4609dc0e7744988799da36dca474d1 : SUCCESS in 19m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e84a619774e437fa506d8fb7a544771 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ef711b9dca854e16a291769aa2a28fb5 : SUCCESS in 19m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/07d8117dfdaa40dbae43f130d59d1d40 : SUCCESS in 24m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1acb9131be784c89aae470947db7578d : SUCCESS in 16m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5be49b6aca2f43c3987ba33635d6bbe8 : SUCCESS in 38m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a6cdbcf15882484090994fcabd90bf11 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/04b14d15cd124cf2b7168d77af4d03e8 : SUCCESS in 22m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6f3c8e0c2b3e4777b22a2cb1be1f3b88 : SUCCESS in 18m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8eba77d3d4744f78a76a8825de701563 : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e8c90edbec8d4f5691b96c52e213c7a0 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/db9296314ef24a7695c82937869779cc : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/06c3e3698949480589f1d857722524b0 : SUCCESS in 8m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1070620fd9b4983abe172a2d7669c2f : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/211b13d42b4846e59dada40399a6fc4e : SUCCESS in 7m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4f7e762f88b4e7a8d4e3ccec9b512c3 : SUCCESS in 54m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/02138842d1734b19bf7833c8378996fa : SUCCESS in 47m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5bddcfecd3d44b2696479118512540d2 : SUCCESS in 51m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/265649ec94ed43118e57b0f2428e7cba : SUCCESS in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d992c4b0812d49148af4a905a71ae72d : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/689a57b909904cdc8e37b31711ae8747 : SUCCESS in 18m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f509b064d2d48dd937ef46f2924e2e1 : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":3},{"id":"3f3e922658e3a07ad2475c8ef4a8e95d894d42bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 15:16:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e55dd87c11782b03504f91580a65284dbf021ddc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 16:32:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe051d0b6ae44f6e853744b9501d3c75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c7b3938f58431c82a0514cfeee1e7d : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/032100707a5b45a2b87eb17015033d22 : SUCCESS in 1h 09m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d966886bdc4c4bc6859b4d5077aed9ea : SUCCESS in 51m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d4fd9d751a746c6b0913d785c20b618 : SUCCESS in 59m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3dda06f1767428890786ceea0d2d147 : SUCCESS in 1h 03m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/020478b25ef44ae9aa3a862719158352 : SUCCESS in 9m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/29de3246676f4328917cf2cd12f3e022 : SUCCESS in 29m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b88192ff61b745aa9f13cdb18a97b028 : SUCCESS in 18m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cf38d31137424776a817f618cf4bed17 : SUCCESS in 22m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d58ef07441764c6ba251dfdae75839fb : SUCCESS in 25m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5008d86a0b29416da25a93d63df3a271 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/25c544c7f0cb4cb3a2f3738fac4da543 : SUCCESS in 17m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8b0c60e8876442a48d2bffdb1928f4c0 : SUCCESS in 19m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/92ea72787a5144df81b145fb9d6b9fc5 : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/403fb1a3ad934791a86e35fd466113cf : SUCCESS in 23m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d3923d3469a94bdeaa2b94872f58ad93 : SUCCESS in 14m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9409c30c15546c887cca10cd7f7f866 : SUCCESS in 10m 39s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/248ff70caad2492e9d91591cb2b0499e : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d060d9da9b574ff8981098539cdd4485 : SUCCESS in 52m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/422b48ac7043415ca903f8ba01f0f391 : SUCCESS in 56m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/614909f4fa5744369a11f664a6c0fe38 : SUCCESS in 38m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e10aa9f1650e4f34a13517cb06a7f93b : SUCCESS in 21m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/150fcbb2c2e04f1dbb9abd770ff9c06d : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb750df950cf4535ac7ed0589dd98a90 : SUCCESS in 17m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2418830bea974405ae03080ab994e0aa : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":3},{"id":"94d98b043092810f59b6aff75a380cbf2a6bbeaa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 16:32:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"215d9fd42156f506cedc64c98d2d5ef0c14b2f49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 16:33:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8766053b65494d2f8f997c0ee031bf17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ee73fc74cb45b283422f6d4e77dda2 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/60e6fb2773e34b0aba6637b408fbd396 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c67e43a6c90be024e13c3d1fb8ff8a351dd44c8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 16:42:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef908aad00f14ebc8ac4789468b766a7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f160aed1a5dc475c852bfc99a483848e : SUCCESS in 2h 27m 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ea1d84c1d7fe671fc1bf1f72f2cb6e3c24087730","revisions":{"3ba3d00b770fb05b8358a307168ea46bca77b317":{"kind":"REWORK","_number":1,"created":"2024-11-06 06:49:56.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/00/934200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/934200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/934200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/934200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/934200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/934200/1"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-06 06:46:44.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-06 06:46:44.000000000","tz":-480},"subject":"Object-server: add periodic greenthread yielding during file write","message":"Object-server: add periodic greenthread yielding during file write\n\nCurrently, when object-server serves PUT request and DiskFile\nwriter write file chunks to disk, there is no explicit\neventlet sleep called. When network outpace the slow disk IO,\nit\u0027s possible one large and slow PUT request could cause\neventlet hub not to schedule any other green threads for a\nlong period of time. To improve this, this patch enable the\nconfigurable yield parameter \u0027cooperative_period\u0027 into object\nserver controller write path.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I1c0aba9830433f093d024b4c39cd3a3b2f0d69f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ba3d00b770fb05b8358a307168ea46bca77b317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ba3d00b770fb05b8358a307168ea46bca77b317"}]},"branch":"refs/heads/master"},"fdc5f8dd56cc28ccea6303bf2606a9f314ee9647":{"kind":"REWORK","_number":2,"created":"2024-11-08 06:02:04.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/00/934200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/934200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/934200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/934200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/934200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/934200/2"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-06 06:46:44.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-08 06:01:43.000000000","tz":-480},"subject":"Object-server: add periodic greenthread yielding during file write","message":"Object-server: add periodic greenthread yielding during file write\n\nCurrently, when object-server serves PUT request and DiskFile\nwriter write file chunks to disk, there is no explicit\neventlet sleep called. When network outpace the slow disk IO,\nit\u0027s possible one large and slow PUT request could cause\neventlet hub not to schedule any other green threads for a\nlong period of time. To improve this, this patch enable the\nconfigurable yield parameter \u0027cooperative_period\u0027 into object\nserver controller write path.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I1c0aba9830433f093d024b4c39cd3a3b2f0d69f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdc5f8dd56cc28ccea6303bf2606a9f314ee9647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdc5f8dd56cc28ccea6303bf2606a9f314ee9647"}]},"branch":"refs/heads/master"},"ea1d84c1d7fe671fc1bf1f72f2cb6e3c24087730":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-11-29 13:59:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/00/934200/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/934200/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/934200/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/934200/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/934200/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/934200/3"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-06 06:46:44.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-29 13:59:25.000000000","tz":0},"subject":"Object-server: add periodic greenthread yielding during file write","message":"Object-server: add periodic greenthread yielding during file write\n\nCurrently, when object-server serves PUT request and DiskFile\nwriter write file chunks to disk, there is no explicit\neventlet sleep called. When network outpace the slow disk IO,\nit\u0027s possible one large and slow PUT request could cause\neventlet hub not to schedule any other green threads for a\nlong period of time. To improve this, this patch enable the\nconfigurable yield parameter \u0027cooperative_period\u0027 into object\nserver controller write path.\n\nRelated-Change: I80b04bad0601b6cd6caef35498f89d4ba70a4fd4\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I1c0aba9830433f093d024b4c39cd3a3b2f0d69f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea1d84c1d7fe671fc1bf1f72f2cb6e3c24087730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea1d84c1d7fe671fc1bf1f72f2cb6e3c24087730"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
