)]}'
{"id":"openstack%2Fswift~934350","triplet_id":"openstack%2Fswift~master~Idc98cab9f1eab75e9e3f4a6b22177a9fc1250120","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-14 20:07:21.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Idc98cab9f1eab75e9e3f4a6b22177a9fc1250120","subject":"DNM: investigating probe test timeouts","status":"ABANDONED","created":"2024-11-07 16:45:06.000000000","updated":"2024-12-03 18:10:59.000000000","total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1419aad6c287f7729dc182f14cffe5d082988178","_number":934350,"virtual_id_number":934350,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-14 21:10:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-07 16:45:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da0915677cc39303bf9484a4513d0ec0d3660280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-07 16:45:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2d8e331fd93fc007491d039c01f7b485513d965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 16:45:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  The job \"swift-probetests-centos-9-stream\" exceeds tenant max-job-\n  timeout 10800.\n\nThe problem appears in the the \"swift-probetests-centos-9-stream\" job stanza:\n\n  job:\n      name: swift-probetests-centos-9-stream\n      parent: unittests\n      nodeset: centos-9-stream\n      description: |\n        Setup a SAIO dev environment and run Swift\u0027s probe tests\n        under Python 3.\n      timeout: 36000\n      vars:\n  ...\n\n  in \"openstack/swift/.zuul.yaml@master\", line 323","accounts_in_message":[],"_revision_number":1},{"id":"b90281ef319558d8830f985b57998f3c299e1ea4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 16:45:36.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  The job \"swift-probetests-centos-9-stream\" exceeds tenant max-job-\n  timeout 10800.\n\nThe problem appears in the the \"swift-probetests-centos-9-stream\" job stanza:\n\n  job:\n      name: swift-probetests-centos-9-stream\n      parent: unittests\n      nodeset: centos-9-stream\n      description: |\n        Setup a SAIO dev environment and run Swift\u0027s probe tests\n        under Python 3.\n      timeout: 36000\n      vars:\n  ...\n\n  in \"openstack/swift/.zuul.yaml@master\", line 323","accounts_in_message":[],"_revision_number":1},{"id":"44e320fe5055bb375c80f3383ecf4f99cb98a297","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-07 16:46:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e48b95014364451e62004f652320ad123ae73cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 18:18:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd8a198aad074cafaf4ad5c9a0984dcb\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/41903811021743bd99c06d83c5e40866 : SUCCESS in 57m 45s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/8ad03a12978f4f0584d339af11edbdf6 : SUCCESS in 43m 54s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/d4f3ad46492249d4a4a4584d10e9bc01 : SUCCESS in 52m 39s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/3e6de4196f9049289c12d6fb60bb59f2 : SUCCESS in 47m 58s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/64e554d46ffd4661beddebae4000275f : SUCCESS in 47m 46s","accounts_in_message":[],"_revision_number":2},{"id":"22dd0f7b63a5d8edbd275daad1538e6007009222","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-07 18:33:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a606c3709b0f33271c0e2cdb4abdf6602f581d7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 19:32:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79c7d6b6fa9f43a99aa6ee422237b5d6\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/14d0224fa09f42f6adde8dc69e1a8901 : SUCCESS in 48m 47s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/7cdf9484a02948efb8f5eb866e1802cc : SUCCESS in 51m 11s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/0df70f7f33d146758ed8bdcad438f5fe : SUCCESS in 52m 26s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/f304c1c20ab14f558496632c4237e7c1 : SUCCESS in 49m 22s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/4082dbe08c79452895681c6696db76f1 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":2},{"id":"0ab5f3aed2e20602a9b4d5f42dc0fb472aec6545","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-07 19:42:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"feb9abb8d5fb5339af33e4069fba75c41aa75242","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 21:12:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5249adcd59442a2a7f6031c3d53048d\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa5885d1403b4802943637540f35434e : SUCCESS in 48m 32s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/f12f25474cc44f44b53211a231cad390 : SUCCESS in 45m 29s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/483a849f7c474638965e86d3f3708f9e : SUCCESS in 57m 42s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/53d212e67ade4f489b0c78dead534c66 : SUCCESS in 54m 04s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/1c8cdf60083940a3b94ed7cef0a660cb : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":2},{"id":"1c4c53e39ca3e35c277c491a8c3d881e447dea31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-08 02:06:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4dc8a04be4962bf10a73f9e1808e5fc892ca441f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 03:29:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7579856adb4b4b6e94617a7647baf493\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/85b08d5cd64348f486edc2b46e987d76 : SUCCESS in 54m 35s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/42ff3bf552f24f7e86083ad45e1b9778 : SUCCESS in 50m 54s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/d472acdec3fa48f3ab67a336bd4db70f : SUCCESS in 47m 43s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/f70c5ea07cc64ec597a2e25a8d688c82 : SUCCESS in 50m 19s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/2e2fa5963e984719a945191f89b50e39 : SUCCESS in 47m 43s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/c24087c7e8314418b8dad439c9fc7ef6 : SUCCESS in 47m 21s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/ed24dad4f9ab4968bb58effaab348537 : SUCCESS in 50m 51s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/59b88afb9b04491b959e8af156a5573a : FAILURE in 53m 36s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/c3668ab8348c455c8919b43a721541c6 : SUCCESS in 53m 21s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/9edafb70b86748a0a0887ebbcdb12ea2 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":3},{"id":"85688985c0adcc0611c74375a6652285e7f43607","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-08 15:30:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"95dd3f2c79b75ed6020048829a2043309cf379d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-08 18:11:11.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"60287a902f8cfa2c9be043ea023e29f14c4bf64e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 19:23:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95490c0543f046a1b698c9664ecee27d\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea143aaba7d746bb91b9f4fa59991641 : SUCCESS in 58m 06s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/601346bc942b42c29811f2e3ccdbb4ec : SUCCESS in 48m 31s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/9b9db7ea71ff4a21af93122f7305d3cb : SUCCESS in 58m 50s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/6ff2cf81439c47c4baac50b5487caaaf : SUCCESS in 54m 55s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/4f194f9868124946bf77a46ffaf10aa7 : SUCCESS in 43m 18s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/a387048c6d67405982f7fa1465e6af50 : SUCCESS in 53m 15s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/ef491370f9fe4eb3852ecb2257e7ac7a : SUCCESS in 47m 59s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/209337ca0c7644a6a944cdcba9ff0c54 : SUCCESS in 50m 16s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/fd106d3ea1f5479c89ca0a143b40d335 : SUCCESS in 53m 43s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/64a47b7729814f898596dc27e20ed677 : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":4},{"id":"0f19b7dae8ee58810b0085e862fb15720636188d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-08 19:40:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f4d65c59886e912a977d74ec04c0ddc73c42b581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 20:55:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/769e959f022b40e0b93254270fc5ac7e\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/675d638af1e94d57a75d7e6069863b23 : SUCCESS in 57m 53s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/d856ac804cd848819627ce85462e332c : SUCCESS in 55m 16s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/1498f9bd2fa04ff78d58ec228be15f3e : SUCCESS in 56m 06s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/d9abfbd079084a648bd011ee8d258012 : SUCCESS in 53m 21s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/e7f66e7d5cac419f8fc9e094c2207943 : SUCCESS in 53m 44s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/a9f16fc16f8440a39a912fd5b59c1b0f : SUCCESS in 49m 55s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/fbee7b8f9fce4582a7d84366f75b1275 : SUCCESS in 50m 07s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/5db7b87b466640c4afc0f8349d94a6f9 : SUCCESS in 46m 25s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/7f87165ce25f4fe88987d9cf1e5ef43b : SUCCESS in 50m 41s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/e37e4abc23e6442ba80c2780685c828d : SUCCESS in 47m 38s","accounts_in_message":[],"_revision_number":4},{"id":"d9ca79bf5643d7960d92ddfc4bb7e8db4d0550f5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-08 20:56:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2e2a3a4e34f87e06ea49378154e89ff4969b264c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 22:00:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/069543ec30704accb7c5ce311f975cd1\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e00ba557e814c348b5be0ec926c2abe : SUCCESS in 51m 11s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/9750d1043562443a96cea48aee226f23 : SUCCESS in 50m 16s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/0443205512c44eec9ce63750eb40204b : FAILURE in 38m 46s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/5d2b4393f3c940a19e69716d8c4c79df : SUCCESS in 49m 15s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/71080bb5e00f42948800cc9b0f7ab88b : SUCCESS in 42m 58s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/81ed4bcb8d8740cb856f2d21834fb00e : SUCCESS in 52m 02s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/7489e44cd0ce483fbb87a3bae7f4f7fe : SUCCESS in 49m 12s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/a0df05e2de31417e924d614f44aad2db : SUCCESS in 41m 14s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/2e7cb31f64c84822bd76ddc1458b511a : SUCCESS in 51m 37s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/a865023af49344499caf49a599ba973b : SUCCESS in 49m 14s","accounts_in_message":[],"_revision_number":4},{"id":"d609d318172544313fb15efc4631edd6b8b4991b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-08 22:42:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1fd1390eca363a6a712814333825c8283b649968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 23:43:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f07c6216cd14ff49dc779de5c24260c\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cc5c1a2324142c8a0a0a0d5f90e989d : SUCCESS in 52m 08s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/7bbf1603849c4d358eb9b248fa03fe81 : SUCCESS in 55m 01s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/2b8b591dd3f34789a77ccbd455b82037 : SUCCESS in 42m 48s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/f090ba4c71b84a9c8d4bea964b458b17 : SUCCESS in 42m 52s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/cbc5d4a74add417e915c68817e689673 : SUCCESS in 54m 02s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/d32d816c1b394ac593f778159ac94ef9 : SUCCESS in 42m 15s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/7d47f78d24b94648be18382bdbc41c81 : SUCCESS in 50m 12s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/e3616902abc74340902de5556986a15e : SUCCESS in 43m 20s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/ee82c4f3e4114f03aff7b70221297f09 : SUCCESS in 54m 04s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/49a8b7315f87454592b5c867b3500f71 : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":4},{"id":"7ca454bd5363f39ae06e0be03b81723678f71df7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-09 00:17:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2851ec125b461b3b5b42ce25576d9baa53d14cae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-09 01:18:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/230a081f683b4d83b7ad9627610ea9be\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f75a86bc714c4ce3a6243f8ea8b24dc1 : SUCCESS in 52m 30s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/b7e021ba266543f895f5d86e6fda70ce : SUCCESS in 55m 42s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/21dcee7c9993478ca7d20059c8647f13 : SUCCESS in 52m 24s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/9bd8b9f9f69c4b6891d216c4ab4b85cf : SUCCESS in 50m 52s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/c5c03b60bdad4548a41e81d90c9156a9 : SUCCESS in 52m 55s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/7e74e99f60c24a99812b079eac825a71 : SUCCESS in 52m 15s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/2498ec9001744ace926a381fc7bf9e2b : SUCCESS in 48m 32s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/f74b976e3275492581c4ccdceb2b79e0 : SUCCESS in 32m 02s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/f2982714495a49e892384b031a8c300e : SUCCESS in 54m 04s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/edf23d3e355d4890abf223396a005310 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":4},{"id":"8aaa081e08353f73129f8fd2495ac6cd0f65b4f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-12 17:39:00.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4be61bd5e6f62ba3e637c6690598adb015aa50a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 18:53:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3840d1bc18b544d3927f39f04959e412\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8b790271ed6458ea98a83c192952c2d : SUCCESS in 39m 50s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/da1b5b0488a0423bb01b6d509a2200f2 : SUCCESS in 59m 09s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/4d29a5a99ab84058932d1330b5ba9e38 : SUCCESS in 44m 31s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/b965ed7a9b284ace9921fb15beed9329 : SUCCESS in 59m 47s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/108f83b394a84988b08ecf0b9b0b27e6 : SUCCESS in 1h 00m 18s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/cc6e480d2a5d4283b7e4f57beaaa2ab3 : SUCCESS in 40m 02s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/79457825077149c2849ba8565edc4341 : SUCCESS in 56m 26s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/bbe1721dc06347ef8cd5e65dab410ef3 : SUCCESS in 50m 47s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/67ecdd9ded1244c5962f0d6417c0f921 : SUCCESS in 58m 05s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/cf10e89537fb4a27ba537a2094f186b3 : SUCCESS in 41m 20s","accounts_in_message":[],"_revision_number":5},{"id":"4a4d62014b9b2985a011d9119730d6de425eaf27","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-12 18:58:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"41484dfaf7fbc3c00edf9a9aafb5fe1e7c5e809f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 20:00:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebff3ad64d6a410daf0673a62b18d42b\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2217cef4479d4835892d26fe5b1a6668 : SUCCESS in 56m 23s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/b5ea33cd0a9f4864b5e57b271f21fd2a : SUCCESS in 47m 20s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/0772b70753854527a920e28d2643c9fd : SUCCESS in 48m 15s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/474de4107048455cb860cf4c930bbdd4 : SUCCESS in 54m 16s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/703565860f8142dea0cb986c9733c880 : SUCCESS in 45m 18s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/786c7f9bd21b42458a5f0314ea3a78af : SUCCESS in 50m 54s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/d2d49e6988c444deb0f0370bd9303b7b : SUCCESS in 47m 42s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/7daa5845e8694e3692ead357d210649b : SUCCESS in 39m 14s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/149bdf2ac8ac4a68be00ce2fcfb74077 : SUCCESS in 54m 22s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/23c8591b759348bd818e1324651622d6 : SUCCESS in 41m 26s","accounts_in_message":[],"_revision_number":5},{"id":"9a7cada4ece4c99bc7dee850cd14372b28553c92","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-12 20:24:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0384de64950b0ae7aee8488a08c91c345dfbf88c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 21:34:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d987a0747d54ac0bafec71881f789d6\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d1b0da94614435e835e367dc64290b5 : SUCCESS in 58m 04s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/8243ea8177e24a3ab9d447014edd1419 : SUCCESS in 49m 41s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/87fcc726cdbc4dbdad9cbdf69c0ff1d0 : SUCCESS in 39m 22s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/d4bd2c5019fa461cafd519bd0fd8b5e0 : SUCCESS in 47m 30s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/abab4dfb5e2a40468ed4ee9b909f42a1 : SUCCESS in 49m 35s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/b0f997e10e9040d69f35bf0167547518 : SUCCESS in 58m 16s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/d36f38455ea141aa83b33e13de188af6 : SUCCESS in 51m 30s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/b429a2c27e1d4050a25023e738d77333 : SUCCESS in 54m 45s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/055ede96ac53442dbf5bf6819a531afd : SUCCESS in 51m 45s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/08f5f11652334fabb3d9ff37ace3c5ac : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":5},{"id":"1d5689289e0bf8d3c7c4c9b94115508c1c3d15d6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-12 23:13:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f13b089f64a3cbecad90272418305c05448bf53d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 00:15:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f12506f4388148cfab0a37ffd4f3846c\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e612fe795f1a4ead8c0ab6883b8b3e55 : SUCCESS in 56m 29s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/7c0155ec336c4f6eac4f5fa6b7bf7061 : SUCCESS in 54m 20s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/25a5cb77e174468aba312b949d2be795 : SUCCESS in 55m 07s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/10bb8a91b7c74fbb8afeac6a0a6c1486 : SUCCESS in 38m 41s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/725cfbb85b45404696d1f8a1ed7a10a1 : SUCCESS in 54m 00s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/3c185e5800354f9e98128e35841993fd : SUCCESS in 43m 11s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/5abae4e6e86745b1a169d8432b1e9496 : SUCCESS in 54m 04s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/014124b040b34b7eb6acf1fe865abeb0 : SUCCESS in 42m 16s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/a659d5e5d0fb460b92a188669ab9ec97 : SUCCESS in 48m 23s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/c3467e2f68484e35b8f3ec26b1adfedf : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":5},{"id":"13f7c92923f7b8ac667de35911dc8da889a413d9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-13 00:42:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4294a4dde01f26fef128f731f0c2bf01bf634060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 01:44:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/369f8a6e941541edabc3bfee2b4d48db\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c49f171c36af4e0680229926d49efa06 : SUCCESS in 55m 06s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/f667d2f7ff2f4691bfe3230375b77758 : FAILURE in 54m 21s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/7df22263334c4359ab39ede3d1c8f263 : SUCCESS in 43m 23s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/9f78c17c290d4aef889d0c746ed7e993 : SUCCESS in 33m 01s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/aa380771320b493b9056979bd862f4e2 : SUCCESS in 54m 20s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/e46568457e8c479b948daa2d0974f23b : SUCCESS in 55m 05s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/0f07ff3c0b5d4dea841922cafbea012e : SUCCESS in 55m 10s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/768f29f4aa53400280c20b79143e845b : FAILURE in 33m 36s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/19bd59cb482e4867b45a6bd9902612e8 : SUCCESS in 47m 45s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/3731d8a2c74c438295b70fc8752bf39b : FAILURE in 47m 06s","accounts_in_message":[],"_revision_number":5},{"id":"92d53609557229668539933b39b953435d926df3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-13 23:15:56.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f1c305bb6b7c522c98371a87509398fe05e8f426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 00:16:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1738b9eb22704c45859ca12c921d49d9\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfa70212fd5943fc97daa6462767bd84 : SUCCESS in 54m 21s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/45bc8757f825485b9b4448bd1ba3f69d : SUCCESS in 54m 47s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/684fb9fc4caf444f91a3650ad497bbb2 : SUCCESS in 49m 25s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/1a857c58654d415c883f9785af495618 : SUCCESS in 49m 24s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/e571bcd87c654a93aeb56e5dc74f88d8 : SUCCESS in 50m 44s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/b0a92343a28b45508840fe1711084d26 : SUCCESS in 51m 48s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/214ae14d17a34aad899bb50d2c8f368e : SUCCESS in 49m 01s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/bb7595f129b1471fb9387918f7dc511d : SUCCESS in 48m 19s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/66cfcca5f6454e45891f9d1a2f776da4 : SUCCESS in 49m 46s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/344a670d621a46cf82db1c8de369c542 : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":6},{"id":"7a6f203e011568330cf7a40aff8f84366a68a93e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-14 00:25:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"101c17a1061f31f9998bc115ad78c5fab183b7db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 01:27:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0de369fad4a044dba76f928da03f03c3\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/57d5e840b3ed42f8a04bd27cda2fe179 : SUCCESS in 34m 29s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/fcc0bda21bcc478e82cfffb47c02b2fb : SUCCESS in 52m 41s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/c113b9dc7757499ba0db4d742a2a7ed9 : SUCCESS in 35m 23s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/8eb89744e65f40f086f1e8cabe4db98a : SUCCESS in 53m 48s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/a73c2960ec6b4a89a9d1a2b24d316999 : SUCCESS in 35m 33s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/12dc7b7ce24b4dd2b305a2b3a9eb2cdd : SUCCESS in 45m 50s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/16f914d02bdd45a8957885cd5c3f54a4 : SUCCESS in 52m 40s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/02daf5a77fac48188f4281713f84e645 : SUCCESS in 34m 10s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/a56a82ca418e4c05ba54ea827385541d : SUCCESS in 46m 58s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/23672ab743e64d2cbcf9ebc97b3dd437 : SUCCESS in 34m 05s","accounts_in_message":[],"_revision_number":6},{"id":"da2bbd5a51e5e67bdf452d9f070f0e7b5837c001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-14 18:32:55.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"80d70c1bddb440e8270182b74f66a5f6fa7ccabb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 19:55:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44f604f2a44e4131b44bb8b70df3a0d6\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/33807dd23e794e26ab2f65b1e5b34199 : SUCCESS in 48m 15s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/476342a2c2c44783ab28a5f20a183737 : SUCCESS in 50m 40s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/0c034e3d907341b7b671fc7976321e41 : SUCCESS in 38m 21s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/50baea2414c1431d8f5e7073d327dbf3 : SUCCESS in 40m 51s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/fa789c5dc9ae4771ab0e397093e3a40e : SUCCESS in 47m 21s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/0045d0f965954558997ddb7d91738dd0 : SUCCESS in 40m 08s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/41aed7d1e4f74fce8a5af65dfc960801 : SUCCESS in 46m 16s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/5e7c7e74ca4d4969ae5a631ea11bc3ca : FAILURE in 47m 41s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/ccd6319d33b842f1af8743f0ef85519b : SUCCESS in 52m 08s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/215c521c5ad8483d9afb0acf4ee80458 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":7},{"id":"fda76a5e7105cbc1bf4f66585b114658986f0636","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-14 20:07:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2d911b8d28ccc0660a51e08c8797ebfaf6ca2bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 21:10:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eceb5b9a4744fd4844b33f56e510189\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2fd61e6ff994ac38b576f09474a264d : SUCCESS in 52m 35s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/8c8bee8d4aef434c93221d7c50fb5fc9 : SUCCESS in 51m 55s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/16cd85bafc1f47b3a34968c837d493e8 : SUCCESS in 57m 23s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/23aadc56467e474eb8f8f5d99aab4634 : SUCCESS in 56m 06s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/c868f1a0829e42b8bb702fb8780a484c : SUCCESS in 40m 19s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/e118c8d136ab49a7a130d5aa63c0303c : SUCCESS in 50m 55s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/912c75b7c08f4c7d9d7bd22157d8ee61 : SUCCESS in 46m 14s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/185458aad7d049b5823e351710659b0e : SUCCESS in 50m 07s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/0f90c316392a4adb9c93e1ab026ee237 : SUCCESS in 46m 48s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/b792d016c31447119d9f74ae4758fd11 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":7},{"id":"9360ac90f26d5ebef3d2efd43f65213c6068b980","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-14 21:19:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"752b6346166fb65fefddd77e50c6791fb5fe589e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 22:34:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48cfa39470a44415b077892fd8bb314b\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d0615ecbc61418d86d7582aa3115657 : SUCCESS in 45m 38s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/ba4158904db94ab8845be68d654f9c81 : SUCCESS in 57m 48s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/23447b4dd9814b738b26f9274f0cb53a : SUCCESS in 57m 03s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/cb3cf41ca69e4dbfa21ce82ce3e7604a : SUCCESS in 57m 24s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/b9e568279aba4802a6bf54a02477dcdd : SUCCESS in 56m 15s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/9a2819c84fd84003b80f2ea3f037d534 : SUCCESS in 52m 46s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/0efa9bff9df049c9a2ff1e9f2863abf0 : SUCCESS in 52m 18s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/7657d54ff5ea42fa8962ee68e6c10ccf : SUCCESS in 33m 40s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/58a7da5153c0471e932b719db8c0926b : SUCCESS in 54m 56s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/36df0c90b5d24b079ac331f4d7bae94c : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":7},{"id":"a6e42afb2826158fab97b244501af0a063403249","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-14 22:59:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"12524726555f5d88d7bc3293d7069612ca125d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 23:55:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f86209edddb4326bc1b73d0b6c66829\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/83e9441ac78a40778d432c5bb38cc4bb : SUCCESS in 54m 13s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/e87c3835368f496cb47dbe97c8d97027 : SUCCESS in 51m 47s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/12028628d98d4220937dfb47f1632365 : SUCCESS in 51m 31s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/65af6f5e75c640ff8e839d4c3aaa7e9b : SUCCESS in 51m 43s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/c4c2be638d3c4e2ea0fe85a426462fca : SUCCESS in 52m 14s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/33100df565e74ddba63042643a26b7a2 : SUCCESS in 50m 01s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/9db1250527584ef285508c275339c1e4 : SUCCESS in 48m 20s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/6853f9d7ca61436aa31bc80d516a1748 : SUCCESS in 51m 52s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/57cf9a43225b4f339596d1ef7fff95fa : SUCCESS in 48m 54s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/0da2065c152541769ca44b4de46249df : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":7},{"id":"10670ef36eb272ae4fb83aab95c360f904a66b67","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-15 00:30:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6ba5c2bc1e1ef6bb0104b8924a1d92d157456f0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 01:34:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa864eaabfbe4c959aa40fe5f3dc1b4d\n\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2090defb420348a7b12aff94ef62b0c7 : SUCCESS in 59m 43s\n- swift-probetests-2 https://zuul.opendev.org/t/openstack/build/8f042bbc9b4c45e99469c053384c2d71 : SUCCESS in 1h 00m 47s\n- swift-probetests-3 https://zuul.opendev.org/t/openstack/build/31cce4e8c6d34190a95cc7df103b0643 : SUCCESS in 36m 53s\n- swift-probetests-4 https://zuul.opendev.org/t/openstack/build/73686d147ee44034aa5c1c38f3ff9ff0 : SUCCESS in 36m 57s\n- swift-probetests-5 https://zuul.opendev.org/t/openstack/build/cf165fb286e34cf8a1940771b164b032 : SUCCESS in 54m 57s\n- swift-probetests-6 https://zuul.opendev.org/t/openstack/build/bb98edde67ae41faa897267ebc25ec9e : SUCCESS in 52m 02s\n- swift-probetests-7 https://zuul.opendev.org/t/openstack/build/e0272d62de64497ba0d3173292b3a919 : SUCCESS in 57m 36s\n- swift-probetests-8 https://zuul.opendev.org/t/openstack/build/c64f2d492e1e4c04bba8894d2c9ddc54 : SUCCESS in 59m 50s\n- swift-probetests-9 https://zuul.opendev.org/t/openstack/build/83b9bda4b2e148c2acbe8d76ad9ed3f6 : SUCCESS in 53m 40s\n- swift-probetests-10 https://zuul.opendev.org/t/openstack/build/28156340981a4593bf4ee64bbbef71eb : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":7},{"id":"1419aad6c287f7729dc182f14cffe5d082988178","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-03 18:10:59.000000000","message":"Abandoned\n\nInvestigation\u0027s complete; see https://review.opendev.org/c/openstack/swift/+/934910 and https://review.opendev.org/c/openstack/swift/+/934813 for fixes.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0c8904c473e1a1176afc84925232a0bc65eea0ee","revisions":{"f426d06502de851bd572afaab62baa1d11968ed6":{"kind":"REWORK","_number":1,"created":"2024-11-07 16:45:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/934350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/934350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/934350/1"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-07 16:44:57.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-07 16:44:57.000000000","tz":-480},"subject":"DNM: investigating probe test timeouts","message":"DNM: investigating probe test timeouts\n\nChange-Id: Idc98cab9f1eab75e9e3f4a6b22177a9fc1250120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f426d06502de851bd572afaab62baa1d11968ed6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f426d06502de851bd572afaab62baa1d11968ed6"}]},"branch":"refs/heads/master"},"c1fc17c21d99e0d0392e13aa4da1467548fafe94":{"kind":"REWORK","_number":2,"created":"2024-11-07 16:46:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/934350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/934350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/934350/2"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-07 16:44:57.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-07 16:46:15.000000000","tz":-480},"subject":"DNM: investigating probe test timeouts","message":"DNM: investigating probe test timeouts\n\nChange-Id: Idc98cab9f1eab75e9e3f4a6b22177a9fc1250120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1fc17c21d99e0d0392e13aa4da1467548fafe94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1fc17c21d99e0d0392e13aa4da1467548fafe94"}]},"branch":"refs/heads/master"},"98921d3e8b61649a599ca5e3333bbdf570f8575f":{"kind":"REWORK","_number":3,"created":"2024-11-08 02:06:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/934350/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/934350/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/934350/3"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-07 16:44:57.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-07 20:13:48.000000000","tz":-480},"subject":"DNM: investigating probe test timeouts","message":"DNM: investigating probe test timeouts\n\nChange-Id: Idc98cab9f1eab75e9e3f4a6b22177a9fc1250120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98921d3e8b61649a599ca5e3333bbdf570f8575f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98921d3e8b61649a599ca5e3333bbdf570f8575f"}]},"branch":"refs/heads/master"},"a34b8042288e94b146d442cf64295d3954cd22bc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-08 18:11:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/934350/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/934350/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/934350/4"}}},"commit":{"parents":[{"commit":"29ff44042bfe51cde0f4169d437dcf86f8411210","subject":"probe tests: Kill long-running \"once\" processes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29ff44042bfe51cde0f4169d437dcf86f8411210"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-07 16:44:57.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-08 18:11:02.000000000","tz":-480},"subject":"DNM: investigating probe test timeouts","message":"DNM: investigating probe test timeouts\n\nChange-Id: Idc98cab9f1eab75e9e3f4a6b22177a9fc1250120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a34b8042288e94b146d442cf64295d3954cd22bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a34b8042288e94b146d442cf64295d3954cd22bc"}]},"branch":"refs/heads/master"},"3415eb3cb88083d86dca5ecdb73636e66c80fea2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-11-12 17:39:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/934350/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/934350/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/934350/5"}}},"commit":{"parents":[{"commit":"130c7ee30b1938b0bd67f94847fb17cd5312c369","subject":"probe tests: Set default timeout for subprocesses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/130c7ee30b1938b0bd67f94847fb17cd5312c369"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-07 16:44:57.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-12 17:38:54.000000000","tz":-480},"subject":"DNM: investigating probe test timeouts","message":"DNM: investigating probe test timeouts\n\nChange-Id: Idc98cab9f1eab75e9e3f4a6b22177a9fc1250120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3415eb3cb88083d86dca5ecdb73636e66c80fea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3415eb3cb88083d86dca5ecdb73636e66c80fea2"}]},"branch":"refs/heads/master"},"528e2592536f53216382b2b11e3b912a077c2045":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-11-13 23:15:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/934350/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/934350/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/934350/6"}}},"commit":{"parents":[{"commit":"2cc56922edd07ff4169af29cff13a3acb2d1fd8f","subject":"probe tests: Set default timeout for subprocesses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2cc56922edd07ff4169af29cff13a3acb2d1fd8f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-07 16:44:57.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-13 23:15:50.000000000","tz":-480},"subject":"DNM: investigating probe test timeouts","message":"DNM: investigating probe test timeouts\n\nChange-Id: Idc98cab9f1eab75e9e3f4a6b22177a9fc1250120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/528e2592536f53216382b2b11e3b912a077c2045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/528e2592536f53216382b2b11e3b912a077c2045"}]},"branch":"refs/heads/master"},"0c8904c473e1a1176afc84925232a0bc65eea0ee":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-11-14 18:32:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/934350/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/934350/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/934350/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/934350/7"}}},"commit":{"parents":[{"commit":"32fbf9eaaed6c52874a0e0c33c44ff869298f6c7","subject":"probe tests: Set default timeout for subprocesses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32fbf9eaaed6c52874a0e0c33c44ff869298f6c7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-07 16:44:57.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-14 18:32:38.000000000","tz":-480},"subject":"DNM: investigating probe test timeouts","message":"DNM: investigating probe test timeouts\n\nChange-Id: Idc98cab9f1eab75e9e3f4a6b22177a9fc1250120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c8904c473e1a1176afc84925232a0bc65eea0ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c8904c473e1a1176afc84925232a0bc65eea0ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
