)]}'
{"id":"openstack%2Fswift~934675","triplet_id":"openstack%2Fswift~master~Ib7b236e50ac844f4e001034abc1d090ca97b2eca","project":"openstack/swift","branch":"master","topic":"slow-rehash","attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-12-06 17:06:09.000000000","reason":"Change was marked ready for review"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-01-16 12:21:01.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib7b236e50ac844f4e001034abc1d090ca97b2eca","subject":"add backoff sleep to REPLICATE","status":"NEW","created":"2024-11-11 16:46:38.000000000","updated":"2025-01-16 15:44:43.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":23,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"50a9c578369ed21585b12fb392e2942e7db4b248","_number":934675,"virtual_id_number":934675,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-24 00:08:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-11 17:43:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-11 18:01:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-12 23:26:29.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2024-12-04 17:33:23.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2025-01-16 12:21:01.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"4ee68edcbb5c575fa7aeaee14f8b6b3027aa60d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-11 16:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1f32a48f1bee35628d764fd4075774a5544b51d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 17:43:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4426ca45f6fa496eac71543d7f1d55a7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/68dd7edd45e14b3590c91450adfd2fb7 : SUCCESS in 54m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6837317e582ca1cd7d2ce390871c9498e3db207b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 18:01:48.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b96aa58528a24c4f8fee15c297d62866\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7b279de4e054f7a87d8294caea6ee52 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/10c6be558bde48b3a5a12ce36bb6931f : SUCCESS in 57m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c9790aba64604cfdace0544db4dbcfd4 : SUCCESS in 56m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3261aeec7cf94f0cab1c6e5977e7258c : SUCCESS in 41m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99355dccfb754b25881e4887f50e628e : SUCCESS in 1h 09m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4055cac53ea64222b4bda2c93029ee00 : SUCCESS in 16m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e625099fce314ab3b0c0b7be80cb915d : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f48a90169a5423e8f91fbef8621447c : SUCCESS in 23m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6d8defb868a140868bb61a31d74a0510 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c441ca7e9d5c453a8078423564f3ad3b : SUCCESS in 21m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d275d8476b5e4273a29ed298291277be : SUCCESS in 15m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7aa9017694c64acd9b42a0b4c69c5792 : SUCCESS in 21m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/14ec51f5a52a47b783b4549f29e7457f : SUCCESS in 15m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8a9b389dcd15413ebb9a351a3d59fda0 : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b1248401a16449aa88dac4634490aad9 : SUCCESS in 23m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9c3e642ea94c4366b19564d0f5b361b1 : SUCCESS in 15m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a139428e50a84bb0817015df4447b176 : SUCCESS in 13m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/814ccd49478b44beae5e6f2669932ec7 : SUCCESS in 21m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5dabaef9ba6b4740b0c09cabdd0d3a65 : SUCCESS in 9m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/56e297d0d8ad4fb4b635a421c538242f : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5a0ccc6b7cc646b4afe5cf94d8226ab3 : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6c9eaf4bf8ac4cc8983508c1f8b84a9c : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/20b0288df01343909159e2a5dcaa508b : SUCCESS in 52m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/70a6a57fa0f54c05a52ce2b4f2211d43 : SUCCESS in 34m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c37b9d060ad4689b3fd3607f71dc6c6 : SUCCESS in 41m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc0af0db6d3542e4b1febdf08c4dbeea : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec685739239343de85828e8c7f947d31 : FAILURE in 3m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93fc8b3ee00d4e78bc920b7252ff7e21 : SUCCESS in 19m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7a4e3675daf4b9cbc6511ea35811a55 : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6fd2dddc164d38079830f63b8afc4a0d38eef354","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-12 23:26:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ec5be0788d52bafb871c7bd9a5967b63986f5f8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-12 23:57:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8c54eb4fc8530bc772f4bb3208614b2c80a6ab66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 01:02:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/501c79a933d54051aaa997f74df71d5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc6488dc47446df848e309e3df6bd55 : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/26737aca4bfa461da4a580930979ba41 : SUCCESS in 40m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7694181838324bed82ce298c303e7a11 : SUCCESS in 36m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e01a53febfa64d1a9167daf0684b8afc : SUCCESS in 44m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/121bc81a1a0e4941910f0693af11bc40 : SUCCESS in 1h 03m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b49581e12c5b4e9bbd4d7fc970ef98a1 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b87c52746f214b37b9b01b92a8a0fb71 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b38da1634b34468a93ec18fbe2b0532 : SUCCESS in 13m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e3d0439f0cac49558c394ecb8ebef36a : SUCCESS in 4m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cebe20c6bf13475ca59322ea19aaf9bc : SUCCESS in 19m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d8b18070bbd243269068c3a0698cba04 : SUCCESS in 14m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c20d869b3b1445f1a0bb2e1ead1dc4d7 : SUCCESS in 17m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/40ffb8923aec407f9911335cbde01ab6 : SUCCESS in 35m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/506af050a725494899881c7b0169c2ce : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8aa014c27db1406da2bfd1be8240b166 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0d9cef4ef3364f22bbe5e13ae90ca4c7 : SUCCESS in 18m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5f01f775e786402898edc16c3bc1188d : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/86d009c00a47433380d9383ad703c8f7 : SUCCESS in 13m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/da64b20e564f407caed02ef5d0e81996 : SUCCESS in 12m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7a9c8a1aafe8479d87cfdea3a9425920 : SUCCESS in 14m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/359071e445cf403281c01ac9219c3d8a : SUCCESS in 13m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1fc5fc3934ac4da5b4e601ffbea5e078 : SUCCESS in 8m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5487093020c4efba2d538fcf8818f0f : SUCCESS in 44m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b54eabb1026c4145af230057946ffdf9 : SUCCESS in 48m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4cc088d8cc7348c192e187da4e007e78 : SUCCESS in 31m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/419dfab1d3a74fa3a1cd0e0165328081 : SUCCESS in 18m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e83e20284a464f399e57e994ff85f53e : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/abaab1907318486ab752fbe1730ec8f8 : SUCCESS in 17m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f7675c9a6b84a4492fdfaf62b1ae626 : SUCCESS in 37m 19s","accounts_in_message":[],"_revision_number":2},{"id":"1498525a1fa084c6465da81b82e45fe9d915c9ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 01:13:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e772ff05ad04c9b9afeb498f8d823d9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/05f16e56c5e04c6980102f3a673937e8 : SUCCESS in 1h 13m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5d5c4335210b277244018c41eafb5f7cdfda1767","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-13 05:59:59.000000000","message":"Patch Set 2: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1ae429377872651d8d38e60c8921ec69336b07c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-13 15:45:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3d685051ef7ea28cb8cd8f3fcda345b347497505","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-13 15:45:43.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"83fe77c83323b2676e6d40daa3148b0a143f413a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 17:00:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/482ac159d5c344aaa627740bd0262c44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/174e90594e04414c8c1a9fb73addfaea : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/38ff14ea3bd541498aa146b1316f4c51 : SUCCESS in 57m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59a49de150614f47bb2c289e4a0ef998 : SUCCESS in 1h 03m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f548b53e8c44ed391e42b5f657fb48f : SUCCESS in 50m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d7bd75600ae44bc837eb7af2697cc08 : SUCCESS in 1h 03m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f88f7d050708443fbdd168febe3dbe66 : SUCCESS in 23m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d992183d0c4e4d0e885db68b03392e6a : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc1a403e8d6e49a8ad74ce0ccd8aa9ab : SUCCESS in 22m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/46c0319940504a4cb9656a8715431c04 : SUCCESS in 8m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a46d6db3e8b44cb688157b911d650469 : SUCCESS in 22m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a730bb777ffe4461b81f24c2b7d3cb97 : SUCCESS in 16m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/083d5f73439e4b2a85f00c4714b0c3dc : SUCCESS in 18m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7699ed88e14d47f58b1d681c27a537e4 : SUCCESS in 12m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/03392fa6ed574f7f904ad1ccfa59e63c : SUCCESS in 20m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ccd2cb0f55a94c4190cacda03e8c91ea : SUCCESS in 18m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9eb6c98e7bb04dca92946fd4fa131dd4 : SUCCESS in 13m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5f1d1795cfa749a5bc0fe0379f92974c : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/63ca673e9d994be683e48cecd9974c14 : SUCCESS in 8m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/057ff1acb7274c539cda14bf29e34a9b : SUCCESS in 19m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a3764d8ed3fd480e8334b8f94f941309 : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c0e9979e038c41e69a6b33b6ec55d956 : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7cbd3fa93add46dfad3b20262f7ee36c : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9da020d1d6a34a75a12825d40a4b0f5f : SUCCESS in 53m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d561caec20e8424fba872b6229aaca14 : SUCCESS in 35m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3da88493f6c54153946c90c912105319 : SUCCESS in 44m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55ed2366162547ffa70d45d47815a6ef : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab91df014dfe48b5aa675d93f4426034 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/25aec9481dbc4ff99319791a43e01702 : SUCCESS in 19m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61072f38dc6647caa2dd8208b8e46ef2 : SUCCESS in 42m 38s","accounts_in_message":[],"_revision_number":3},{"id":"de1a4f90e45ffa8ef677c49739e41b03257f1947","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 17:23:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6931618fe004afdaf06b2892d24b877\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f472926a961546ed8ff110d4c5eee146 : SUCCESS in 1h 23m 35s","accounts_in_message":[],"_revision_number":3},{"id":"fffedc47f087e500e4c0743b3a3b3d08fde0b00e","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-13 22:48:21.000000000","message":"Patch Set 3: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e06317afac01040112490e22702f7d2809fb7452","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-19 00:42:35.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_1179\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"_revision_number":4},{"id":"a3c78e180bf08f3cd5e9c0e097a14117d0d52175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 01:49:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57576f0a3ac84a5993d7e3b24d428280\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0aedf3479a49969a536b43e2310689 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/e847d6dc41f1402886515c1f7d281a45 : SUCCESS in 1h 00m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c462c95a5a8483b86513e87eda98c46 : SUCCESS in 1h 01m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/211dfd81675b49d8a43be0236e3646d0 : SUCCESS in 51m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcf29c9795244490bc9edcaa0e40884c : SUCCESS in 59m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dac2f124af93458b9aee151569c28ff0 : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32297491908b4798a3e5bb4b72fc6712 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88c47482556a4a2386db8c76a54a1d0d : SUCCESS in 21m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/31ce72d3dd044acfac89d8b2a0d2b3ae : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5f276057d3e743cbab8b3de7126b8cea : SUCCESS in 20m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bffc80f321e749198c0214ab37b61ff6 : SUCCESS in 18m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1f8f71a950a04d01832eebc091a6d4f6 : SUCCESS in 20m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/aec7944aca4f450d9fbcac11a2ad091f : SUCCESS in 21m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/940c9cdd42d349849ad855f516e68468 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d454a04ee8254f4cad4b8cb845c6f1d7 : SUCCESS in 17m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/73ae923b57194fe197a170482d9188d4 : SUCCESS in 17m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0f6f341cc55a4148b200728c10f2b638 : SUCCESS in 14m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cb4da2bb39f5463fb101683304bcd183 : SUCCESS in 15m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ddc18ce7bb0841fa92a52ef287663eda : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aed9482b2c2348e499295f6847cd35f4 : SUCCESS in 7m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab5eb7880010464580017083e499cd6a : SUCCESS in 12m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d422935c0270484ea61ef76b49cc4ea7 : SUCCESS in 6m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bcf6792bb9b148ceab2b9083d4b8942f : SUCCESS in 54m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a76ad3a8bddf4895acbc7ab5b00e0145 : SUCCESS in 38m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c3a5bf934a64f788adad8c85fc68fad : SUCCESS in 31m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab9c67b898784f08aaf9f079d0d7cdce : SUCCESS in 17m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9369742d06a94658889033e289cd677b : SUCCESS in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/38e5c88d70b441d7a68ae7460ccc303a : SUCCESS in 14m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/561c6ff86db74143b653e85e9812f35a : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":4},{"id":"c90d3f47cf0f0ac41f45b92799406cd71a4323e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 02:49:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8557721d865b491bbea826673a767399\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e278d0cdc274578b201682eb6535b73 : SUCCESS in 2h 04m 55s","accounts_in_message":[],"_revision_number":4},{"id":"fbb181f051909135c98a52c837c1c468bfa5ad6f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-04 17:29:04.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nThe following files contain Git conflicts:\n* test/unit/obj/test_diskfile.py\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"e420f6f5ad088c00cfae23303d624d9c93c3ea96","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-04 17:31:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"399a4586fa5f2a15e268b23226b4e8cadf3fc84d","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-04 17:33:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"608b1b1bfda6f0a1f24f763caab4c591938ea76c","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-04 17:33:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f2f2ff896ca5ac3f16f8ebc76fa792ed343d5acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 19:02:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc43426adbe542d4bb06185d10728f00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971946c84b4748a98913d7ae2f4bb976 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/1dd39d536f3f4d4388b14a1fb5346a91 : SUCCESS in 1h 00m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2b9e85ade32d42eeb842262ab89092c9 : SUCCESS in 48m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b796e3ef34c246749c37ff7500ff365c : SUCCESS in 37m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee738c65a87e45a8b06e7620c223f222 : SUCCESS in 1h 05m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/452ac6401a064555bc1d1d27f4a4fc79 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/527056fe978c483ca4aaaa0bd4bb2d24 : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9ba0b930b0f4640865912f5b2908af2 : SUCCESS in 21m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a31f6c6a6dc74f7fb8eb11def9f538b0 : FAILURE in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/61f4b597308744198363e4d37d27b493 : SUCCESS in 21m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/534141bea7ad47078177fb4ccecc945e : SUCCESS in 17m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2bc92ee4943c433fa28bb6a78bc892e1 : SUCCESS in 19m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7ea998f65c074ed19756cdfc68ceacff : SUCCESS in 25m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/218d796463864899b5bdb7aef648906a : SUCCESS in 20m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5c8ad80168ce42dbb9843853508f4225 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/35c5b87218594b9ead6282e52f1ba3eb : SUCCESS in 22m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/07dbf68459954a4c9275f80bc2ff49b4 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8d948e4462164f19852949825883d8ac : SUCCESS in 17m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7774b63114584d2482fa89b10efebbda : SUCCESS in 19m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/22a869e8915449b88cb6bb077ffd0d12 : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2459aee8e5ad41bbafa8c01b9728950f : SUCCESS in 15m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4a5f553ae7594d238567a58436c4b5af : SUCCESS in 8m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ef4c5e7aa1543f59e9feb3413499ba8 : SUCCESS in 56m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1deaa64c61884dd2a6b086282c8f5e09 : SUCCESS in 54m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8a82ef7ba63a49c69c5ade587a913785 : SUCCESS in 57m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b15d52f4799c4268806dda2d2cf6b052 : SUCCESS in 22m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcba61df6743448c9441b14db67b7894 : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13b9ef36ae5d4ced8a20448d6f148d88 : SUCCESS in 18m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4842be88b6594b10b3f43ccdfa7de427 : SUCCESS in 51m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/087f2ef461254884936db066f9092259 : SUCCESS in 1h 06m 24s","accounts_in_message":[],"_revision_number":6},{"id":"59f4b2d054a5f6fef1fa8c4f4e74d803a142e125","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 19:02:52.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9727ab497094fbba21d730b9979462a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9771176b93834b7fa2e541d1c7b8559b : NODE_FAILURE Node(set) request 300-0025849970 failed in 0s","accounts_in_message":[],"_revision_number":6},{"id":"4bef29fc72dc2f52f314ab048404eea699ce2159","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-06 17:05:27.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"fc1655734d6b7ca9e12e022ff511b1ae786f0bff","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-06 17:06:09.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"ce3ec180559f96ada39202f8ef1b46a52d558f71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 18:08:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b592fda5fb944dc857f434b30c79c81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1859e9b2aaf64254b30b9b997fb1f7c1 : SUCCESS in 12m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/a81e1335e79b4801bbcb6bef5ecab303 : SUCCESS in 56m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/adf1a00c98034bb68e03aa82f7d2af9f : SUCCESS in 57m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0239cb520d1341a596ee322795a80899 : SUCCESS in 52m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d423481f39a41369532bfdd0ec1465c : SUCCESS in 53m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/333b9c8fd71440488337f6dddd89cf46 : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b94e0f7e8170480ca53c9444bae67cf2 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e284116d2f348c29f8a931e752ea438 : SUCCESS in 16m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/58108d419d0641e88ee807a27b3526b2 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f1e00e3aef3429abe917e0ff2740eb9 : SUCCESS in 20m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/860cee67ce0541e5aa9330b6517ee6b7 : SUCCESS in 18m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0faf683506c04649bcee7ae1ad54c4e0 : SUCCESS in 16m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/10e83207436b45f39b9b6588c5f135c4 : SUCCESS in 20m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8972c756b70c4adb840b3aa6363135ee : SUCCESS in 12m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8f968a0d2dcb45e4bee4d340d57236ce : SUCCESS in 15m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1986adbd7d444ec3abf1a244e3970892 : SUCCESS in 20m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d2f72524de0a47bd88ea0690626f8248 : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/384862e9a7474cb9b50d4ceb91be8f3e : SUCCESS in 14m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/beaac3fbf7114c5b833a27dd3a97e68b : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/02e60a5c7fb94fec8d003c82f36eefe4 : SUCCESS in 4m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1638bb0cad984e95a04a1263a91a9859 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aca0338f72064c608330f27506dd0c57 : SUCCESS in 8m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8368eda02c3488db6d4915372e4554c : SUCCESS in 55m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e7a9c728b4314334b019cea05ae25d64 : SUCCESS in 49m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/294dc60284fe4da1ad3dc5bfbe0ff62b : SUCCESS in 54m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96f0840a31a14deaa1b2444912cc8db7 : SUCCESS in 19m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e38f45c147545bca963ba16c98baea8 : SUCCESS in 7m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9294342ad77c49b99de24132e0757966 : SUCCESS in 26m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34cf73a1ad6a4c9c87bb4302bbab7f71 : FAILURE in 34m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/165bf4841c08406085fbf335b7aa0278 : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":7},{"id":"ea0a415ef5919ffff0d3c41202b3fdc68a93188d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 18:50:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ce1795b530648e79ae8e0f93cae89e0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2cf852a2404a40e081a9e1b302f27242 : SUCCESS in 1h 19m 26s","accounts_in_message":[],"_revision_number":7},{"id":"65fdf924f272f893f9e821fd0ad8a714d71aa652","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-09 17:11:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bc60475dd199bbc1e141fc5cba83e89bacc04ad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 18:32:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec5225dd10c3465c9a43901d9c22cda5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06dd06a9a8354f00b1771aef837ee02d : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/021d9d33fe1341019ccf25926b62ba0d : SUCCESS in 1h 01m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8b5f836dff64069abedb881f5549b7a : SUCCESS in 1h 02m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da51308d5f1d43b3944d7c4d5eacc551 : SUCCESS in 41m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96a17afede9b4c3a9e51eae0060f30f8 : SUCCESS in 1h 02m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c629a79b23b94a17bead0add04cd3b77 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b642650c3cb64a3c8882f719e2b6d908 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/406b653a69b74ea89951c86151137001 : SUCCESS in 17m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4674337ed6d34d49ba26c1bba1b31062 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81cb283ba8c749ce9c261330aa68709e : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/594d61b56c9f4fe8a00c0567fa442897 : SUCCESS in 16m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59b64a3e38a246b18159aef7508197c6 : SUCCESS in 21m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e73d2b8263ea4ec5830adb442ec51cce : SUCCESS in 35m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/45d77bbbc2ee4c45930d169f86c638a8 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ea4f65b1d2748619a016e26448f4750 : SUCCESS in 15m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d6a9a6b7361242828165be5d37fb2820 : SUCCESS in 18m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/de80122818674b60bd83bfcc0840862d : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e901c73ab0634a0396107560c2451b9a : SUCCESS in 13m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9fe987de7d4b4deabe997c2dff5fcea5 : SUCCESS in 19m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2b8e00791b11494d8ec872fd0526c219 : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc89027356234a1f8e1bb3ba496af262 : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/42ff8dbc23974e35be2d98a083aa5622 : SUCCESS in 7m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f24205ee6a34c83af588f5b7b22a8f9 : SUCCESS in 1h 00m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ff39b28631b41939e475db2dfddecfb : SUCCESS in 46m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e924b285d9d94434b7764deaba817d51 : SUCCESS in 1h 05m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1d0ecfbdf2d417a908fdba99319ba05 : SUCCESS in 18m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0710a0b4e64466fa0366fbb4c4291a9 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e911061856184bbfb36ced047b674ae4 : SUCCESS in 11m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4877e2f33d6f4057ac5f824ac199e510 : SUCCESS in 41m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8c715df1c994da09c645b63c7c45421 : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":7},{"id":"d6601ee213e2591067261cbe3c15039ca56b1793","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 19:17:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/135a82ad9d2a44bc83f5904e8145fa32\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/af6a8de724b146beb12f05c834e1284f : SUCCESS in 1h 48m 31s","accounts_in_message":[],"_revision_number":7},{"id":"32b52a9c1fc7d5de6aa45fca1791a7164d441789","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-12-23 23:09:32.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"222298977609d289a7d9efccca21550847a46d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 00:08:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/255eadb72ed04194b387e6a06c961c50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84a03378096d43fbba23abae4082957b : SUCCESS in 7m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/2f1462f53f3a406d97cd4e3b84bdbbad : SUCCESS in 39m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0540cd8daef045458baf74b7299aad03 : SUCCESS in 41m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/95483aebe2154790876be8b627faaeba : SUCCESS in 37m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc451e4df06a47b89f4f235f7efd0e30 : SUCCESS in 45m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ff02d8cfd264bb4b93923afa9f7b20a : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9407865a45884ece8c101c81986f275c : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59af769e5db9423aa65a76b9745f9192 : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fef4a5d0d90648bea1d1ee93ecb7def8 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/777a5422002648358f4ada0f672779ad : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/74bbb2b1bd9c43d7aba33c50f12c5e7e : SUCCESS in 17m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b2f06e7bace74e639fc7f83dab8d60f6 : SUCCESS in 17m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0aaf1276ffc3416a85686fd0b0522f05 : SUCCESS in 18m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3d17528aa4564fc5b1531bae756abca9 : SUCCESS in 10m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5c1b2b3d548a40388829ed276ab5bf4e : SUCCESS in 16m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8693cedf8ae1462d97e5f833bade4157 : SUCCESS in 13m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4d7e39c87c6d479fb294fc0e7dd38da9 : SUCCESS in 9m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/614c87a252974d5ea72f3b08f39ea01c : SUCCESS in 9m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3d1b818ed67e4f7cb24420c7b441436d : SUCCESS in 10m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c975cc99abb04c8ea95f9e2ab800fe91 : SUCCESS in 5m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84c1ae15698f4c6d85c12a1709bab309 : SUCCESS in 11m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2173b3688c92421fbbe4850cbc5eefeb : SUCCESS in 4m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5b153adef5744e3a57e212463f0dc5d : SUCCESS in 40m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e6d5e19e9a404d84b916c2c1b07a6fc9 : SUCCESS in 25m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/21146249b47545d49b5b42086ba3e71d : SUCCESS in 28m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/454d018e83434445b5526a2173598a38 : SUCCESS in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a82166e87474603b7ab48af6d390a1b : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71de3a636e49406c8b3d476df66e9b1e : SUCCESS in 17m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d009ed57da34596a993a3871fb7bfda : SUCCESS in 51m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5dcdd3b38e6942ee8b8d29a19c375c2a : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":8},{"id":"fd3fa7f66cf0461f1d137dc61b0e3451838b1443","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 00:42:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6ea7e303bb7422b966488733fb75251\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f5b291c2871417db483177afc221668 : SUCCESS in 1h 31m 26s","accounts_in_message":[],"_revision_number":8},{"id":"ab1fa93f3587490a7654708633820713f8982d4b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-16 12:21:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"289ebad60e4abef8c9c0bd34f5338d513698f8ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:39:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a3b6b93a0114d34b73f3cd0eca6dabd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c86f0d47ad724b478c36bed4176d0803 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/8235e8ce03eb4886adc3972a6b49c7f3 : SUCCESS in 57m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b00f1dfdfd3643e8a04924a0430ca95f : SUCCESS in 1h 07m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ca0fb9386a834dc0af896f7b18f9b6dd : SUCCESS in 58m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d2032fb38b34c92a44cf8da9a2aceb2 : SUCCESS in 1h 11m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58d407f43f2f4938939e77fe9743b7ba : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f166c169ef944529e7d794e1c962b04 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1bab8bc279f4f049229bf9836a3e06e : SUCCESS in 12m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9633f98e755c4cf2a939dfb092cf806f : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de871ef9be1b4e09bb546b2a997a4973 : SUCCESS in 17m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/96890e21d5004b0584117a7f8f53ab99 : SUCCESS in 18m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eab1e27f0b55427cbf2de91dfffba2cb : SUCCESS in 25m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c2637718c45e4158baf836d96b7a19d4 : SUCCESS in 13m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bbff06af32564c018b157ce1b3c81625 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/12b0d89e1cf5448a8f347e541521213e : SUCCESS in 14m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fd68a136cbf0475fbcf4493cb3afe400 : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a66bfeac492143e78f43206d83762e5b : FAILURE in 19m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bfd2337cd0004f35a5c03df42787dbc3 : SUCCESS in 6m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2f2d62292f4418a9e8ca43d7d621868 : SUCCESS in 1h 00m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/646b081b6ddb49e8a88592e895674c68 : SUCCESS in 46m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa217d2ee1ee4ecfa5c3b7b7219a8395 : SUCCESS in 48m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a3fbaad47f848f397883656b37c2c07 : SUCCESS in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e5a4f2b2334cfa8f488d60c9c5cce6 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ca63939193741518b53c821f01ce660 : SUCCESS in 17m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dd9dcb223b94657919ad198a3743a15 : SUCCESS in 54m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5895364b52df4c1b8d6b5ab3b1dc836a : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":8},{"id":"50a9c578369ed21585b12fb392e2942e7db4b248","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:44:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cdbe0a8e1ff4352bfdd50d5c8da0cc2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b0998a74e69b4d0496f7884344808553 : SUCCESS in 1h 43m 34s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"33694599a30c0d51e6c657446b1ab24cdfc1c0cd","revisions":{"91b4e877c6005173b3a2376d35af1bbc7dbf4067":{"kind":"REWORK","_number":1,"created":"2024-11-11 16:46:38.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/75/934675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/934675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/934675/1"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-11 16:46:23.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-11 16:46:23.000000000","tz":-360},"subject":"add backoff sleep to REPLICATE","message":"add backoff sleep to REPLICATE\n\nChange-Id: Ib7b236e50ac844f4e001034abc1d090ca97b2eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91b4e877c6005173b3a2376d35af1bbc7dbf4067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91b4e877c6005173b3a2376d35af1bbc7dbf4067"}]},"branch":"refs/heads/master"},"a8922ac52a96123fb8e8144e3aef6cf16620aa16":{"kind":"REWORK","_number":2,"created":"2024-11-12 23:57:44.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/75/934675/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/934675/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/934675/2"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-11 16:46:23.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-12 23:54:55.000000000","tz":-360},"subject":"add backoff sleep to REPLICATE","message":"add backoff sleep to REPLICATE\n\nWe\u0027ve noticed latency spikes in the object-server while REPLICATE calls\nare in-flight with some of the object-server requests failing with a\nPartitionLockTimeout.\n\nThis change attempts to limit the possibility that any individual\nREPLICATE calls is unfairly holding the GIL except while also holding\nthe partition lock in an attempt to smooth at the observed latency\nspikes by preferring 500 in the backend service call over unlimited\nrecursion.\n\nChange-Id: Ib7b236e50ac844f4e001034abc1d090ca97b2eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8922ac52a96123fb8e8144e3aef6cf16620aa16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8922ac52a96123fb8e8144e3aef6cf16620aa16"}]},"branch":"refs/heads/master"},"66eff91d3b7fbd32826a7414f793dacbe9ea1da8":{"kind":"REWORK","_number":3,"created":"2024-11-13 15:45:18.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/75/934675/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/934675/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/934675/3"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-11 16:46:23.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-13 15:44:54.000000000","tz":-360},"subject":"add backoff sleep to REPLICATE","message":"add backoff sleep to REPLICATE\n\nWe\u0027ve noticed latency spikes in the object-server while REPLICATE calls\nare in-flight with some of the object-server requests failing with a\nPartitionLockTimeout.\n\nThis change attempts to limit the possibility that any individual\nREPLICATE calls is unfairly holding the GIL except while also holding\nthe partition lock in an attempt to smooth at the observed latency\nspikes by preferring 500 in the backend service call over unlimited\nrecursion.\n\nChange-Id: Ib7b236e50ac844f4e001034abc1d090ca97b2eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66eff91d3b7fbd32826a7414f793dacbe9ea1da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66eff91d3b7fbd32826a7414f793dacbe9ea1da8"}]},"branch":"refs/heads/master"},"12b12f67874e358a7426ea2a91c2a14ae889b553":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-19 00:42:35.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/75/934675/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/934675/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/934675/4"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-11 16:46:23.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-19 00:42:35.000000000","tz":0},"subject":"add backoff sleep to REPLICATE","message":"add backoff sleep to REPLICATE\n\nWe\u0027ve noticed latency spikes in the object-server while REPLICATE calls\nare in-flight with some of the object-server requests failing with a\nPartitionLockTimeout.\n\nThis change attempts to limit the possibility that any individual\nREPLICATE calls is unfairly holding the GIL except while also holding\nthe partition lock in an attempt to smooth at the observed latency\nspikes by preferring 500 in the backend service call over unlimited\nrecursion.\n\nChange-Id: Ib7b236e50ac844f4e001034abc1d090ca97b2eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12b12f67874e358a7426ea2a91c2a14ae889b553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12b12f67874e358a7426ea2a91c2a14ae889b553"}]},"branch":"refs/heads/master","description":"Rebase"},"45dfc9accac8ba0fabc32b71e5a2abb810cc2f89":{"kind":"REWORK","_number":5,"created":"2024-12-04 17:29:04.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/75/934675/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/934675/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/934675/5"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-11 16:46:23.000000000","tz":-360},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-12-04 17:29:04.000000000","tz":0},"subject":"add backoff sleep to REPLICATE","message":"add backoff sleep to REPLICATE\n\nWe\u0027ve noticed latency spikes in the object-server while REPLICATE calls\nare in-flight with some of the object-server requests failing with a\nPartitionLockTimeout.\n\nThis change attempts to limit the possibility that any individual\nREPLICATE calls is unfairly holding the GIL except while also holding\nthe partition lock in an attempt to smooth at the observed latency\nspikes by preferring 500 in the backend service call over unlimited\nrecursion.\n\nChange-Id: Ib7b236e50ac844f4e001034abc1d090ca97b2eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45dfc9accac8ba0fabc32b71e5a2abb810cc2f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45dfc9accac8ba0fabc32b71e5a2abb810cc2f89"}]},"branch":"refs/heads/master","description":"Rebase"},"b5abf3aeb5cf5ffc1f7184032d23d17c767a3d98":{"kind":"REWORK","_number":6,"created":"2024-12-04 17:31:46.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/75/934675/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/934675/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/934675/6"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-11 16:46:23.000000000","tz":-360},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-04 17:31:36.000000000","tz":-480},"subject":"add backoff sleep to REPLICATE","message":"add backoff sleep to REPLICATE\n\nWe\u0027ve noticed latency spikes in the object-server while REPLICATE calls\nare in-flight with some of the object-server requests failing with a\nPartitionLockTimeout.\n\nThis change attempts to limit the possibility that any individual\nREPLICATE calls is unfairly holding the GIL except while also holding\nthe partition lock in an attempt to smooth at the observed latency\nspikes by preferring 500 in the backend service call over unlimited\nrecursion.\n\nChange-Id: Ib7b236e50ac844f4e001034abc1d090ca97b2eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5abf3aeb5cf5ffc1f7184032d23d17c767a3d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5abf3aeb5cf5ffc1f7184032d23d17c767a3d98"}]},"branch":"refs/heads/master"},"ddbc202e3adcf4feaa28d51a63b455e1aff6895b":{"kind":"REWORK","_number":7,"created":"2024-12-06 17:05:27.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/75/934675/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/934675/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/934675/7"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-11 16:46:23.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-12-06 17:00:46.000000000","tz":-360},"subject":"add backoff sleep to REPLICATE","message":"add backoff sleep to REPLICATE\n\nWe\u0027ve noticed latency spikes in the object-server while REPLICATE calls\nare in-flight with some of the object-server requests failing with a\nPartitionLockTimeout.\n\nThis change attempts to limit the possibility that any individual\nREPLICATE calls is unfairly holding the GIL except while also holding\nthe partition lock in an attempt to smooth at the observed latency\nspikes by preferring 500 in the backend service call over unlimited\nrecursion.\n\nChange-Id: Ib7b236e50ac844f4e001034abc1d090ca97b2eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddbc202e3adcf4feaa28d51a63b455e1aff6895b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddbc202e3adcf4feaa28d51a63b455e1aff6895b"}]},"branch":"refs/heads/master"},"33694599a30c0d51e6c657446b1ab24cdfc1c0cd":{"kind":"REWORK","_number":8,"created":"2024-12-23 23:09:32.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/75/934675/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/934675/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/934675/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/934675/8"}}},"commit":{"parents":[{"commit":"cb745837536196371bd6d0465d1dd03538d3a6b7","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb745837536196371bd6d0465d1dd03538d3a6b7"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-11 16:46:23.000000000","tz":-360},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-12-23 23:08:48.000000000","tz":-480},"subject":"add backoff sleep to REPLICATE","message":"add backoff sleep to REPLICATE\n\nWe\u0027ve noticed latency spikes in the object-server while REPLICATE calls\nare in-flight with some of the object-server requests failing with a\nPartitionLockTimeout.\n\nThis change attempts to limit the possibility that any individual\nREPLICATE calls is unfairly holding the GIL except while also holding\nthe partition lock in an attempt to smooth at the observed latency\nspikes by preferring 500 in the backend service call over unlimited\nrecursion.\n\nChange-Id: Ib7b236e50ac844f4e001034abc1d090ca97b2eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33694599a30c0d51e6c657446b1ab24cdfc1c0cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33694599a30c0d51e6c657446b1ab24cdfc1c0cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
