)]}'
{"id":"openstack%2Fswift~934737","triplet_id":"openstack%2Fswift~feature%2Fmpu~I0c05f3c14bab7bc7b04f4293e7ba2178af5ba6bf","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-sess-refactor","hashtags":[],"change_id":"I0c05f3c14bab7bc7b04f4293e7ba2178af5ba6bf","subject":"mpu: refactor MPUSession class","status":"MERGED","created":"2024-11-12 11:12:46.000000000","updated":"2024-11-12 13:12:36.000000000","submitted":"2024-11-12 13:11:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934737-p-mpu-sess-refactor","meta_rev_id":"4d93c0363e445dc5a73d10b96146374e0011e881","_number":934737,"virtual_id_number":934737,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-12 13:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-12 12:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-12 12:19:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-12 12:16:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c56ba73561e8ae04390549e00a40f6124f909443","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-12 11:12:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3a72e361b75ea969d7c8c56e9cd4e91d1dc483c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 12:16:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e05175e216b342f79f20ccdb5e7daec2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8f13f7a4644d508b73a1cfc5fa462a : SUCCESS in 8m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e910fc9de550479d8d2855a7cca850bb : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1844109b2434761ae96756916618bc8 : SUCCESS in 22m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f33190d41b6f43c0a4d3febc6477bce2 : SUCCESS in 18m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c97bcd049d024aba875c1dbc52f2c6b1 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4df28f1f09e948fe85015ef1ad828df8 : SUCCESS in 18m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f6fba421f9b94893b733c57377754a35 : SUCCESS in 18m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7bdc8201601147258785d0836a916c24 : SUCCESS in 21m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6cf7b83eaeb5446f82fbdb6a7e6eb57a : SUCCESS in 22m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06c75a10cbe4436d8551c64bb419fc6e : SUCCESS in 19m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/320e881caa614871bd04cb96498cbfb3 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/01b9c07d0c9544489652534f009c0385 : SUCCESS in 18m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e904c8ac222c4e11aa4a23d77ab3847a : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/73b0c6cb0a4644239b1000e0e7fc4d39 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/974e39e6ee0149eeb1664c135f795072 : SUCCESS in 17m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3853aade6ec74dd4b58d266cc72a546b : SUCCESS in 8m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d7f25cc161e54c77b6d3e66e2828d15d : SUCCESS in 12m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fb6244ed5b144c0488cda039c5020076 : SUCCESS in 6m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe70ea5e7d0545a59407c44e3434ec98 : SUCCESS in 51m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/437125868449495a9be759b26057efd8 : SUCCESS in 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/247679c2a3d24b1d990b3a502d2d2143 : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c147910b69ba43c9bf2e12c6233d6ba4 : SUCCESS in 16m 37s","accounts_in_message":[],"_revision_number":1},{"id":"01dfea99187e19438933035385cfeaa8c16f948e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-12 12:19:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"572fba0d853dac4a73f7dfd62727e0a05b44a0a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 12:19:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50a9394dacbb4abce8b60c2dc025468ab8207ac1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 12:53:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f0ec7688ebc4e0ebfb5254056c7c01b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ebc9922db34348d69b0fe925ced96db3 : SUCCESS in 1h 38m 36s","accounts_in_message":[],"_revision_number":1},{"id":"31b6b5e1d799da6f4d41d4c973ba6a83a03a3720","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 13:11:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/582b59098620429984388da73649fd6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace33d2c044b45058f064ffe0329ef3b : SUCCESS in 8m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/37f73bfdec5e43b2b4f19ff0bd72cbc4 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cfe6452914074a6c94919f7d725b524d : SUCCESS in 19m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27a4e1db9ed9412b937039ce59e5596f : SUCCESS in 17m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/711831985ced4dddb4df8a5b90b2ea8c : SUCCESS in 20m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7634f752bf6a4d9783f1a4672cb5836a : SUCCESS in 20m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/87cc769ac0f94d06a283bd7c47776b59 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c69e7ae5e64b43df8b92b60362f9e7ad : SUCCESS in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8e8fcc086f4f430baa7e2b950f2fa5e9 : SUCCESS in 19m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/47cecf3bce1a4f549947eafa0e95eec0 : SUCCESS in 11m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d40b00a53fc94910aaa02bd484630eea : SUCCESS in 15m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e79607ae50c54790823995dcbc9a359f : SUCCESS in 13m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6dcc301686a0446b86767982b95d9368 : SUCCESS in 9m 22s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c2ae1a08d1e44665bf1aea5e86129192 : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e64f537c0e74fff9602ce48a449e65c : SUCCESS in 47m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f3a363ee2b248628fa08ee425ebd92c : SUCCESS in 18m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f1634367fe4e4fa7299f4381f38654 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e2f262ae4fd9412a96eec87cda28f33e : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":1},{"id":"46a3a1f2ce263409f98f16a770c5a417d9e40595","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 13:11:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4d93c0363e445dc5a73d10b96146374e0011e881","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 13:12:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/174a0a342bed46328fce511ebd5bb2c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3752e1395411492799b7da1aed3ab6db : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/43138b36d8e44f1b8b3ca9acb959859e : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67859418b14a26d54d7178e0ad78e857a4c145a6","revisions":{"67859418b14a26d54d7178e0ad78e857a4c145a6":{"kind":"REWORK","_number":1,"created":"2024-11-12 11:12:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/934737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/934737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/934737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/934737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/934737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/934737/1"}}},"commit":{"parents":[{"commit":"50dfff7db1251aff0b549449b4e41633bacff4b7","subject":"mpu: fix param passing to sub requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50dfff7db1251aff0b549449b4e41633bacff4b7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-12 11:10:01.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-12 11:12:14.000000000","tz":0},"subject":"mpu: refactor MPUSession class","message":"mpu: refactor MPUSession class\n\nInherit from MPUItem.\n\nChange-Id: I0c05f3c14bab7bc7b04f4293e7ba2178af5ba6bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67859418b14a26d54d7178e0ad78e857a4c145a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67859418b14a26d54d7178e0ad78e857a4c145a6"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
