)]}'
{"id":"openstack%2Fswift~934910","triplet_id":"openstack%2Fswift~master~I4f7dd0755b8dc8806d9b9046ac192d94ca705383","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-13 16:21:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f7dd0755b8dc8806d9b9046ac192d94ca705383","subject":"probe tests: Run relinker via subprocess","status":"MERGED","created":"2024-11-12 23:39:49.000000000","updated":"2024-11-13 16:22:16.000000000","submitted":"2024-11-13 16:21:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"934910","meta_rev_id":"96eaaa7b9984ab4d3fd600c293107a65bd6a24e9","_number":934910,"virtual_id_number":934910,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-13 16:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-11-13 15:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-11-13 15:10:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-13 00:35:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-13 15:10:26.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"1646b069501faf464f6715312a94fb1b34eed42e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-12 23:39:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e0d2758f5fc4176fce754220078967024ce132c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 00:35:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e623fda006e241bb9a09d4ba5319fecf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efda538803b94c0abb8be0dfcd570b2a : SUCCESS in 9m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/869d105a47bf4be69866ab70a8d793ae : SUCCESS in 53m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/11c673788ddd4752a4b7d7f24604324f : SUCCESS in 45m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1e1a4d158ec453db025dd5f681ba93a : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f8279ba72c1101663f5f369de3f247efc1808a96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 00:41:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6fb2598547f43469d38be00c0553959\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/252f4aa845434b76969e84657ec622e6 : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2cfca25e7bdcddaeb1d418602b46d91a29be6bd2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-13 15:10:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06513b39cb4380cb91b7a9299165efc50c6bd4ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 15:11:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1324a81baf9ca5e34f01b4d67ccfb11141990ef3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 16:21:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c3d39a790ef43e78c48d3a25ab6f78d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/423ac22c52ab43a8b5392e225ef5b52d : SUCCESS in 8m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e30ea6e2d9845e58f3348cea76bcef8 : SUCCESS in 1h 04m 23s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5d2af947c7d147e1ad38f1f1b5b701cf : SUCCESS in 10m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/53406bc5da9d4cbaaab35702668f3719 : SUCCESS in 19m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f20195754897443baa9b02fd7b426d9f : SUCCESS in 20m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1e9bfca165d84772a2dfe442a3e7f139 : SUCCESS in 23m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b685396d28884fcaa8e106fcef9ec0b5 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c9ff2f1b7d6043d294f1a5850f89e7f7 : SUCCESS in 21m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c59f12054473422e80ce59b82847eb73 : SUCCESS in 18m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8b85fc6c6d75461b922b02db199db748 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3f7f469414fb4e8aa61156430a875bd9 : SUCCESS in 18m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/edf2d026ae8b4a42be60fb35b248dca6 : SUCCESS in 14m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0497e3c6ef964d259cd9873e18454035 : SUCCESS in 6m 32s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/62b931790f5c446c9a4bff1f6217e64a : SUCCESS in 7m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/90c0ce6cb5124c07b96e24856772e790 : SUCCESS in 55m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30443793b6de4fc799d8f97735a73046 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6eead74d13950ac9aa222ff36f9c7ae1a53a5374","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 16:21:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"96eaaa7b9984ab4d3fd600c293107a65bd6a24e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 16:22:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdc4fa66f515481e92af9451ba036eb8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d2904a10ff748819be4bab8307049ed : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1fb7a5a3527e47ea845f5deb6d7b7356 : FAILURE in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b262b16d55059cd24d788fd1e3f4facde88271d1","revisions":{"b262b16d55059cd24d788fd1e3f4facde88271d1":{"kind":"REWORK","_number":1,"created":"2024-11-12 23:39:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/934910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/934910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/934910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/934910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/934910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/934910/1"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-12 23:39:15.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-12 23:39:19.000000000","tz":-480},"subject":"probe tests: Run relinker via subprocess","message":"probe tests: Run relinker via subprocess\n\nOtherwise, test-run state (eventlet.tpool initialization, in particular)\ncan bleed over and cause hangs.\n\nCloses-Bug: #2088026\nChange-Id: I4f7dd0755b8dc8806d9b9046ac192d94ca705383\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b262b16d55059cd24d788fd1e3f4facde88271d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b262b16d55059cd24d788fd1e3f4facde88271d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
