)]}'
{"id":"openstack%2Fswift~934943","triplet_id":"openstack%2Fswift~feature%2Fmpu~I2e14c623e864ddaf12751a855ce66b31eb7a2832","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-copy","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-21 10:13:14.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I2e14c623e864ddaf12751a855ce66b31eb7a2832","subject":"mpu: forward headers with uploadPart subrequest","status":"MERGED","created":"2024-11-13 10:37:57.000000000","updated":"2024-11-21 11:47:06.000000000","submitted":"2024-11-21 11:45:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934943-p-mpu-copy","meta_rev_id":"0d0307f78ff51bd27dfd0a34e8cb94fbdcf3177b","_number":934943,"virtual_id_number":934943,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-21 11:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-21 10:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-21 10:13:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-13 11:28:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0643c94a526eb38313c0df47ca6304217fd5e941","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-13 10:37:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02859d861b3e2563e06d6a4f075232bde42cea15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 11:28:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5c8cca6dc6240fba8ecf381670ecdf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9282ef689da4609b7be2f855288d79e : SUCCESS in 9m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ab7912ff95142c2b5b4141e0c73f6eb : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27b530d6a7db474e9fbafd753dee1763 : SUCCESS in 22m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01a4e514493e4913b4627d5d537d1d69 : SUCCESS in 19m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e3a081cd85e043ff959693f4f7d5dae4 : SUCCESS in 4m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/201aeb10276b48498d396172ece1cb35 : SUCCESS in 15m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/efe99bbfce214e9cbbe3a3d8e5d0471d : SUCCESS in 14m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/67e8d1b7086c408f8b49beca213238af : SUCCESS in 23m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bd011bc910464843afa86fff23b60012 : SUCCESS in 34m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/07c63cae207a412abda98958892108a7 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2d720982d76d41ea8937c99fd0f1c7a7 : SUCCESS in 19m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/36ae63be95da4dbea2e8a8b6a3459eaa : SUCCESS in 17m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f56f70075d2a4315a4d30ea663f6a4e1 : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e044f924c9244f359b8131be40840937 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a30c6660af1548eab8eff8ec5e722c6e : SUCCESS in 14m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/998058dd801d41e3bb7f99faf7ceeabf : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5c289b8603db4995a52824be8a606ce9 : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/84cbed7bfa0646e097f0a591b7279576 : SUCCESS in 7m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dae77f8f9f4c4ebd93e7e50be123c90c : SUCCESS in 43m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d10a88f23c0487e8325b05c423a90b4 : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4342399c5514234b9fa52585b37e08a : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39935707e5d04484b7ffa0e70311d3c5 : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"dc3b099d08e407ffb5ba6551ea96f090ef84f1df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 11:43:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d39252201fb3458a8b47649f3b86b684\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/86499e7c44a34e96bd4260e98964b489 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6f780c201894c36a29b1bd5f4b9c248c05e99511","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-15 17:16:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c37ed1cd265aed1bad1024120e56110c9755ad49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:13:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aad92392ab654ef39616999d66a8c264\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c9341f95ce45639af85dba2f27f5a0 : SUCCESS in 11m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9168f55e1ca4469caada9be9bef61483 : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3c6aa20231745fa9bd8eefb179b9e70 : SUCCESS in 22m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/857ed466b00e4cb69cc14e3a077240d0 : SUCCESS in 22m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e160f0b84b53445d80845254c3c0fc3b : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ba2edbd35f104986ad63008f6b782ac7 : SUCCESS in 23m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4e274907fb564193969ade4fcdf30ffe : SUCCESS in 21m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/20b6a25532894b2980b590d029b8efac : SUCCESS in 22m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f83e790f9b3843f98e5614572db43af9 : SUCCESS in 31m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2b2582957a68476d86ed5780016bf9ed : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/45c54923edb441ca92d3aa3810cce364 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bccd58f4aeef4428b2ceb21659d795bd : SUCCESS in 18m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ae219023bda84a0bb2316cc00e855361 : SUCCESS in 13m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0fdba25100af4d8da30b708d658a9b45 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5d3a5f59f3ee4c8c874f48b47a751ca1 : SUCCESS in 14m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/189e0fb95b784139a5e1dc243b3223f7 : SUCCESS in 8m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bbae3ffedd7a4fe0ac71fb963cf63113 : SUCCESS in 16m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ab239c8907054cc5bbb37d4bb25797c7 : SUCCESS in 7m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d53ad84dde034ab899bede40f0c040c4 : SUCCESS in 49m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8f6100041a546edaa5de03e11c19f36 : SUCCESS in 19m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cd4821723714ad39266ab15cf8cb69e : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb36fa42338048a2b94ad4db8d970b0e : FAILURE in 20m 57s","accounts_in_message":[],"_revision_number":2},{"id":"d372ac02642e473bb9ec6dddc980d1b369465df6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:56:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3103aa84dddf4a8490b3e553b44ae158\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/20710da0fb6b443cbee1a2816ca66854 : SUCCESS in 1h 37m 43s","accounts_in_message":[],"_revision_number":2},{"id":"0b2831dc97db36ac6df0d0b4a14dbb39f3c866c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-19 13:12:37.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"981b4c311e1b5609e05fc6ae19606680a3059bc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 14:34:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7ca21e96a574c25ba23850a4a0e76e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643deeeda103410db9a5312fa97f2d3a : SUCCESS in 10m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e317d7468f944f8b55aeade0d705ec2 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/acae3530f7c84a2184438add17a8ef2b : SUCCESS in 23m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e108c853027845109c19cbad8b1b1c75 : SUCCESS in 21m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7fe212cb50934ae38878af1ac85d334c : SUCCESS in 7m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/420d9881646d4275af71dbd72ed55e09 : SUCCESS in 22m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9b9e49d84e2a49bd9a20c40889c7ae0c : SUCCESS in 19m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ce5a4afe2820406c9b02c44192fd8af7 : SUCCESS in 25m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6da2ea86e579400f849f2df07a7317c1 : SUCCESS in 26m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fa560ce622454227a6a95b2f42f09c0c : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b90ee0ff28bc47d7bb0f6f32719aca28 : SUCCESS in 25m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7de563b5ce7f43539855b5978c923b69 : SUCCESS in 17m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/94627908f24a44399a5441b98caef995 : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/81fbfe36f13045fb83fac1880abc028b : SUCCESS in 14m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7d258756e02b4b0b9ae3797d0cf7ec36 : SUCCESS in 15m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/af70a522731741c4950767b02d03fc6a : FAILURE in 8m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/abe46d4c730f43968a3c2c9620578488 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a024324d5244b8a99b0aa5fb35ee899 : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/11c63bf9cccf4d0a9d893ef0dc8d3004 : SUCCESS in 49m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed7833ba60374611a3a90d7e5bc10de7 : SUCCESS in 20m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c0f7b115da44da2b65514e86966df1f : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/61956b671d664de28cf27193866586be : SUCCESS in 17m 57s","accounts_in_message":[],"_revision_number":3},{"id":"e4b468c32455d9bf38d667e4e0d32110a2cfcece","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 16:16:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b98361ee68147e9b36ac456cb55503e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f704b10d402349a3a0ad02bbadd5c4d5 : SUCCESS in 3h 00m 21s","accounts_in_message":[],"_revision_number":3},{"id":"d3d599ba6f5b785a95e7de6e9d3129e25d19fc32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-20 12:32:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9d57ff693d0c4efad091f07314de2bf0dbc84916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 13:30:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0b60f8d95f548a9b91fd80e5cb41c07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/711f65fc88814759913ab3f6a454939f : SUCCESS in 8m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17ec7c550ade464fbbc2ee52175b6766 : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58b5662954104d238b3ce916fb190df6 : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/30aa958c2b324f70985106599794e26e : SUCCESS in 22m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d117e44e9c9840a0855179c362c443ea : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/96d877e8351243679e17d9a23ddd97ea : SUCCESS in 15m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1f60557ebc794f8fb657996e59016f32 : SUCCESS in 18m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ea438c4471464347b63cb5e34493b3f1 : SUCCESS in 18m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/112ff35ccb3d4a12b7d95c1e918c54da : SUCCESS in 30m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c1ecf746be5545719d43942654ba36e7 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5e17d2433158432d88d67ce733e39ac4 : SUCCESS in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5bd4da05b63d40fea08f115401ca91c1 : SUCCESS in 16m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/759a1ad8cea742ef9154a3f2509ff1be : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f875bc1d01e34944b479fabc55afe6ec : SUCCESS in 15m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/20a3e95559a144149412301d14b58311 : SUCCESS in 9m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3f40e3a4d3574076a5ccabdca95a0c40 : FAILURE in 5m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/336da9f5ca844debb03a9f0ad1bdb71c : SUCCESS in 15m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/31f762c736c8429e95dad9fecaa6163d : SUCCESS in 6m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc5e6984e52e440eb4934c184e6dd270 : SUCCESS in 39m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ca429a211eb4332970900378a885443 : SUCCESS in 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1876c85fbae4e99a26ac806e7d38c8b : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec7451d13cf9440ca1bccd4d0af2b81b : SUCCESS in 25m 07s","accounts_in_message":[],"_revision_number":4},{"id":"f0c25a2d209b85d6ebee4e4ee1e8801da2ec5117","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 13:42:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4107670739ea412ca884600205e6b92c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79a23f84e9914e7abd92df8de98deb56 : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":4},{"id":"ed410ea5336d535a7bb0a79c5a51269ef179b6c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-20 15:43:02.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f8a8c6e66c65c9775b75487d5145eb4d9a101c61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 16:43:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/702c1ef2284441a58e2a50f4230a586c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382a66d83e8545e68eb503a1e052ae1f : SUCCESS in 7m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd5028031efe477cbb3fbdcb5fa8ae7f : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35ad7233db69484aaede02ab6b608867 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b40c904f29b4d5e8b1fcc3e5873359a : SUCCESS in 26m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d5150d3bc394af2bba05ef42c4b142b : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8e30ad3672074b48abc3cc8aa6e90e29 : SUCCESS in 21m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6e02941e552a46399127effc499e2db5 : SUCCESS in 20m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b21180e2429f4cbd9f6d8e525dbe4208 : SUCCESS in 23m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f7282ed2a71e4a17a13fb314a6b690a9 : SUCCESS in 12m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6aeeb5737a67486e85e188a62c601e8c : SUCCESS in 19m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/511a3d9854854171b1b3de652589d657 : SUCCESS in 21m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1c36d4e6a6e6488ab527bf601d412ce5 : SUCCESS in 17m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a4223cd783674007ac2c906bdf9ddfc8 : SUCCESS in 8m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0033b5852d6147dbbb44794915da04d4 : SUCCESS in 17m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e5a5d706304a4f3c9c29455f033e4c33 : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9742862e79d45c58e9da534c6da7c6f : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/32b2a98081784d4582e16ec84abafffe : SUCCESS in 9m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b94f41352f34fbfb3cdb42025cfcbf5 : SUCCESS in 7m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c2e0becb7b5466fbad01fe560b66971 : SUCCESS in 58m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1247670d8c08442384ae7d62157d57d4 : SUCCESS in 18m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52c12b43602d42c6af4cacf4850f2ed8 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/964cbfe6011f4de3ba9e37f4e4ff7902 : SUCCESS in 24m 30s","accounts_in_message":[],"_revision_number":5},{"id":"fded5c358b0d5b9d83b9c0a6e5fb5c2fd616c0f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 17:54:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b550202656741fc879e8d443608f834\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/14e75c868ed44f1a9ffabad5ecbaba51 : SUCCESS in 2h 09m 01s","accounts_in_message":[],"_revision_number":5},{"id":"f4838ebbb110c78b191e951d784802669e2cb7cd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 10:13:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"079ae65edcffb22e2d23a14ff02805f7380aec05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 10:13:36.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3769d03eb0983b63c0e2cfaf31889bd2a08794b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 11:45:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a410774b77f4cabb5053e31abc284eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bcb1569fdaf4949b13ed0e142259c50 : SUCCESS in 10m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ea7a9a0aefae43dfaf40bb59fc58a2b0 : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e330be45cf89475bb3991f6e750da381 : SUCCESS in 19m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/171f9db3a87340a497de844c390880b6 : SUCCESS in 16m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5793551a1b99422d9040efa25b50a941 : SUCCESS in 21m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a704f8eb808b447cb4a0b5b9028b3efb : SUCCESS in 36m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4ec98531a86b45b49071b7e0e524ae74 : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8e6c5cfb48ee42f1bfdbec4d4b181e5d : SUCCESS in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/067e86ec99474bc2a7d3f472ad7a6912 : SUCCESS in 18m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/66a35e90aaea4bc1b05e1922a822f8aa : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6ececd5cd366474e96adabe92f96610f : SUCCESS in 14m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cc094a4c6a404fea8a9507bae0506f17 : SUCCESS in 17m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/708b7af3f0f54756b54a5ec1e5ad9efb : SUCCESS in 6m 17s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b18e373befb4e6eba37eb776acfd11a : SUCCESS in 7m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/371bb751589940658a78aa8235b60145 : SUCCESS in 54m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5748e1126424fae8739161ed2becd86 : SUCCESS in 22m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac142b6601cd47ca8fc715ee387353e1 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f7b48a76407043a8a740e467cea3da18 : SUCCESS in 25m 49s","accounts_in_message":[],"_revision_number":5},{"id":"17df501114e6095399b489dbd58f9d05b23cee88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 11:45:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"0d0307f78ff51bd27dfd0a34e8cb94fbdcf3177b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 11:47:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb3c07c02d4841e3aa114123815996a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4e19b42a06141518ea737348250c199 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fac45c3fd7f64e1eab65bc8bef3b49d6 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5a5e0f7999f66896678f89cf9fa21df74b70c26f","revisions":{"fdcdc24a5e8c0400003e283bab92f2d5721f697b":{"kind":"REWORK","_number":1,"created":"2024-11-13 10:37:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/934943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/934943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/934943/1"}}},"commit":{"parents":[{"commit":"7d60f954a8c43c42f516bd5c372dedcfe13a8164","subject":"mpu: copy more user headers to final user object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d60f954a8c43c42f516bd5c372dedcfe13a8164"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-12 16:29:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-13 10:10:03.000000000","tz":0},"subject":"mpu: forward headers with uploadPart subrequest","message":"mpu: forward headers with uploadPart subrequest\n\nInclude client-provided Etag, Content-Length and Transfer-Encoding\nwith an uploadPart PUT subrequest.\n\nChange-Id: I2e14c623e864ddaf12751a855ce66b31eb7a2832\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdcdc24a5e8c0400003e283bab92f2d5721f697b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdcdc24a5e8c0400003e283bab92f2d5721f697b"}]},"branch":"refs/heads/feature/mpu"},"d38cfd2dcf142b10a72705e92e2f267cf994abc7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-15 17:16:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/934943/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/934943/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/934943/2"}}},"commit":{"parents":[{"commit":"b34701df03b0d2057d8d8e9793c07603c73eb22e","subject":"mpu: copy more user headers to final user object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b34701df03b0d2057d8d8e9793c07603c73eb22e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-12 16:29:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-15 16:57:12.000000000","tz":0},"subject":"mpu: forward headers with uploadPart subrequest","message":"mpu: forward headers with uploadPart subrequest\n\nInclude client-provided Etag, Content-Length and Transfer-Encoding\nwith an uploadPart PUT subrequest.\n\nChange-Id: I2e14c623e864ddaf12751a855ce66b31eb7a2832\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d38cfd2dcf142b10a72705e92e2f267cf994abc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d38cfd2dcf142b10a72705e92e2f267cf994abc7"}]},"branch":"refs/heads/feature/mpu"},"c775e0c11bdbf660aed09ea71ce4b0b6c5aea6a4":{"kind":"NO_CHANGE","_number":3,"created":"2024-11-19 13:12:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/934943/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/934943/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/934943/3"}}},"commit":{"parents":[{"commit":"452b30814bc7cdd63c174c4025395aeba4bfdea5","subject":"mpu: copy more user headers to final user object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/452b30814bc7cdd63c174c4025395aeba4bfdea5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-12 16:29:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-18 15:38:00.000000000","tz":0},"subject":"mpu: forward headers with uploadPart subrequest","message":"mpu: forward headers with uploadPart subrequest\n\nInclude client-provided Etag, Content-Length and Transfer-Encoding\nwith an uploadPart PUT subrequest.\n\nChange-Id: I2e14c623e864ddaf12751a855ce66b31eb7a2832\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c775e0c11bdbf660aed09ea71ce4b0b6c5aea6a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c775e0c11bdbf660aed09ea71ce4b0b6c5aea6a4"}]},"branch":"refs/heads/feature/mpu"},"db4e8aa8621b9fd95059a2a5b5a82b32cda16a2e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-20 12:32:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/934943/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/934943/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/934943/4"}}},"commit":{"parents":[{"commit":"71d178a5aea15dafcfb60bdb52bb25fa9c12433d","subject":"mpu: copy more user headers to final user object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71d178a5aea15dafcfb60bdb52bb25fa9c12433d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-12 16:29:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 14:49:05.000000000","tz":0},"subject":"mpu: forward headers with uploadPart subrequest","message":"mpu: forward headers with uploadPart subrequest\n\nInclude client-provided Etag, Content-Length and Transfer-Encoding\nwith an uploadPart PUT subrequest.\n\nChange-Id: I2e14c623e864ddaf12751a855ce66b31eb7a2832\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db4e8aa8621b9fd95059a2a5b5a82b32cda16a2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db4e8aa8621b9fd95059a2a5b5a82b32cda16a2e"}]},"branch":"refs/heads/feature/mpu"},"5a5e0f7999f66896678f89cf9fa21df74b70c26f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-11-20 15:43:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/934943/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/934943/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/934943/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/934943/5"}}},"commit":{"parents":[{"commit":"bfbede94eb76e0eadd480d1f297822a3c8a63f0b","subject":"mpu: copy more user headers to final user object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfbede94eb76e0eadd480d1f297822a3c8a63f0b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-12 16:29:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-20 15:42:32.000000000","tz":0},"subject":"mpu: forward headers with uploadPart subrequest","message":"mpu: forward headers with uploadPart subrequest\n\nInclude client-provided Etag, Content-Length and Transfer-Encoding\nwith an uploadPart PUT subrequest.\n\nChange-Id: I2e14c623e864ddaf12751a855ce66b31eb7a2832\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a5e0f7999f66896678f89cf9fa21df74b70c26f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a5e0f7999f66896678f89cf9fa21df74b70c26f"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
