)]}'
{"id":"openstack%2Fswift~934964","triplet_id":"openstack%2Fswift~master~Idacf841538093f0b147832d433d45301985db96d","project":"openstack/swift","branch":"master","topic":"removeLoggers","hashtags":[],"change_id":"Idacf841538093f0b147832d433d45301985db96d","subject":"sq: fixups for test_logs.py","status":"ABANDONED","created":"2024-11-13 12:46:18.000000000","updated":"2024-11-14 11:16:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fdc0e536c87f5a019e4de0c3ba013f9f219f1667","_number":934964,"virtual_id_number":934964,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-13 13:50:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-13 13:50:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc6fb39dc2f1dc474c8eb16b0825d7a0c6c8fabd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-13 12:46:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0c11c0f8e93256f344708c5f736f23437c701b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 13:50:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/671f3fa1a4d94447a1bb53aacdeae6e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/615eb4349e644b0b9dd20639f02307e4 : SUCCESS in 8m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14b48ad7b0544988a96fe6d330e11cf5 : SUCCESS in 51m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cc01ebef7cd4c939aecf5322ca6a231 : SUCCESS in 12m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f69846a536d0418bad01dedaaf314097 : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ea3f57c5d684ccda2eea69d5151b86c : SUCCESS in 21m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c289db5154ad4cb4b0804ea41490c632 : FAILURE in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aeac354958464450ae6156a2779c29f1 : SUCCESS in 19m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2f507fe9d2cb45c0b86b1f6a87227998 : SUCCESS in 17m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/883be56d5044462a950b85cebbc05581 : SUCCESS in 21m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/46bc1d582d914929823c96b2a9479723 : SUCCESS in 14m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/855976ee548c4a21adfad517c9fad3f4 : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/83ca4b3db1574a3a8eb6bd1c3b668547 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d8c6df826a0a4ee2ad63221d5e3e9bd5 : SUCCESS in 17m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/80b0cdb7f2f24aed8820bab5716676d4 : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7a5aebc173df4aacbfb9d2da4ff79eab : SUCCESS in 12m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/86ab5f8940c440fabe68007eb40715c7 : SUCCESS in 17m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8bce965a95e747bca465eece27dbc155 : SUCCESS in 31m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dd63df8821684759a8db22fd0019c240 : SUCCESS in 1h 03m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a9b9af33a17428c9808401656ae7319 : SUCCESS in 24m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/112e668968fc405680af72f0f70926bc : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ee05b070d7a48c599ed4d4c3f78e33c : SUCCESS in 18m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5cc5e0ec04f0d06e6ceedff30aeefa65565a94bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 14:46:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/185f540248fa489ab1a225e632fc221b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57c5986cdf994e918acfc16bbd714e41 : SUCCESS in 1h 58m 38s","accounts_in_message":[],"_revision_number":1},{"id":"fdc0e536c87f5a019e4de0c3ba013f9f219f1667","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-14 11:16:33.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdddc2f732074beba1b6ed95c35b4277d08e4d58","revisions":{"bdddc2f732074beba1b6ed95c35b4277d08e4d58":{"kind":"REWORK","_number":1,"created":"2024-11-13 12:46:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/64/934964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/934964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/934964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/934964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/934964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/934964/1"}}},"commit":{"parents":[{"commit":"44fea4420e517b398bf02b9afc55fbc948a0da67","subject":"Remove PrefixLoggerAdapter and SwiftLoggerAdapter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44fea4420e517b398bf02b9afc55fbc948a0da67"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-13 12:38:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-13 12:46:02.000000000","tz":0},"subject":"sq: fixups for test_logs.py","message":"sq: fixups for test_logs.py\n\nChange-Id: Idacf841538093f0b147832d433d45301985db96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdddc2f732074beba1b6ed95c35b4277d08e4d58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdddc2f732074beba1b6ed95c35b4277d08e4d58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
