)]}'
{"id":"openstack%2Fswift~935279","triplet_id":"openstack%2Fswift~feature%2Fmpu~I9ddf1ae40daffd20016f9551f29c7c7081abcc4d","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-18 11:34:01.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I9ddf1ae40daffd20016f9551f29c7c7081abcc4d","subject":"merge master to feature/mpu","status":"MERGED","created":"2024-11-15 11:19:16.000000000","updated":"2024-11-18 13:45:10.000000000","submitted":"2024-11-18 13:42:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935279-merge-master","meta_rev_id":"d34477b996e10d59f364e6f500d6bb4e9a58c606","_number":935279,"virtual_id_number":935279,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-18 13:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-15 16:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-15 16:18:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-15 12:12:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-15 12:17:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79efd5e7fd51c77b35a1015ed34251c805aee91c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-15 11:19:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7422a48eb85deb3e202a039af70182777081066b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-15 11:21:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9774b9836ed33cc8028c266b5687feee79e047fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 12:12:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8d466739bec41ab885a045abee94c90\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/669e2a7629504027b2bdb39f09719ea2 : SUCCESS in 49m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e44843668988ceb68beaa6f41a506dd1da4696f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 12:17:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b8c6842a02a48c8bb26b4170c84481a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd575809d2b74e12b6eb087d5c6759e4 : SUCCESS in 8m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93e71816c4e74844b81f83e5673a70d0 : SUCCESS in 27m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c72ad541747d40d395958a54d01736db : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/436bd66d543e4e6382dc48a8bfc90d85 : SUCCESS in 26m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d41073c37f5a423bbbec1267b8ccece3 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a487ab49a21d47bca126a19126d94259 : SUCCESS in 10m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0f75797ba6094f94acbf2d44646cac55 : SUCCESS in 18m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/40bfc64f33dd401781472198e46eb643 : SUCCESS in 10m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4eca1cabe629431eb94c653fd59e16bd : SUCCESS in 34m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e5828374817a49c39137daee74c81471 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/57a3c1d867894d0c9edd6d411921463c : SUCCESS in 10m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ce725f53cea44da6b610be66511e248e : SUCCESS in 17m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0a918d364e6347509148d9c23da97aa0 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6919211cc3be4d048f59f099eb5c7dff : SUCCESS in 13m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a6bdb928995c4521aaff41ce52056c2b : SUCCESS in 15m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c31d851a9294f08bd72707376c98c5b : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c8802289ac840f085e77544b41d971a : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b2d4a39a550e40b798de3fafa4eac1be : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3e10983c6b643f687eb329256ba03ed : SUCCESS in 50m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10365059aca04d77b008628a291e6c7b : SUCCESS in 21m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8afb33481ce94fa2a763ffbed0669584 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/995560fd805e4c3ab94b30eaf679b5ff : SUCCESS in 18m 00s","accounts_in_message":[],"_revision_number":2},{"id":"23d3eddbe9df454dcf50ca8dde9eee87ad525787","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-15 16:18:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"42121f760c5925eff220f1521b9dbd4fad7077eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:19:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3bedb9e662cb2a1b5f3ff106a1b74d819be8344e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 17:26:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e6a0ad7eeea430084e28f6f349a151c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dddef69fa134877966d1528af52b2e3 : SUCCESS in 9m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/773b2197822e4dbbb9de85c22486f48b : SUCCESS in 7m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d03f98b599734fe1af331ebe94a396d2 : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b0200dd55a049c7aefda9ca699d1180 : SUCCESS in 19m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a296bb459b2e49fca1a78b041fa11562 : SUCCESS in 23m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0b54193171e44925a36f90b615300c72 : SUCCESS in 29m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e9d169aaad5f41c7ad6398d26c4b7a45 : SUCCESS in 22m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/59635ff16f92423ab3405af3f51866fa : SUCCESS in 17m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7b237b370bad4be68de50096688db04b : SUCCESS in 22m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7ac77a2101a94465b44c80f5690de12d : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a85789b8c48941ed9606d94b4e66a50d : SUCCESS in 18m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/85f02a03cfc3438990ead0d58a0a3779 : SUCCESS in 18m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3146eeda77ff4bfa817a04352a07f07e : SUCCESS in 6m 30s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c8d0e61215bf4a548e36ebda01713013 : SUCCESS in 7m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd42db4024b34215aba275ca4facc3a0 : FAILURE in 45m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5af26a5cdf449a7b7b1e95b86cb0be5 : SUCCESS in 18m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2909f75b1a04895a60bf105bf68320a : SUCCESS in 3m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f825f10163834faf8cd8a1cd70734cd0 : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4f421fe9a07b584043376ca7208b774d01de63cd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-18 11:34:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"01064010d83cfa52e30ae6f59a7597c2f760e58f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 12:30:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fa73c39ce324efb84bf4a5872bda701\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac12ae16b61447089803abd329c804e8 : SUCCESS in 9m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/704afd9298d84814867cbd211f40dac9 : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bbd32ae21ce4237b045ea68b184c91f : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b176c6f21c44f18b5fe0ba9200f3649 : SUCCESS in 20m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6ec241208614286a19f6077150332b8 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fabfc1fac1a14a809d5e7d53bfa19540 : SUCCESS in 18m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6dbbbd1ef7db4105bf997652e5b006c3 : SUCCESS in 17m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d136333b3fae41ccaa5b04be134e6b6b : SUCCESS in 20m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3fbf5e6d5bf24139a3a6ac270402b433 : SUCCESS in 37m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bd95c624394c4a7aa542bda5f88ac758 : SUCCESS in 20m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b74dbbc0e5784af29bfdb3a0d39cad38 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2359bd9dfe9746958694447f45f5b127 : SUCCESS in 11m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/075edc6e7b074c88bf705714f6bd35a6 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1853f6b16441483daa4489e0f8fa2884 : SUCCESS in 16m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fc24cf16450944219a33c2d274448ebe : SUCCESS in 15m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c78238c710d84bf4a3a312f2a19e7ed8 : SUCCESS in 5m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6cb1aeeb93a4a8a81cfab61999e7b47 : SUCCESS in 10m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2566090094854a94b30ec52fc73ecfd1 : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/be8a8482f5224745b11be8a90056dafe : SUCCESS in 56m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c7a06d8329f4519a631d7aec6866da6 : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d97e6ce4c44cdb8677d35f95de16e4 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d929da72a2a640259a180e58608d4e97 : SUCCESS in 19m 01s","accounts_in_message":[],"_revision_number":2},{"id":"fef97f1f5fdaf490a2c7a3b821ab50c20ecf307b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 12:31:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e75e0edda1604d0b7066dec5f8a9c60bd0d974d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 12:38:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2391555cbe184deab585342791a4c220\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe94f3fec91d4a3eac2d616a849944d1 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":2},{"id":"84a301c150f13b8811d82970a55a414bb57f1537","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 13:42:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b39f342df48f42148e308656e7bb02ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f468c378e1140f4b5bace3fecc19faf : SUCCESS in 8m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/230cd2cdb9ba4215b923ebfdb0c7b215 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8855e56b2ee04f4caec2cce050e22e31 : SUCCESS in 19m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57cdf20df13f4b829dd33b8d626eb920 : SUCCESS in 22m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cf9f6fc1f5144959a2d68051bad7eb2e : SUCCESS in 21m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/54c6fd351f3445139801d431a10eb30c : SUCCESS in 20m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c9cdf71b2e654f25b062eb3b443d08de : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69ce1a30eac240efa2ae977f9277ed27 : SUCCESS in 13m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/493c92f9580249d6b301c8b33a3ad0a4 : SUCCESS in 19m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/508cfb93061848dab487191351081336 : SUCCESS in 9m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7592504ced84431e87aadd6270308300 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cc7054850df04c1a8611241970c9efd3 : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3403fe5f97ae436e8ca5ae3a2f1c8356 : SUCCESS in 8m 19s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0d1419c3ab9b46d5a539ee28a52bb781 : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ac1f8a1f6e245bbb18ab265b33ae80a : SUCCESS in 55m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af0055f9f15d444f87e99a2a446fd60e : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9913f63b72d642be95020e56c8d2dd60 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5cb4e3ce503841728005778a873e4055 : SUCCESS in 18m 34s","accounts_in_message":[],"_revision_number":2},{"id":"27ec2e2f5ea8d4cefeed09a43a255243a2b9e528","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 13:42:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d34477b996e10d59f364e6f500d6bb4e9a58c606","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 13:45:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/077c353bc083420ab51ca64580ea33b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/313700b91c194376b762b2cee24b15e7 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ef31c9b33d454664be5408777451fafc : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af22dd3d254984f90e310d2595f0ab276f34988c","revisions":{"e19da0795b2518c1fd9222ff33f60e763247c77a":{"kind":"REWORK","_number":1,"created":"2024-11-15 11:19:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/79/935279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/935279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/935279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/935279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/935279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/935279/1"}}},"commit":{"parents":[{"commit":"97d210c13a3b4b482ca8ceb0d7fb13f974e4b377","subject":"Merge master into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97d210c13a3b4b482ca8ceb0d7fb13f974e4b377"}]},{"commit":"89815389d5376195d9bc1b0bf65f88e9fb7ce637","subject":"Merge \"probe tests: Set default timeout for subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89815389d5376195d9bc1b0bf65f88e9fb7ce637"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-15 11:18:58.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-15 11:18:58.000000000","tz":0},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: I9ddf1ae40daffd20016f9551f29c7c7081abcc4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e19da0795b2518c1fd9222ff33f60e763247c77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e19da0795b2518c1fd9222ff33f60e763247c77a"}]},"branch":"refs/heads/feature/mpu"},"af22dd3d254984f90e310d2595f0ab276f34988c":{"kind":"REWORK","_number":2,"created":"2024-11-15 11:21:25.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/79/935279/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/935279/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/935279/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/935279/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/935279/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/935279/2"}}},"commit":{"parents":[{"commit":"97d210c13a3b4b482ca8ceb0d7fb13f974e4b377","subject":"Merge master into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97d210c13a3b4b482ca8ceb0d7fb13f974e4b377"}]},{"commit":"89815389d5376195d9bc1b0bf65f88e9fb7ce637","subject":"Merge \"probe tests: Set default timeout for subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89815389d5376195d9bc1b0bf65f88e9fb7ce637"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-15 11:18:58.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-15 11:21:04.000000000","tz":0},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: I9ddf1ae40daffd20016f9551f29c7c7081abcc4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af22dd3d254984f90e310d2595f0ab276f34988c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af22dd3d254984f90e310d2595f0ab276f34988c"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
