)]}'
{"id":"openstack%2Fswift~935387","triplet_id":"openstack%2Fswift~master~I5c9af73cc66e17dc662d8c4d97cbf71ded1fed6f","project":"openstack/swift","branch":"master","topic":"test-bad-digest","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-11-25 21:59:38.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-11-25 23:24:25.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-25 23:24:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c9af73cc66e17dc662d8c4d97cbf71ded1fed6f","subject":"Add error msg validation for BadDigest","status":"MERGED","created":"2024-11-15 17:13:30.000000000","updated":"2024-11-25 23:26:30.000000000","submitted":"2024-11-25 23:24:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935387-test-bad-digest","meta_rev_id":"b8b34aa9c3e20afc27d8ce06f1b5999f36b6dfa2","_number":935387,"virtual_id_number":935387,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-25 23:24:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-11-25 21:59:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":2,"date":"2024-11-25 18:21:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-11-25 21:59:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-15 17:13:51.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-11-15 17:46:07.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-11-15 18:53:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-19 19:32:20.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"2221de8dce40f6040496190e41ff8beb0b332d93","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-15 17:13:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f859c234d703de4d5fcdab6cd2377e44c089557","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-15 17:13:41.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"6b648e29c1500bc7784cc1b6d625c91a9efe2fce","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-15 17:13:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5f789eeb70dd9396ba661d8d5f6f7a23a6ccd00c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-15 17:45:55.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"55e72a90852df3e9cfb7a5607b4e461ac0f7a811","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-15 17:46:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7b920e725cda05e41b969aafa9a4a789f5bd5643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:53:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfad68dc1d3e4536b7a543d1f990556a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167607cd689d4a5492c9052f4b4922d1 : SUCCESS in 8m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1bc2f1a466b048d3afa605f6c598a1ae : SUCCESS in 1h 01m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d4af61f773041f19bca0a19fc0654d2 : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e96d846415c64a0aaf234ad3cc9cc48a : SUCCESS in 16m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a11d861e6206449db5e5d54351e3fcfc : SUCCESS in 21m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5d75a181532d46d99ffa9c9896c6d698 : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d01cbf37fa374597a40e7f52b235e548 : SUCCESS in 10m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f18e269af20142a48572fe4204a57248 : SUCCESS in 10m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f156f00cb9b44e6b9fb1ee0070ddd4e5 : SUCCESS in 11m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bd8f8de1cb9f4b2aa5ca8c41e8dd86de : SUCCESS in 14m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/034c0ac1026d41dd859d2aec403a4d79 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1b1ed8a2e1d341d8956b2efd1b82ac5d : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b175e7a9f6da42f1904294a72885266b : POST_FAILURE in 16m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bbb3045a85804e27b49e069506d2958f : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d75280df05204f7da397c5738b2c3d34 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7c701fff43b74dd38fdd3c5d46ea7a74 : SUCCESS in 18m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9b18aa0d61434c4d8432f5d669dec4fa : SUCCESS in 48m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d161aac9887945928775797dabfbdd5f : SUCCESS in 57m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da2315c40e6a40338c058add8c5abe25 : SUCCESS in 20m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b473f7cea9e4418cbf02f1601409ff9d : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ae8e591ecda84ea5aa78581e9c1da2f8 : SUCCESS in 18m 34s","accounts_in_message":[],"_revision_number":2},{"id":"83f7ebdb7d33bc434c75d3e445b99259a348e878","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 19:37:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9490490a3e14c6ca2657ad418a036cf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/24ee18f5f0e74bbfa14c82a67f4f2911 : SUCCESS in 1h 36m 03s","accounts_in_message":[],"_revision_number":2},{"id":"4f41071d383d62aa2f84646011af3867f12c529d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-19 17:23:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":3},{"id":"df58acbf4c06eae2c32cec2cbe1f7f81629c2e40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 19:10:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79eec06adc0d4cc5b2820f2da7868d1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebdac94f5f484f848b0681e56c09ad6f : SUCCESS in 8m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f5bf47173a44cb69d017f9639b40400 : SUCCESS in 59m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf60f5f5b8244fad9d14129ffc45b10c : SUCCESS in 21m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72d054708e9445f2bf1dccf15a449ffd : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/52faedbe10cb4ace847b948aa2675577 : SUCCESS in 18m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9250f610fa4c4ef9be42c2c9a1fe9bc3 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae8dda2efa264b20a46f8c06477a284d : SUCCESS in 20m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/625bd23de8924625a467edec51ecc266 : SUCCESS in 18m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/48032f931dd74436a7455b36ddbe22a1 : SUCCESS in 17m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/910d3116319f46b9bc24e282cc11391a : SUCCESS in 23m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b128364597da48b7863c66907996b198 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/da65fc66181c428eaaaae6a41c27478b : SUCCESS in 20m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ba6926b7277b4794b038fce79a8597e9 : SUCCESS in 18m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4b7f6e92d33e42258a31cdf752ba466b : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6d0b8195053446da804ef24a63d668eb : SUCCESS in 14m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dbb2d226301841f78e6d451463614c7f : SUCCESS in 15m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df2ec6b43c884865a7f977abddd769d5 : SUCCESS in 1h 10m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a97574f684144c9697e99bcb8b5b008c : SUCCESS in 46m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e59fc4ceb4d46398a66a6e0ebc0d93e : SUCCESS in 24m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b6c17b24d2648de94343024f4baddae : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/255efe6930604a5f9e0c8496ed31f331 : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":3},{"id":"b707f93434b9aca12061f16b8959a9ce81a214fd","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-19 19:32:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0944a632b699501c3aeaf6d1d909dee7198f6e70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 19:32:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"093634d8f2902c915a385478ad28932cde1fc06d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 20:43:27.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f65717380e1b4ea5a55908dbc9523114\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f4820ad8ead4c608c1e2d4be7a89b79 : SUCCESS in 10m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b065bf0f608b41e8a6fa838d58a703e9 : SUCCESS in 1h 10m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/947ac31ded6d4300b714ed11422c69f4 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5cabd14a7dcd4398abc6017bf0f98360 : SUCCESS in 20m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f6cd141e39f84ffc9e025cc2e9da5da7 : SUCCESS in 18m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e097e6e56bac42a5bf68c53062b6004a : SUCCESS in 22m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9b3415f736ed458ba46ddc6001370634 : SUCCESS in 20m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d31950e990d3415083e9836956c66e23 : SUCCESS in 17m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69c95617ea2d4c61a72555b270cf2573 : SUCCESS in 17m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae82d37e6b084f1d9341024985890d87 : SUCCESS in 17m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a1c4fba010d340da9668be4e051514bf : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/57feff0447e040a0b3fbfefe87e28e4a : SUCCESS in 16m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fd29c236b9ce4bdba2898121de014523 : SUCCESS in 18m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8b75f134436a4453bb243e6a3a2fd775 : FAILURE in 7m 13s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8fdd1c1c96814fa2a4b23d2db3e917ec : SUCCESS in 6m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b60ec2214c6447d5914b05a23f056699 : SUCCESS in 46m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c29667b052f74bcdbcdb87bf83d7b6a5 : SUCCESS in 58m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d83317765e74bd8a5986eb77c7528e1 : SUCCESS in 21m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2c6b1a5c564b33bbc2c1de3c87ebeb : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6ccabd2e5a2b45d38e663837203fea92 : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":3},{"id":"51fc2156c9fa3919c2a9a0a1f8473617ea2945e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 21:55:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/307190b3675d4d26b3647ecb9b046535\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/152a4087bac74da8b96abe0569195729 : SUCCESS in 2h 05m 03s","accounts_in_message":[],"_revision_number":3},{"id":"b2af7fe29e660962a6cd05894422e334f2be0d01","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-21 22:16:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6708758c89e6d96ec90d8efe4c5c60e78d3f6146","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:13:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf26ec31c2534dada2f27cf65f2e4566\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b2a4d5089c645dc828901d164fe5659 : SUCCESS in 7m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fbd9c8ac78674a938da2d7390df99434 : SUCCESS in 44m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23ca60eebe464e38b4975e8458c97a11 : SUCCESS in 21m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81f1310166224cadbd77d795b2ff3d96 : SUCCESS in 23m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a5bb267004244133b9ff16d6b24545d6 : SUCCESS in 20m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/45611574eb8b47eb8968d61ff94e6fc9 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd755af6b0494cefa12fcf598ac83fea : SUCCESS in 18m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e438c5740bf14e2099505c21ee068d1c : SUCCESS in 20m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a2e38d0c833b407c86a2ea746806af73 : SUCCESS in 9m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/82d5dcec391f4d02989d087a6848a5be : SUCCESS in 20m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e94be4a3d4874df09bf71f070e46ddba : SUCCESS in 19m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5d0d505aa7d84a8ea400f4634cfd031f : SUCCESS in 10m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2cf048ac50f34ef98982e19ba84a27b3 : SUCCESS in 10m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/83583f34791b4cd6abe0f1fbdd3a379a : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8cddd3a77f2442d190876df6a99eb670 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3ff1a424281742c489952a5b86c12bc0 : SUCCESS in 15m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6f9f76744a248429c2c5d6bd7edf394 : SUCCESS in 40m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/889bd6ca83984e6b9362536f08173c20 : SUCCESS in 52m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f51c8cad01147f780e0c5c95fba6320 : SUCCESS in 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3033d69b38143f9bf12352712ce62d5 : SUCCESS in 2m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/147da51ea4754716b0d85081cdb6b813 : SUCCESS in 21m 52s","accounts_in_message":[],"_revision_number":3},{"id":"8efa76d819d257514a11e697fe0c9b6465678374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:14:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"94773e76b58ef29a54f72b6a3571a7be0774cf80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 00:13:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d6d615d69bb40c7a946835fff09cb78\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/beae43ad70634e48b338d8b02b3a3a5d : SUCCESS in 1h 54m 38s","accounts_in_message":[],"_revision_number":3},{"id":"acb8995ac0a992ed03812f671e1069497cfa1a1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 00:14:48.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00a63fbdd86f44c298cc717d69a164c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b86869fdbf4db8a4653ae6ff355eec : SUCCESS in 4m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ffb95d044113477ab1181adc5a9e09c8 : SUCCESS in 31m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5aef2d2e4b154c108c3897e3f1ab96e3 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af71dc80507a4c56bbbf720775bbcaa3 : SUCCESS in 20m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1b659850bf5c48aca28527923da080c4 : SUCCESS in 11m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c541880bc00046fda343d950c8fd3361 : SUCCESS in 22m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e28fa95d6e334207a5bf1cf8bcefab79 : SUCCESS in 22m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a186c01a87634bb09f9a274bda3eae9b : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a8c05d2abec848538ca97fc66c24c0c0 : SUCCESS in 13m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/34531622bf7045e1beeb25e8d371adf5 : SUCCESS in 17m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/670c78b2197e4319a035b9e6276f2181 : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/657c142c574749b09596f4a94ccb44ab : SUCCESS in 11m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fcdbb2ce805341dfa28ccfab3c69fb3f : SUCCESS in 14m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/792334d33e514b0191bf1b14284374e9 : FAILURE in 7m 13s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f5d4457f465c4f0fafa7525ffa9f1f87 : SUCCESS in 7m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15b171057c184fdd8c5b0171fb92f0f7 : SUCCESS in 34m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6417800ace9542e9ab417c07173b8dfa : SUCCESS in 50m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/891c65f537bb43d49d0a3c51212dbea6 : SUCCESS in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01dc9a833a65407aa71c65cbe0603bd8 : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa640a3459ac42a78fdfb0e1025ec928 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":3},{"id":"43e910b4945b2f0337ca39c1a02e72062ed7b855","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-25 18:21:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":4},{"id":"ca25ba8609f252e198046f263f3d816187d0f559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 19:04:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/400030b09bc54933890073db16332e3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5069a21bb2c8443a9c1085117635b2bf : SUCCESS in 9m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e74c9013f6d74ae39153c759fad06dd8 : SUCCESS in 41m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a47d0ba0e0bc4fdb9aba3d4a53e4049f : SUCCESS in 14m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7578f99c11694425a6da82cb64c4d393 : SUCCESS in 22m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78c1ac35e7764d1d8ec7d3eed71e4d22 : SUCCESS in 23m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/51ea51a7ac394cedb1be74afb20867d9 : FAILURE in 3m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f54d9c20b25040e49d67f90fee257258 : SUCCESS in 20m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/72d48a7724b6495d81c446d780ca6e36 : SUCCESS in 20m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/40e764e9009b4a20b32660338d25597b : SUCCESS in 21m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/219b70adbbff46eaaef26d4432c35c75 : SUCCESS in 24m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c50ba884d7a4f2f86694d19d02051b3 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e99f50c982a94c1790f43aa9bd9ff2bc : SUCCESS in 20m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f9dc2190967c4c69bad80287c30ea19c : SUCCESS in 10m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e35f358886b74bf3b15f377d2dc637c7 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7aca927e8eb64edf84bdf9dd9c790967 : SUCCESS in 14m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e0eed00e12e649499e026e8b880f80d0 : SUCCESS in 15m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2edf16c15753429b866169a10cfe0869 : SUCCESS in 34m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/382b9df42d5446e292da6f94a8a8a3f9 : SUCCESS in 37m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a5be3aeeaaf40aaa61ab2487c8052d9 : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3710f6296e6e4bf7ac3908b35dda4605 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24cf8849e9a04d02b85217128272c01c : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":4},{"id":"a9b23c11e8827df530fe29729388fbc913c7808b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 21:21:15.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01a2fecdf6cc4d13bb15d87fafbfc0da\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b6687450d5a4eddb05909fb61339d85 : NODE_FAILURE Node(set) request 300-0025775862 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"a39bdf0c7634691ccf1d8645481904050f967d44","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-25 21:59:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f77499f83a65ac51b95d6ac176cef8d31eda582b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 22:00:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"44a69c06c7ba56c67238241fbe87b3b0926a1ee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 23:24:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e114c77992e245cc9a4d8b7819b45d32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80bc7b90ace54eb3910689bb9d620b7a : SUCCESS in 9m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2551c6f1f1f44290bb2cc573aa26dcfe : SUCCESS in 38m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5ac16f19727d44389fc058d5e50104c3 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c97b60527664452b8b2879991771392 : SUCCESS in 20m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/618a9489ad9e426d8f2a0b3cb5b9656b : SUCCESS in 16m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b91773a48a1346a8b67d9e9e0e6dbc9d : SUCCESS in 19m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c6a62b3699de4910a78ae56737b695c6 : SUCCESS in 33m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b013e96ecf1e4d17a844050ef0b9b457 : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b892d256f4f24905846d3e4309fdb5df : SUCCESS in 18m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76f683a80e1f4fa39f9dfcc2e017dae5 : SUCCESS in 10m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c6bb1fe49c9645bd816c88d60ea79a26 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b80617dfe63244fb93c0136decc561d3 : SUCCESS in 9m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/52153898aa744b558322007ca817c153 : SUCCESS in 18m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/85a6abc7ae8f484bb39d564da42c4dd7 : SUCCESS in 4m 29s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eff93b81c910466b94df6616aab2f53e : SUCCESS in 6m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1ecf57a3cfc41378f93fceb09bf6ac4 : SUCCESS in 27m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c90a8dc73eb4da383b89b247425a54c : SUCCESS in 38m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4232c12a6ec44b0bb52ca4b7b010b1d : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b113a3565644fa9e9176dc15c3494d : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b88ae5de570a41dc8a1484d5d5e5d2b1 : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":4},{"id":"143bee68432216461083a50356bde1e9be993a6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 23:24:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b8b34aa9c3e20afc27d8ce06f1b5999f36b6dfa2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 23:26:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c7e2ade68fd44f7a6abfb3c66387281\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957bb104d0874e33a442201ef67cbcec : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/66755bacbb2f4336953b28302bbb2c7f : SUCCESS in 1m 33s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"41c614bbb55b5da7611ce825213eaba61688eb28","revisions":{"33b98d688f10d3fb7af5e3edbb0a752ad7d75f45":{"kind":"REWORK","_number":1,"created":"2024-11-15 17:13:30.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/87/935387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/935387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/935387/1"}}},"commit":{"parents":[{"commit":"89815389d5376195d9bc1b0bf65f88e9fb7ce637","subject":"Merge \"probe tests: Set default timeout for subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89815389d5376195d9bc1b0bf65f88e9fb7ce637"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-15 17:13:19.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-15 17:13:19.000000000","tz":-480},"subject":"Add error msg validation for BadDigest","message":"Add error msg validation for BadDigest\n\nChange-Id: I5c9af73cc66e17dc662d8c4d97cbf71ded1fed6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33b98d688f10d3fb7af5e3edbb0a752ad7d75f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33b98d688f10d3fb7af5e3edbb0a752ad7d75f45"}]},"branch":"refs/heads/master"},"adca30785263cacd4e5f873939098b96dc06cb6b":{"kind":"REWORK","_number":2,"created":"2024-11-15 17:45:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/935387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/935387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/935387/2"}}},"commit":{"parents":[{"commit":"89815389d5376195d9bc1b0bf65f88e9fb7ce637","subject":"Merge \"probe tests: Set default timeout for subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89815389d5376195d9bc1b0bf65f88e9fb7ce637"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-15 17:13:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-11-15 17:45:54.000000000","tz":0},"subject":"Add error msg validation for BadDigest","message":"Add error msg validation for BadDigest\n\nChange-Id: I5c9af73cc66e17dc662d8c4d97cbf71ded1fed6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adca30785263cacd4e5f873939098b96dc06cb6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adca30785263cacd4e5f873939098b96dc06cb6b"}]},"branch":"refs/heads/master"},"1f8ca567f3e5f6aaa10bfa8dbcbf972d5da1a35f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-11-19 17:23:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/87/935387/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/935387/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/935387/3"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-15 17:13:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-11-19 17:23:37.000000000","tz":0},"subject":"Add error msg validation for BadDigest","message":"Add error msg validation for BadDigest\n\nChange-Id: I5c9af73cc66e17dc662d8c4d97cbf71ded1fed6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f8ca567f3e5f6aaa10bfa8dbcbf972d5da1a35f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f8ca567f3e5f6aaa10bfa8dbcbf972d5da1a35f"}]},"branch":"refs/heads/master","description":"Rebase"},"41c614bbb55b5da7611ce825213eaba61688eb28":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-25 18:21:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/87/935387/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/935387/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935387/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/935387/4"}}},"commit":{"parents":[{"commit":"fa889358acf7675efba2de095777886b4c1ff7f8","subject":"Ensure correct content-type in container HEAD response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa889358acf7675efba2de095777886b4c1ff7f8"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-15 17:13:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-11-25 18:21:12.000000000","tz":0},"subject":"Add error msg validation for BadDigest","message":"Add error msg validation for BadDigest\n\nChange-Id: I5c9af73cc66e17dc662d8c4d97cbf71ded1fed6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41c614bbb55b5da7611ce825213eaba61688eb28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41c614bbb55b5da7611ce825213eaba61688eb28"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
