)]}'
{"id":"openstack%2Fswift~935388","triplet_id":"openstack%2Fswift~feature%2Fmpu~I2ed37a81295c61e229b6e00b13fb92eb86b855ab","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-copy","hashtags":[],"change_id":"I2ed37a81295c61e229b6e00b13fb92eb86b855ab","subject":"Remove test/probe/test_mixed_policy_mpu.py","status":"MERGED","created":"2024-11-15 17:16:16.000000000","updated":"2024-11-18 12:32:24.000000000","submitted":"2024-11-18 12:31:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935388-p-mpu-copy","meta_rev_id":"32c531d340099dc4c686f51aee39fe8512be1539","_number":935388,"virtual_id_number":935388,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-18 12:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-18 11:34:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-18 11:34:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-15 18:13:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a0d62b933e68a98f2c2122377453810d096c594","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-15 17:16:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b0ea509208de2a308a337c287fa7c7d1acaf06c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:13:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef974e386410463aa8638ac28456acc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae4b92181f924ad984a3e37d6a1fedee : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3fc72e9cdfe41dbbb4993efcdcb660e : SUCCESS in 48m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd5fd843cc448cda4a7f0c92e9ae56c : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6b3514fc962d59782e13f8d30c9602652051f80c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:40:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42dfe1f35f184c1898ec06142a57d269\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dd16235f9986407cbc5714d52c9686b8 : SUCCESS in 1h 22m 10s","accounts_in_message":[],"_revision_number":1},{"id":"572427c25964df7eb9d89b8a554ea832417a8738","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-18 11:34:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a38cef91414720d4e376225168b40e310daa3b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 11:35:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1cd7d427b803adb9b8c2d8222625ac20445c27fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 12:31:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fcd0f6d573e485592cc3673778281ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d433690b454ea6896b34fcca2ecc3f : SUCCESS in 8m 27s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b733a082ee2f4b48b4ee9235cd9c33cb : SUCCESS in 12m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ff41bb4844447e2915803456caf7f47 : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3d8aae7acd36420b86d0bc4b02aafc81 : SUCCESS in 23m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/15db73efa94d4f5d9da358b8f4eca21a : SUCCESS in 30m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21697252bd82404db97c4556b163047d : SUCCESS in 13m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b01f0580bc04b5c9e09f11b4b359fa1 : SUCCESS in 19m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/744c581966194491a74931c92054cce7 : SUCCESS in 13m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dde9ad141c9a49cdba424b655e4b2417 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/44e241e7b8154cb9bd50d9aefc2a314c : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/25efea8ef3a34c448197397b61fe650e : SUCCESS in 12m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fce783880360458dac50cdf9f6cc1a38 : SUCCESS in 7m 18s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/db5d1dfee06d482987f843ebbced3cab : SUCCESS in 6m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/616ed860485d472f94a65bd77707aefc : SUCCESS in 52m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac2ff053608249d6bbd28305ab28aff2 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9bcbc166551aa5b3a01c58c6ecaf161b20079806","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 12:31:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"32c531d340099dc4c686f51aee39fe8512be1539","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 12:32:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/549270f8f2864d70a8756a9757b496f7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a0bb413e47444d9046a2237a4fbf13 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c8536a7b77ce42dcbf0039922121fede : FAILURE in 28s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97f464aab97829e3cf96e8a716b78b739f5e560e","revisions":{"97f464aab97829e3cf96e8a716b78b739f5e560e":{"kind":"REWORK","_number":1,"created":"2024-11-15 17:16:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/88/935388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/935388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/935388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/935388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/935388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/935388/1"}}},"commit":{"parents":[{"commit":"97d210c13a3b4b482ca8ceb0d7fb13f974e4b377","subject":"Merge master into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97d210c13a3b4b482ca8ceb0d7fb13f974e4b377"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-15 16:52:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-15 16:57:12.000000000","tz":0},"subject":"Remove test/probe/test_mixed_policy_mpu.py","message":"Remove test/probe/test_mixed_policy_mpu.py\n\ntest_mixed_policy_mpu.py was previously renamed to test_mpu.py [1]. It\nthen inadvertently came back on feature/mpu with a merge from master\n[2] bringing an import fix [3].\n\n[1] Change-Id: I3f78e9bfaf2eadc77114ec3c21c7f15fc3ff99b9\n[2] Change-Id: Ic865f2fc7b32638dc55a045f91c84b0ce930f548\n[3] Change-Id: Id2662bfc5ef2f21f901f1c98e6389c4cb01818a2\n\nChange-Id: I2ed37a81295c61e229b6e00b13fb92eb86b855ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97f464aab97829e3cf96e8a716b78b739f5e560e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97f464aab97829e3cf96e8a716b78b739f5e560e"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
