)]}'
{"id":"openstack%2Fswift~935502","triplet_id":"openstack%2Fswift~master~Iea01928e1daad25f3425f486b9dda4c6fb58510c","project":"openstack/swift","branch":"master","topic":"p-rename-mpu-probe-test","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-03 18:09:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea01928e1daad25f3425f486b9dda4c6fb58510c","subject":"Rename probe/test_mixed_policy_upload.py to test_mpu.py","status":"MERGED","created":"2024-11-18 11:41:51.000000000","updated":"2024-12-03 18:11:13.000000000","submitted":"2024-12-03 18:09:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935502-p-rename-mpu-probe-test","meta_rev_id":"e5c19d22a7d16c3629751cea75a6d57c441b1449","_number":935502,"virtual_id_number":935502,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-03 18:09:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-12-03 16:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-12-03 16:17:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-18 12:33:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-18 12:40:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-03 16:17:02.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"9e47185e836cf7200c8c33b032ba4d20c0ec8d76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-18 11:41:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f48d5cc5722cf50ad65e64a993ee45c361421fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 12:33:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a19ba2ef92f45dd88a014f484aa5182\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f5bc10710244378918d5f22cdf9793b : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6ae765d1cbbb117dddc1348e5e61088741afa8a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 12:40:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/223249a067c341bb971350b3a31f5dd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1318c80d0e463fa1e6e4cddb5b3d4e : SUCCESS in 8m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/24c151c567a940bfb3eb03fb28f5c68c : SUCCESS in 53m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4da0a37fbb7488db9e3b8b9eb8e6dce : SUCCESS in 39m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9acdd32d35da4bdd82dc76ef5de0e462 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c74f91446f3cb0da443ec2aa8d1f3ec1eedeeb84","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-03 16:17:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0588741ebc8aa364c9697793e9c9cf1a4a54475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 16:17:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72f4602e841214e1cb00bf3fd0ac4d59606ee2e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:09:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4a1fa78198c452d8e2d5c8669fc5c9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c04cdb133c48f196a23fe78e6313e6 : SUCCESS in 10m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a746456f978e4bb4840b17f9603b222c : SUCCESS in 53m 49s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4d933e8bc84442a1a87062e448d655b0 : SUCCESS in 19m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e2e0ab27761a4f4db94769093ae1fdfd : SUCCESS in 15m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a377ff67573c47bf8dd341740d4c46e8 : SUCCESS in 20m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1bf59e19a54c471d9cf28f30c4804d04 : SUCCESS in 23m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/10d111367e534cb9bd8edeb11c6d57d1 : SUCCESS in 19m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2a3dd7de2021484bac822d87ca614ed7 : SUCCESS in 9m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a0be234344fd45a490cfe6b1a6bc74c1 : SUCCESS in 16m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/671149b29c884cd698424b0e679eb1f1 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8bd1bf39e6d94d82b76a0ac42c522304 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/03f7fbdd17ac4e3da0da892be13a4237 : SUCCESS in 8m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e23fb4659a734c6d89039afdbf8bf470 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/76d09e6e1d1545b988cadcb3d3d86393 : SUCCESS in 6m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e1b03ce3ac145c981818f5e2c0ffd11 : SUCCESS in 49m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25bbaa8123a04defa06e0662311cbf29 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1884bd007d6321dc2b29224180b23b057b536af7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:09:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e5c19d22a7d16c3629751cea75a6d57c441b1449","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:11:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ebee49593e44a7b46f705d52d30ed5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d026234ef843fdbf9af3eb2deb726c : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/11fc52d3d99d4003a41d8b1ba49a29d3 : FAILURE in 29s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66e4887c91a43325f827ebbb78e21c4e846848d6","revisions":{"66e4887c91a43325f827ebbb78e21c4e846848d6":{"kind":"REWORK","_number":1,"created":"2024-11-18 11:41:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/02/935502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/935502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/935502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/935502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/935502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/935502/1"}}},"commit":{"parents":[{"commit":"aac4c14574a30df6ae47604709ec883eed7df59b","subject":"Merge \"Differentiate unlinks and outdated unlinks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aac4c14574a30df6ae47604709ec883eed7df59b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-18 11:37:48.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-18 11:37:48.000000000","tz":0},"subject":"Rename probe/test_mixed_policy_upload.py to test_mpu.py","message":"Rename probe/test_mixed_policy_upload.py to test_mpu.py\n\nGive this test file a less specific name in anticipation of further\nmpu-related probe tests being added.\n\nChange-Id: Iea01928e1daad25f3425f486b9dda4c6fb58510c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66e4887c91a43325f827ebbb78e21c4e846848d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66e4887c91a43325f827ebbb78e21c4e846848d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
