)]}'
{"id":"openstack%2Fswift~935650","triplet_id":"openstack%2Fswift~feature%2Fmpu~I1e53a8f28ad3cda2562aa281806d571a4d211fa6","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-copy","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-21 14:34:42.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I1e53a8f28ad3cda2562aa281806d571a4d211fa6","subject":"mpu: detect bogus upload-id","status":"MERGED","created":"2024-11-19 13:12:37.000000000","updated":"2024-11-21 19:39:33.000000000","submitted":"2024-11-21 18:08:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935650-p-mpu-copy","meta_rev_id":"f5e0faa58ed0d6d9cfbce29904376c933fc4ec44","_number":935650,"virtual_id_number":935650,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-21 18:08:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-21 14:34:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-21 14:34:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 14:24:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c52ecf0b71aa66ebb29856c6ef05340af105df3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-19 13:12:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5c9db798a20391a235b677d03a418c2ed8b8c27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 14:24:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d1f41fc38da44178343e68c4bba4e7e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b3cca10e8e488985b7273a731a7680 : SUCCESS in 8m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ebf54c297d84020aa7e425b8e29cdc4 : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d71b4d589f840f5a7776a7fc01b8d7f : SUCCESS in 23m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28db7d3f9106478495bac139db5ff60d : SUCCESS in 19m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/245783248acc48cdbb5ca3dfe14e23d4 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e2c99732294141ee918ebe7e97e0fbc5 : SUCCESS in 18m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c08d48c41a654d84865c226900583b2a : SUCCESS in 18m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d71e48929f2c42a1876c90a4405d3a9f : SUCCESS in 21m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3a66d86f2af4428b85fdd45d1bfab372 : SUCCESS in 23m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/78fd4a91140b4093ba171a2da74e454c : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8e620a312ab847df945234b9fe20e5ca : SUCCESS in 21m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/08297f10062645feaa9fa7e2af9c324a : SUCCESS in 16m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/200ee414eebd40ae910b4992436536ce : SUCCESS in 13m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/65cd768689db42509b0ce3846a6202ab : SUCCESS in 16m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1dcd7298cb8e4b9bb5c4f26ec0546fe0 : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a982ce1488248088b65e7a22fae7cf6 : FAILURE in 8m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f99b254ff7814d03b8d888782004e557 : SUCCESS in 21m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ee84e9451c146ea925b1e4eae08329f : SUCCESS in 8m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ccea3fc543f14b4fadc9bee742787ef9 : FAILURE in 1h 01m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56f253c9a0584c9b8f86b1993627f8c1 : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/572ee0440c2b44e78c38e4beccb24d1f : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aaa78c548ad4412f8a315dce5671eb93 : SUCCESS in 17m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d63ebaf2aecae121f5d9b8eb720082b84cbfe1c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 16:17:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f4fedd8afdc4a61b0dfaef22ca9861a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/518e16625d8a4bda8fba1b8d6ef8f514 : SUCCESS in 3h 00m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8bb885fe181060177798f912f9b2bd891a4f3937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-20 12:32:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"340467ccb8ae7b437d5730622445663498576bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 13:36:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd06e63cdc2443809b00aace5fa6c6c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c47eb44dbe4f8085d5e96b12da2219 : SUCCESS in 7m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/164e6050a723497394d8398b96662463 : SUCCESS in 15m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e13d4b1e13074195a2304648ba44305b : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08aaf3e03bf84d0b9899e808de21793c : SUCCESS in 19m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5d7875cb3b8844c48ed911de8e702c40 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae40d3581c0c4456965f231e5ac40383 : SUCCESS in 19m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ea4409e3474bdfbf8062de0466feaa : SUCCESS in 18m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4104609fe6974d1999654e8afa7ff302 : SUCCESS in 22m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/712e8eb6492d4a6296405d810feffd43 : SUCCESS in 29m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c51c05943a941a9a3006a7c86e30696 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8f666d1419884a9bae5aee810bf83d76 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/203306cac4b24ba5a436492893186f6d : SUCCESS in 17m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/53a8dc891a7b48c79d8d483091c88b20 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3fbfa37b5d0449fa97462b94b63145ea : SUCCESS in 14m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e8d368afdcf94eb4bcca73d801ee7c17 : SUCCESS in 14m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/430cf5210e07473487ea6f33e1a61c77 : FAILURE in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9b6130ee635f41878fb3a2e19730c98e : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/820251c6e3cc40c38d330fed9435abd3 : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/325a8a8a7de340bd8b8b2b30a17e963b : FAILURE in 45m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/828334c3f6244d60b5635c9d7cd57774 : SUCCESS in 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7098b845000744d3a910eb44a4ec5091 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7b1be6e384b34daaa313763f41592680 : SUCCESS in 17m 01s","accounts_in_message":[],"_revision_number":2},{"id":"982f2e4726206f886e96a9ccdd3dfbfff0cbf956","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 13:56:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10dfdaff18664f37a65a80b6dd6957c1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/04520ab4bb3346da816fd004474abc9d : SUCCESS in 1h 21m 40s","accounts_in_message":[],"_revision_number":2},{"id":"1920c11ec3599dedc26ec0ef9a5ae3439b8a44e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-20 15:43:02.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9bea01d1f81316c17c30a456818f8dea3c8c89ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 17:01:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d27aa811bc274b94ac6277c9c07077f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d960900f1b1e40dc8387ee36c57ea7e7 : SUCCESS in 9m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b32ab0f23be34637be69b916647ddcba : SUCCESS in 16m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7539c4380cf743dbb59cd4f1debb941f : SUCCESS in 17m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84764f2e662547e1ac1d6fbc6b473a54 : SUCCESS in 19m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3cca6c847c0042d59ab6f5650de95bac : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0e6fb43169354d548d8e39aa93135abe : SUCCESS in 15m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7613032da8cc4ce5b5b1dbde4a1abfa7 : SUCCESS in 19m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c91ff2ee378546299a3274d858167e31 : SUCCESS in 20m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b6eb2707cabf442ca98bb788cb45de71 : SUCCESS in 31m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3cae7f87afe84dff9c9008984ca33b8e : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0f13c05e6ea94742a177453e90c89b7e : SUCCESS in 16m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/85bd012d08354f3aa3778755b081adb9 : SUCCESS in 19m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/faa2e1abe1f845aa98e42a4131f9f6a1 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/65a92cc0446d48a189a47e70dcdd0a05 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/202f03cb51274381af5c211cfb0efc32 : SUCCESS in 14m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a13024552a2b4059b4be55af9cc41d56 : SUCCESS in 8m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d3b55a9fb804e5b9310b0ef661d05e3 : SUCCESS in 14m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2ff7681e72224bccb1268bb5ca358579 : SUCCESS in 7m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/325919909a564e9daaa7ac7cf6e9b8fa : FAILURE in 47m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b211d146a44946d5b4c9732ab9b38d0e : SUCCESS in 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffe4dd88c49541f8afc67a2095d05bbb : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba98a6355ee94f5b86360df338604da0 : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":3},{"id":"1f89fb35b1d6d595f4605d0513ecabb7e0f1c65c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 17:37:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38f87b121f0d47e58e35716ebaaee010\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fdc0a232f7484ae294df8362043efd17 : SUCCESS in 1h 51m 19s","accounts_in_message":[],"_revision_number":3},{"id":"a8b36ec05ff60e1ef08598d8801aff2862db9d08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 09:50:21.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2bc27da1a49dcb0835c1e7d2fdf4bbd6c81d4b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 11:39:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccd9ae183cf64eb79c2c984d9e6f5d50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a0351e6f904fad887123ecee95bc47 : SUCCESS in 9m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/885d8525d1874dff907a80be6bdaa4e0 : SUCCESS in 22m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b04d0d8693a244fa9bd30bbbdda972be : SUCCESS in 22m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de086ddf6506472eba374c99b72f31a6 : SUCCESS in 23m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67c7fa1e653e4b539b2dd224faad88a5 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/91234fba62964a9ca3ca781ddb9fffda : SUCCESS in 20m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/be5633f9caf44140ac1d0f018b7a9bc7 : SUCCESS in 21m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8a501331d12749d4a45af9e6eb8e2aad : SUCCESS in 21m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ea9f77e794f64cb2b875d4950360edad : SUCCESS in 25m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21fdb28904f74dd08c7dcccf93a27b89 : SUCCESS in 18m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4d0c3f4898464e41a528505e6c66c6ba : SUCCESS in 16m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76401184617d4e8794c9c96fff439034 : SUCCESS in 20m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/077061622bc84d33860a4e3a349194d9 : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/45722044335f463f97a16fa02b7e38d6 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b3414b7d65ce4f159965a700dadc10df : SUCCESS in 16m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3472fcb495e4488587a3f96433a5a7b1 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3631a99174fb4512a45f8fcd468344b1 : SUCCESS in 13m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6567153da00048c5bdda454ac35401be : SUCCESS in 6m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/40294a1cf9a14258acbebc747ea9ea58 : SUCCESS in 47m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d33a7c61d9a74926a74eb672daecfc5d : SUCCESS in 21m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363506af7e4c4a69a7212eca33e704cc : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c3b7cf2adf5f45d982716e307274e8af : SUCCESS in 18m 59s","accounts_in_message":[],"_revision_number":4},{"id":"0d1b4c0a6ba141c74042b1730c52abcb2004fc8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 14:13:33.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c3d17503be04e8ac2c37cb139ba9a52bbb438033","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 14:34:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5fbfa6f2cb98fcbf5052d33d877f6a8a42735f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 15:09:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e32891380dcb440aa66be7de4fb7c741\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81dbbede88a4efda694b9c7ddd0125e : SUCCESS in 8m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a683e7e79f748ddab490e2fd270732f : SUCCESS in 22m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9b119568a2e47ecbbd52f6e9596ca99 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1d0853b58704945ba75bdb81c5646df : SUCCESS in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/176605e4cd8146e599274834b95b2932 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/60909a308cd44d3ca398d6405ff00fb8 : SUCCESS in 23m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2f6594f94e14a11bbebeb50ecdec3d0 : SUCCESS in 21m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e6571c1213484f0a9cce6888f2839336 : SUCCESS in 20m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5741cf01078a435c82f8c7bc0b08ee2b : SUCCESS in 27m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7ade863fabbb4f72b56279f22e0b7fe6 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2dc1896784ce48db99a0f412a8a7ecbc : SUCCESS in 17m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/67347c99b79d434895cf43a78ed212a2 : SUCCESS in 16m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/83c5673f4c0b4d7d8493882bc965cdb2 : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1eaecff510f245fc801f61dcce741f39 : SUCCESS in 14m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cd611844f0e145b3a7c2be50b495ce8b : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8bb67212a92842058673594b604ffe10 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69b0296404a04743916daa85471d5a99 : SUCCESS in 13m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9fe4240961954c71bc438a61885d59a7 : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4152631da8cd4ec99496a94c8f017cbe : SUCCESS in 43m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89f7e03f3d254d7aadf3545b2b7bef95 : SUCCESS in 19m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce219a19aa564ffdae57b62e4aca17a6 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bfd5963865d64ec184519a75a5936263 : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":5},{"id":"ba043b08e62a30a5e4cee968ac56c1f218feef3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 15:23:36.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"01f8e5133215270ba017fcd2c7c3da9b9031912f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 18:08:00.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa9eb0c001b340bdbdc92f4b0b599ae2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b4844c262d74260900a5b0b9187fd61 : SUCCESS in 9m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c00027c9290446f681361aa401cbdd13 : POST_FAILURE in 6m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d5b74360d08d4c0da411b3db1db7fb77 : SUCCESS in 20m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd95fcc0851c4845aa75c44200022546 : SUCCESS in 19m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/92a7559f02284b8284e5061f47151cdd : SUCCESS in 18m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/260f83d77efd44d98086dc3081703f7b : SUCCESS in 22m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/90b66f0f313843bea1a1a4cf7f34b78a : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/68c74d1f3c794ab3a85dc2b73c9ed8c9 : SUCCESS in 15m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3698480ef5fc4450af3ed94aa2ff9c9e : SUCCESS in 18m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7053c6b72b0748f78e2dc7a44284f7f5 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/226b56234dab46b5afd426217931fc68 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ba580cce2b7044f48b7b31bae506a484 : SUCCESS in 17m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9267b04146a74f248fb3845a7daf5d89 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bb949f4f26ad45109595924c3f3a10cc : SUCCESS in 7m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e29f604b90d247eeafc51ca90be63bb4 : SUCCESS in 52m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb77c27b5dc3431ba6156eacbcf4f90b : SUCCESS in 22m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebf1c3f7ad3240e998317bbc0fb5e8dd : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/371465d419114a23bff04b78008ab91f : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":5},{"id":"67549cebc1796a7256ccb04c2cc2316ecbd58413","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 18:08:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3479d104f8a2bf86f4fe49212329e8ac19999619","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 18:09:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c8b80790d1d45fb81eb9b8ab3c82406\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf20b128ed2419eac27139f40e4d28a : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f794302b37d84c6c87426316edcbd67f : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f5e0faa58ed0d6d9cfbce29904376c933fc4ec44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 19:39:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eead0bf68b254ba2a4d6cb8aee070da8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/67c56b621c014746a889f5ef7e08e11e : SUCCESS in 2h 03m 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"10c5de5b892d172f90f71d97b39afe34fcd17072","revisions":{"677ff80144886263c7a2309d95df480c823b030e":{"kind":"REWORK","_number":1,"created":"2024-11-19 13:12:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/935650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/935650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/935650/1"}}},"commit":{"parents":[{"commit":"e7b35cc0fc4b873ce07749970f63724bd5d1e37b","subject":"mpu: detect content-type during createUpload when not given","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7b35cc0fc4b873ce07749970f63724bd5d1e37b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-18 16:43:14.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 13:04:55.000000000","tz":0},"subject":"mpu: detect bogus upload-id","message":"mpu: detect bogus upload-id\n\nAdd a hash of the path + uuid to the upload-id and check that\nthe hash matches when a client request sends an upload-id.\n\nChange-Id: I1e53a8f28ad3cda2562aa281806d571a4d211fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/677ff80144886263c7a2309d95df480c823b030e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/677ff80144886263c7a2309d95df480c823b030e"}]},"branch":"refs/heads/feature/mpu"},"df9d51b9bb56fd9eb413a4af6dcc713afc6ef147":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-20 12:32:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/935650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/935650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/935650/2"}}},"commit":{"parents":[{"commit":"0f69696cd8a3ccc46a18d0472044a06b326dbd26","subject":"mpu: detect content-type during createUpload when not given","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f69696cd8a3ccc46a18d0472044a06b326dbd26"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-18 16:43:14.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 14:49:06.000000000","tz":0},"subject":"mpu: detect bogus upload-id","message":"mpu: detect bogus upload-id\n\nAdd a hash of the path + uuid to the upload-id and check that\nthe hash matches when a client request sends an upload-id.\n\nChange-Id: I1e53a8f28ad3cda2562aa281806d571a4d211fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df9d51b9bb56fd9eb413a4af6dcc713afc6ef147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df9d51b9bb56fd9eb413a4af6dcc713afc6ef147"}]},"branch":"refs/heads/feature/mpu"},"fc7d7fdffe962d7f173da44057aa4d8bb7fb17c0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-11-20 15:43:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/935650/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/935650/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/935650/3"}}},"commit":{"parents":[{"commit":"394e566f5b8e16699fb0e9317862d81ea814453c","subject":"mpu: detect content-type during createUpload when not given","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/394e566f5b8e16699fb0e9317862d81ea814453c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-18 16:43:14.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-20 15:42:32.000000000","tz":0},"subject":"mpu: detect bogus upload-id","message":"mpu: detect bogus upload-id\n\nAdd a hash of the path + uuid to the upload-id and check that\nthe hash matches when a client request sends an upload-id.\n\nChange-Id: I1e53a8f28ad3cda2562aa281806d571a4d211fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7d7fdffe962d7f173da44057aa4d8bb7fb17c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7d7fdffe962d7f173da44057aa4d8bb7fb17c0"}]},"branch":"refs/heads/feature/mpu"},"62ebd2c35672ddcbb58a63e4ba593818261cc29b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-21 09:50:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/935650/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/935650/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/935650/4"}}},"commit":{"parents":[{"commit":"7c1587f06c758579efca593f2ffc71209963a846","subject":"mpu: detect content-type during createUpload when not given","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c1587f06c758579efca593f2ffc71209963a846"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-18 16:43:14.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-21 09:45:11.000000000","tz":0},"subject":"mpu: detect bogus upload-id","message":"mpu: detect bogus upload-id\n\nAdd a hash of the path + uuid to the upload-id and check that\nthe hash matches when a client request sends an upload-id.\n\nChange-Id: I1e53a8f28ad3cda2562aa281806d571a4d211fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62ebd2c35672ddcbb58a63e4ba593818261cc29b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62ebd2c35672ddcbb58a63e4ba593818261cc29b"}]},"branch":"refs/heads/feature/mpu"},"10c5de5b892d172f90f71d97b39afe34fcd17072":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-11-21 14:13:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/935650/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/935650/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/935650/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/935650/5"}}},"commit":{"parents":[{"commit":"63539afe6288cf502cc1e113b667718a942b1720","subject":"mpu: detect content-type during createUpload when not given","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63539afe6288cf502cc1e113b667718a942b1720"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-18 16:43:14.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-21 13:33:13.000000000","tz":0},"subject":"mpu: detect bogus upload-id","message":"mpu: detect bogus upload-id\n\nAdd a hash of the path + uuid to the upload-id and check that\nthe hash matches when a client request sends an upload-id.\n\nChange-Id: I1e53a8f28ad3cda2562aa281806d571a4d211fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10c5de5b892d172f90f71d97b39afe34fcd17072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10c5de5b892d172f90f71d97b39afe34fcd17072"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
