)]}'
{"id":"openstack%2Fswift~935651","triplet_id":"openstack%2Fswift~feature%2Fmpu~I249e9e32ecc1fb2817d1b253752c876fb2674369","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-copy","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-21 14:34:50.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I249e9e32ecc1fb2817d1b253752c876fb2674369","subject":"mpu: copy sysmeta from create request to manifest","status":"MERGED","created":"2024-11-19 13:12:37.000000000","updated":"2024-11-21 19:51:36.000000000","submitted":"2024-11-21 18:08:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935651-p-mpu-copy","meta_rev_id":"831966ab532e13eccb491d682208331c3d1c8b59","_number":935651,"virtual_id_number":935651,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-21 18:08:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-21 14:34:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-21 14:34:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 14:36:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1bd03d288c13ae78c37cf06a3dabd514f42cc9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-19 13:12:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01162ebd372cadbfa4f595aaa592f52e78d96c05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 14:36:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fbe45ad6b0849c49461bc738bf3842c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc54ff1d3984f4c83248d8faec242db : SUCCESS in 8m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76080c8b8b904269bae7dbf4ea8b49f7 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e485324b46724e8cb01c8bfe05b89993 : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2618068700fd4997ab92d2d2844ee6ca : SUCCESS in 21m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c96da34e22994299b47d4232590b9bef : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69597ff0149a45008af94eef3a3038c0 : SUCCESS in 19m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/109ad884afc545ba8ab210f42c972a29 : SUCCESS in 19m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2e7034739ed740ccb2aed98822052b70 : SUCCESS in 22m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6cf4228460d84af58e1291c3608dc801 : SUCCESS in 25m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4c439d66d70e45eaaacad968d1e28167 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3fa45e72d29041c7a05405f06727cbf2 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1c2e0957ab3a48adaf9de2472fb9f956 : SUCCESS in 16m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6336a4c204484c649cfc9b96fbd6ca09 : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f9ddcd9094f04b5290fa24d5be9591da : SUCCESS in 16m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0335cd1e15034e25942841e9ccb3e27e : SUCCESS in 13m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b0d2de29fa214bab921204cc7dd15555 : FAILURE in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b307550dcd1b4fa58bf2a3fd6b8210ae : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dfc4edff211a4bc59c3f220225fbaa0c : SUCCESS in 9m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c48d5855b48644f6932304f463e41dca : FAILURE in 48m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd6a3653caca469cb436508235cc35b3 : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ef35628b5d949b48f85bb35621caa61 : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e79b562e6bbb45238a7ad65dabf57631 : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"dd8eee60e289ed933edd9ca1595be766e61eff82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 16:07:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20a4fa7b324e4aeab0e2c1a5f7d54831\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/985eccd0e8bb434e86d4d9926dc0b5ce : SUCCESS in 2h 51m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0d97ea22bac34b4b99652f08044f44c3503acb26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-20 12:32:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f170195a0fdc5387ed3dfd2432d3312562a1bee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 13:33:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19cca8751687483ca27a80f9d687c773\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763fab3580d84cdfad283efc3464fdc0 : SUCCESS in 8m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd1c0e2ba60c47eb9939240c12e70ba8 : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc48be40a29249fda7bad24cf6e99f22 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/350a8b83301744ae92eaf6dad5a06ef0 : SUCCESS in 20m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/07232f06dbcd467e869ec100e32386cd : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38366ce9a8d54e5e91bad657d591b213 : SUCCESS in 19m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62668ea103884838a5388fcea5b35604 : SUCCESS in 19m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2ed68920711e4c03a5c921df5ad6af0c : SUCCESS in 20m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f0ba01b2464c41e1b3ef29d8c43f3d59 : SUCCESS in 35m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c1445c47cbb74ee9ab120ef1ab131778 : SUCCESS in 18m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/46fd83c9846b4b60a99f2d3513de8c94 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9dfa02041c02476681dd99b058f9dd27 : SUCCESS in 17m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2868f6e0d32d4c0c90809719fcc11f46 : SUCCESS in 13m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2756f6ef1f054ad8a1ecec915a3bf8ac : SUCCESS in 14m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a330f8e537db436b9651ca5ea1aaf60e : SUCCESS in 18m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/16db596df5994a6ca985157c0f042e09 : FAILURE in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6c2bdd2501f467c94640f44f564329b : SUCCESS in 14m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6ee16aba2f0b4b96a74a05935405615e : SUCCESS in 6m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/881f6b0b0da74b64ad14087f01fea372 : FAILURE in 53m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9909daadcb804b048f63a92c4403eb91 : SUCCESS in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75fcd85bc8dc4809a60c67b56e5ef4e9 : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3415451f4c804583825c774b61c022a4 : SUCCESS in 18m 44s","accounts_in_message":[],"_revision_number":2},{"id":"ab2100412ca5e70addafed75fc7a1c29063d6aa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 13:55:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e90038e13cdb4e2aa70e116cdfd9488d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8fc2879e79a445bcb393720076f767c7 : SUCCESS in 1h 20m 38s","accounts_in_message":[],"_revision_number":2},{"id":"abc51f1152f7488bf29d8c440b458c28925ba87c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-20 15:43:02.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b062b2fc4d74714f64c6bc92a6b0faa201088c8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 16:46:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c6809f3217f46f7a512ac7d0929d225\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c86546b40b4cb09d54d81637249815 : SUCCESS in 7m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7b7914767d7445d858f7addee982b4a : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b32b7967e81a4878afe445721557921e : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79865ffa86694aa3ae584dad2184b8b4 : SUCCESS in 18m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c721013bfc4f48f8baac6fcd60416cd2 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/602e57c9e30043489a098f69e41a2429 : SUCCESS in 18m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cab802605e6a49b79e50b588accfcedd : SUCCESS in 18m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ebe0c30f60bf40dfbc6b83f7ae49b301 : SUCCESS in 20m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7b0f8ca2048f442eb4ceaf9ba266b4c0 : SUCCESS in 27m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bdeada8f54e54179b8d9ab02662af2c1 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/67f07fad2aa547339ecd8e9fd6097e5d : SUCCESS in 16m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3add6c03031d4e8094a258a4af74c06b : SUCCESS in 9m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f2dd8e76237f496ba31702709e1efe7d : SUCCESS in 19m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5cfaddd9bfc24caf8679432a9108b23f : SUCCESS in 13m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5db4002cb1194d81ae7bfc45f35eb4ec : SUCCESS in 17m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bef38f3d0c054c02b26d637fcce56559 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f0607a42a404fb7a97c2f3e677d1286 : SUCCESS in 14m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8ad9010b45364a9c9532d4576dce2432 : SUCCESS in 4m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/664ee813dadc49e88e6d74b4c34e0cb7 : FAILURE in 51m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac9ef786b749442d834f854de43415c8 : SUCCESS in 18m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08ad513923ce4d31987973404af7ba4e : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17edd1a4e0024d63bfc3a32f7e76db41 : SUCCESS in 17m 03s","accounts_in_message":[],"_revision_number":3},{"id":"77b8c9ac89d38940f666e771d032c4e6f2114425","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 17:32:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c2870e756ec4a9bbab3ec1f2659e787\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/33f3503714b247ee80943688a5035d25 : SUCCESS in 1h 44m 29s","accounts_in_message":[],"_revision_number":3},{"id":"c4c6f93fb0791aaf5eb5d0b2a1a614ab4c47fed3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 09:50:21.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"85c6fb84dc0543c8f10733f1bcb6f0fb908dcd42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 12:20:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb088bc984b94611aabaefd3b948366d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d15088f3ee94d6fa7d49ca2c806ad90 : SUCCESS in 8m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0121846dadcf40288b17a569ece11214 : SUCCESS in 22m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ecdadc3b79644c6db47b59968bd480ab : SUCCESS in 20m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9701ed6e69f46ba825fb9d17a9574b0 : SUCCESS in 24m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/815638e2f66d4dd8a67045d84a95e44b : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/79dcb0c344cc4d1d8e93a3858e85b6b0 : SUCCESS in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac759d9ebdc640b5831448d01386e74b : SUCCESS in 23m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9074c9f10fc94a6c92622ce7af860a46 : SUCCESS in 20m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f6e947f59da04d4d80ffa2ed46333991 : SUCCESS in 35m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3b49f50b322e45a3aadebe55b1441b16 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/381e102fd0794ab5b5495f0790cb0e33 : SUCCESS in 19m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/81593bf9c55c45858cd87192f602ed29 : SUCCESS in 12m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/70a6965d101943edb3f1d5e75acf39c3 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/64df5a8945ec4c27abc2e0f34582212b : SUCCESS in 13m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/58eccf3e58a343f6a5795c584aa99a70 : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e317a4ac693447a0a1e9911091675f16 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb589b1d43ad4e9fb5cadc4e9fed7a14 : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/27564f41044442ec9028708e5aee7017 : SUCCESS in 6m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fed2686fa2ee4d04959a22646b694729 : SUCCESS in 52m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86fff64d9953426183ad8442a9dd7932 : SUCCESS in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce00b1dcde6f47848e1f4c87c7234fdf : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d6ac15c421a240acadde8e065b89ac58 : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":4},{"id":"f27b0714b296de3e98c228a0c6fae3c6297e7d95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 14:13:33.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"870559398b4c1f5d4e8f66e42065b8ac0506a83b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 14:34:50.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a8e08749cb3c9d6c737c60bb762817a815027d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 15:08:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94aa349fc6ed4dd7a557066bcab2dcb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b89d100d3884ab594708e19af180dca : SUCCESS in 5m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/520e4d344a9c44dfbdea1933c9c59799 : SUCCESS in 22m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/faccb6a510494acfac87166a00f89d50 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7ab10f28f304879b59738a57a2d95d4 : SUCCESS in 23m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/afa2d8364e444ac6a93a625ac66878c8 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6743dc212cd7497aa8c53299115f22d4 : SUCCESS in 21m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/39de8c4d160e4c90affac3a409c73e0c : SUCCESS in 20m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a1a62e669a274c3ba144431822bca8d6 : SUCCESS in 22m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4b016e1ec0404e5db042855d34b9bf0c : SUCCESS in 29m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/46b29516e76843328dc95296e05fdb6a : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f26d35f79d4f4943a84d9099eeed8f24 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3ade218c0acb4beb910ead813002a7e7 : SUCCESS in 18m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/37aa75e66f9844149a43aed1b6d01b73 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f97cd8e2928f41cc90950a09e63d8bd2 : SUCCESS in 13m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5d5cc8694f414affa98b427d96c12936 : SUCCESS in 15m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3af7101eb89746c98a3c76810240e2b0 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/daf3768b0cbd4e3bb2e33fc0175d6cec : SUCCESS in 13m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/921e5a6fb6fe4491b5c53c2a53b4ddc6 : SUCCESS in 7m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b78c063a8cfb4de5bc1e3d0de772866c : SUCCESS in 46m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2078d1c3501468b928c8b2876f84d4f : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb7602e5f45b4785be4e1e84d2e34d03 : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c71e56e66364128bbf6b6d33d42f3e9 : SUCCESS in 17m 56s","accounts_in_message":[],"_revision_number":5},{"id":"b127f14fcf20891a547e232653e2dee449bae53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 15:23:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d6d3585c78dae76d4fa4630ed79af7adaa05afc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 18:08:02.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9775eb528708490d8bcf444e30111b51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4fb107e29344b8aded04a850386c40 : SUCCESS in 9m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1822b983d79946a7aa240a5bcc5daa94 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d997073b3b724e32ac78286bc7576ee8 : SUCCESS in 16m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a67d6371a7d543c889bc627c358a497f : SUCCESS in 10m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ba2874706536483ea1caa54e3689d4ac : SUCCESS in 18m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5ea1a73791f84b0296a035f9fff322f1 : SUCCESS in 29m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/80ecb659d1184230bca0316125ae86c3 : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/057d8228e71f47328762dd4381c3e8b2 : SUCCESS in 19m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/758b9dcad2a44531984c5cce586c06b9 : SUCCESS in 17m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/124d1f936bee4294934486ffea31a89b : SUCCESS in 20m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3da09921008948569d5b7756bcb46ad7 : SUCCESS in 16m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a6ddb16099ea454599a85cf9309076a8 : SUCCESS in 17m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ec94d9242ca4304b1c0a28450a8b812 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3ef8d5c93cdf42a9a3ade6f9b04c3f2d : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc33ebbc5800464a8fab3ee775c9a542 : SUCCESS in 53m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c2ab06efeab4307804146e898e2ce80 : SUCCESS in 20m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/652c775c06ee4fbaa4cfb28101836b60 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d555477e1d343678ed6ebb196da4d74 : SUCCESS in 19m 10s","accounts_in_message":[],"_revision_number":5},{"id":"59a9aadc639ec6beefddab8f44e00903db91a763","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 18:08:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2326efaf237d8bc41ceacfc9344185e433bf6739","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 18:09:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b92dd7f4c394b69a7384adff4c797e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e2059939ad846a6a406bf7a8a6c4d05 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4a4c30efec164e14a9a501f213db3e6c : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"831966ab532e13eccb491d682208331c3d1c8b59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 19:51:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01bb4fc43f424da48047239fda1b614d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0f792850f373443daf196a9aa6d4c816 : SUCCESS in 2h 13m 04s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f35234193bafd24e414093589666f45ec31922ee","revisions":{"abec3d3113934f750d8c712a1524a3896692abeb":{"kind":"REWORK","_number":1,"created":"2024-11-19 13:12:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/935651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/935651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/935651/1"}}},"commit":{"parents":[{"commit":"677ff80144886263c7a2309d95df480c823b030e","subject":"mpu: detect bogus upload-id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/677ff80144886263c7a2309d95df480c823b030e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 12:10:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 13:05:01.000000000","tz":0},"subject":"mpu: copy sysmeta from create request to manifest","message":"mpu: copy sysmeta from create request to manifest\n\nOther middlewares such may add sysmeta headers to a createUpload\nrequest which must be persisted with the session object and later\ncopied to the manifest. For example, s3api middleware may add acls as\nsysmeta to the createUpload request.\n\nChange-Id: I249e9e32ecc1fb2817d1b253752c876fb2674369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abec3d3113934f750d8c712a1524a3896692abeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abec3d3113934f750d8c712a1524a3896692abeb"}]},"branch":"refs/heads/feature/mpu"},"20f17c75b6311846f8cc434ae16066a6f2a41e48":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-20 12:32:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/935651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/935651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/935651/2"}}},"commit":{"parents":[{"commit":"df9d51b9bb56fd9eb413a4af6dcc713afc6ef147","subject":"mpu: detect bogus upload-id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df9d51b9bb56fd9eb413a4af6dcc713afc6ef147"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 12:10:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 14:49:06.000000000","tz":0},"subject":"mpu: copy sysmeta from create request to manifest","message":"mpu: copy sysmeta from create request to manifest\n\nOther middlewares such may add sysmeta headers to a createUpload\nrequest which must be persisted with the session object and later\ncopied to the manifest. For example, s3api middleware may add acls as\nsysmeta to the createUpload request.\n\nChange-Id: I249e9e32ecc1fb2817d1b253752c876fb2674369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20f17c75b6311846f8cc434ae16066a6f2a41e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20f17c75b6311846f8cc434ae16066a6f2a41e48"}]},"branch":"refs/heads/feature/mpu"},"763afa0e2e8a40dcf699004b92cb348b1ee6bc37":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-11-20 15:43:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/935651/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/935651/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/935651/3"}}},"commit":{"parents":[{"commit":"fc7d7fdffe962d7f173da44057aa4d8bb7fb17c0","subject":"mpu: detect bogus upload-id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7d7fdffe962d7f173da44057aa4d8bb7fb17c0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 12:10:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-20 15:42:32.000000000","tz":0},"subject":"mpu: copy sysmeta from create request to manifest","message":"mpu: copy sysmeta from create request to manifest\n\nOther middlewares such may add sysmeta headers to a createUpload\nrequest which must be persisted with the session object and later\ncopied to the manifest. For example, s3api middleware may add acls as\nsysmeta to the createUpload request.\n\nChange-Id: I249e9e32ecc1fb2817d1b253752c876fb2674369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/763afa0e2e8a40dcf699004b92cb348b1ee6bc37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/763afa0e2e8a40dcf699004b92cb348b1ee6bc37"}]},"branch":"refs/heads/feature/mpu"},"a835b5a0b83cdfc9c95352814dc244e4151ef559":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-21 09:50:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/935651/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/935651/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/935651/4"}}},"commit":{"parents":[{"commit":"62ebd2c35672ddcbb58a63e4ba593818261cc29b","subject":"mpu: detect bogus upload-id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62ebd2c35672ddcbb58a63e4ba593818261cc29b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 12:10:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-21 09:45:11.000000000","tz":0},"subject":"mpu: copy sysmeta from create request to manifest","message":"mpu: copy sysmeta from create request to manifest\n\nOther middlewares such may add sysmeta headers to a createUpload\nrequest which must be persisted with the session object and later\ncopied to the manifest. For example, s3api middleware may add acls as\nsysmeta to the createUpload request.\n\nChange-Id: I249e9e32ecc1fb2817d1b253752c876fb2674369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a835b5a0b83cdfc9c95352814dc244e4151ef559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a835b5a0b83cdfc9c95352814dc244e4151ef559"}]},"branch":"refs/heads/feature/mpu"},"f35234193bafd24e414093589666f45ec31922ee":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-11-21 14:13:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/935651/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/935651/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/935651/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/935651/5"}}},"commit":{"parents":[{"commit":"10c5de5b892d172f90f71d97b39afe34fcd17072","subject":"mpu: detect bogus upload-id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10c5de5b892d172f90f71d97b39afe34fcd17072"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-19 12:10:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-21 13:33:13.000000000","tz":0},"subject":"mpu: copy sysmeta from create request to manifest","message":"mpu: copy sysmeta from create request to manifest\n\nOther middlewares such may add sysmeta headers to a createUpload\nrequest which must be persisted with the session object and later\ncopied to the manifest. For example, s3api middleware may add acls as\nsysmeta to the createUpload request.\n\nChange-Id: I249e9e32ecc1fb2817d1b253752c876fb2674369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f35234193bafd24e414093589666f45ec31922ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f35234193bafd24e414093589666f45ec31922ee"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
