)]}'
{"id":"openstack%2Fswift~935799","triplet_id":"openstack%2Fswift~feature%2Fmpu~I885ce60d28f8bb28ee0deaa263d2496573a324df","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-copy-user-headers","hashtags":[],"change_id":"I885ce60d28f8bb28ee0deaa263d2496573a324df","subject":"mpu: add placeholder for deleting hidden containers","status":"MERGED","created":"2024-11-20 14:39:05.000000000","updated":"2024-11-20 18:35:41.000000000","submitted":"2024-11-20 18:34:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"935799-p-mpu-copy-user-headers","meta_rev_id":"08c50f9fbd4cb6941520ab7c116d0e2ac46c5ebc","_number":935799,"virtual_id_number":935799,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 18:34:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-20 17:45:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-20 17:45:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-20 16:08:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5eae0308c6b6c3c95ca033f5f0b439044aa929cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-20 14:39:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0909e95b62e86ff47e9be6ed2421877f992029f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-20 15:21:09.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a3b3fdfd92be27a65be95c2684d86f224029ad25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 16:08:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc7c058765c743ab8a313ad1beaea0bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0bff408bfe44c59bbd10f6a1a283f08 : SUCCESS in 9m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e94b26874057413ab6af1b30b5468409 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3e8872a5f4c46bcbf57e38cde14917d : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd85099161e0461f9eef08115d89527c : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/afdeded1143b44c4833bbe702cf151db : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f2be08fec7934aeb861655591b5bf86c : SUCCESS in 21m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d49c11de646a4b75becb1ba356fda450 : SUCCESS in 22m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/958290ffae4c40a0889a7a41ebee4f85 : SUCCESS in 10m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/60cfc00a5f7a4d20aeb173315764e012 : SUCCESS in 22m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/212dfab9b2e24410ac8cbc9576d0b914 : SUCCESS in 9m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54f228a4548b42adb01915d98f483d13 : SUCCESS in 18m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/86de5d589a684f7c944466e23a852d11 : SUCCESS in 19m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ce24533b03e74babab5da9bb4bcd1bef : SUCCESS in 13m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2896cd9b1a0743a790f15f024a7a834a : SUCCESS in 9m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6025c43c927c4c19ba5ca09ce1e6a2eb : SUCCESS in 14m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7fbdcc75838a471390136822fce835ad : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1af1623fc56f487d84850e94cc57e7f9 : SUCCESS in 16m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9ff61733285648b5b590e81048e6934b : SUCCESS in 7m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/acaa519b80874d0e96148a3255a243fc : SUCCESS in 45m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5590f87e43ec4237b2180cbd5d10e4e5 : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfbc218c203e4cb5913bbc099edca779 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/113c2fa6eaf9488a8c8d151c75d5ab26 : SUCCESS in 11m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3c26d978ae5530a84ea33f973daa9083a8a37df6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 17:36:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e74715734e114fbbb75bd47469288ff2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/87e9a98f236b4c3d841a30ff29c7568c : SUCCESS in 2h 12m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4b7f70d4b9c3915aef83c77399ef295087951d50","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-20 17:45:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"327d68144eca08c00200294912225f8d4ebee1a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 17:45:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e94206830ab86346289ffd98f1cacb856dac3636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 18:34:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8862eae2d0041b88182021232e761d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df8385773d044b5abf2b1647ea944433 : SUCCESS in 9m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d3aab8d5175042de819e39a2b89ccfd5 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e214f77fe12240a4a740b970c49a3287 : SUCCESS in 22m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1cd3f8bd7e09409a8d03dd0d1c37f744 : SUCCESS in 20m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8f1268876ed64ffbbf3c80c2cbf9a9b4 : SUCCESS in 23m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1a4246125a77483c823266e183ff1e26 : SUCCESS in 22m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/30448db42c124cf39a3a277ab90e6f22 : SUCCESS in 20m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/180a2a2c644f420b8334cfc6709d725b : SUCCESS in 20m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f1f9cae5dbfa4bab8fddbfabab2c7078 : SUCCESS in 21m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dd8bf150e4e54d15a48dbb08f7a43dc0 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6dbea18f29ba4c72a6d521ae9cd097c1 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/abf47dc914544f5381fcd4982f66b914 : SUCCESS in 17m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/51e9704c8e6a424b928a33494ffbbebd : SUCCESS in 6m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6df19e25c1074570be3d60edb78fb47b : SUCCESS in 7m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/67cf203d9f8145818309b9736048c561 : SUCCESS in 33m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57517ea1c26045b999582e1e04cb1470 : SUCCESS in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87de780b7ddb4999a3880f383d96a424 : SUCCESS in 3m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c7dbf8889974f5493d58865c239f086 : SUCCESS in 20m 36s","accounts_in_message":[],"_revision_number":2},{"id":"aa3f1727c01f63962da551b35a7c736ae27c201a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 18:34:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"08c50f9fbd4cb6941520ab7c116d0e2ac46c5ebc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 18:35:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b4d00761a9643bd932f3508b78a81af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25a5eed6be1146d8a45257cf98705e2c : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/04cfc46ed56f4a1eb7c884d7ba79bbf2 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3dcf40af0445d86cc6040d414c070c974dbeae90","revisions":{"851039f280b4a73f4280f5f9f53dba3d5e8c4922":{"kind":"REWORK","_number":1,"created":"2024-11-20 14:39:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/99/935799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/935799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/935799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/935799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/935799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/935799/1"}}},"commit":{"parents":[{"commit":"efa8db397a4e303a07f659dd77214129dbf588c8","subject":"mpu: tolerate complete and abort mpu operations being retried","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/efa8db397a4e303a07f659dd77214129dbf588c8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-20 14:36:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-20 14:36:08.000000000","tz":0},"subject":"mpu: add placeholder for deleting hidden containers","message":"mpu: add placeholder for deleting hidden containers\n\nChange-Id: I885ce60d28f8bb28ee0deaa263d2496573a324df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/851039f280b4a73f4280f5f9f53dba3d5e8c4922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/851039f280b4a73f4280f5f9f53dba3d5e8c4922"}]},"branch":"refs/heads/feature/mpu"},"3dcf40af0445d86cc6040d414c070c974dbeae90":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-20 15:21:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/99/935799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/935799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/935799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/935799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/935799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/935799/2"}}},"commit":{"parents":[{"commit":"a62892c4db3e337e60849e9a31d6e84bb73c8936","subject":"Merge \"merge master to feature/mpu\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a62892c4db3e337e60849e9a31d6e84bb73c8936"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-20 14:36:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-20 15:20:58.000000000","tz":0},"subject":"mpu: add placeholder for deleting hidden containers","message":"mpu: add placeholder for deleting hidden containers\n\nChange-Id: I885ce60d28f8bb28ee0deaa263d2496573a324df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3dcf40af0445d86cc6040d414c070c974dbeae90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3dcf40af0445d86cc6040d414c070c974dbeae90"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
