)]}'
{"id":"openstack%2Fswift~935808","triplet_id":"openstack%2Fswift~master~I9d626db7e600ac468c99dabc89893719070d6996","project":"openstack/swift","branch":"master","topic":"cors-non-voting","hashtags":[],"change_id":"I9d626db7e600ac468c99dabc89893719070d6996","subject":"DNM: trying to zuul a little","status":"NEW","created":"2024-11-20 16:27:28.000000000","updated":"2024-11-20 21:05:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"860c8572797ccf61724c949da2cea3797ad0e91b","_number":935808,"virtual_id_number":935808,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-20 20:43:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-20 17:44:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebcb9752e5dff717199b5dacbb2a39ec57b8a11d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-20 16:27:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f968a4ee72cab00347cafedc31da34c6e7aa6c0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 17:44:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b034fe616f0469d98f555617d2b6ecb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9360015f934327a2a64a643b26ce23 : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/6c88f1ecafbc4565ada199d460d9e59a : SUCCESS in 53m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d808173a2a7f42f1ab343913799ffa27 : SUCCESS in 1h 10m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d97027238564fb59c1818735d22e531 : SUCCESS in 57m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fd59c5ff7b44bb699ce71177fd15a22 : SUCCESS in 1h 01m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a79da6e2d7334cd78c2471c3c1ef4a53 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5bf8e474b3b949938027f522647c0d65 : SUCCESS in 23m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afc5775b1bc34904925d62162b134d12 : SUCCESS in 20m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/62d1e240cd1c4b358db712e4295fa967 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/870593cdb91b4fe4aed15e02a3911817 : SUCCESS in 9m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00a922597f06450ab01a291f300d064f : SUCCESS in 20m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b03a95d9387d4f728f8682c7eb5867e8 : SUCCESS in 22m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ca5887d2fbc240c39dd61247b8b90207 : SUCCESS in 26m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1fc9f884b1f8401bbe62d6af20da7c1a : SUCCESS in 9m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/465040b049de421898055cc28d5c642f : SUCCESS in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/31a6395350ba4b7d8b9563159e1edce1 : SUCCESS in 18m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2958d9cd94074d0ca0bd2944075ded29 : SUCCESS in 9m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/faf1d7c16c3548c48d51860b847fc217 : SUCCESS in 13m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/200fc5f4466341fb8d3aa11497959691 : SUCCESS in 15m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7c0672af16a471cbc526f545a13046e : FAILURE in 5m 53s (non-voting)\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/98df715e82d94500a86fd61d05d64113 : SUCCESS in 14m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8058840b97084f45b092114abe1ab052 : SUCCESS in 4m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d81eda7933434761b1134bc5093fb261 : SUCCESS in 55m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dfcb0d6c81104010ae56ac3564f309f3 : SUCCESS in 41m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d1dcd205c606455f8e4c8ca740e2d087 : SUCCESS in 43m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/439de1cdcfe54e5db5b285bcbb04e87d : SUCCESS in 21m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e945eb48da224ac8a5aa984e80f53282 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/84ff398af3b94c66ab985efbe61adc73 : SUCCESS in 23m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12fa4af48c2d406497074469b93c439c : SUCCESS in 47m 47s","accounts_in_message":[],"_revision_number":1},{"id":"754a9bfd7a7d291a80698d9e52b8fbd114edefca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 18:41:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/930bb0aa524f42c9a37fc2cff9eebde8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84cd8c8ee8d945a29a9d5bd464cd6567 : SUCCESS in 1h 41m 35s","accounts_in_message":[],"_revision_number":1},{"id":"34b09268f2da9fcaac2b0648e7aca67b6522e28d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-20 19:29:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"15f2775906e28f28b0fe3b17ec65c42b0a608a56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:43:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7163fb47146b433f8e3748c1c24113e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336ee9a5dc7a4b4d891b92d89e186f23 : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/724a0fb001a642e282f7615d93f35464 : SUCCESS in 54m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bcd55ff145b54500bc4f970e01bf48de : SUCCESS in 57m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d1e291e2f884f5a8a034f9bcec1279e : SUCCESS in 43m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c1b8581347144fc8cc110735c539de6 : SUCCESS in 1h 07m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b13919c340a34f6888ff4feab1cbdd49 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a26095fb57e453bab2edd8fcf874402 : SUCCESS in 22m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e1b8307e33c480e8b1c948d2b619ed2 : SUCCESS in 22m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b808eb8c04ec4f8e8d2e66a2fd634ffc : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83fb45e846204dd9928ad9295d12d9d8 : SUCCESS in 19m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1d34aefb2ed240579f089db1c8da797f : SUCCESS in 8m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/84615e41685d4d829b7abe4a42c84b5f : SUCCESS in 23m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e323e0f551bc4c208f7f50d857ce114a : SUCCESS in 30m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/db5ce307283343029d34c946d6297100 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5062df3e247b4647abc24de3b754a100 : SUCCESS in 10m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60639425f0cd44e6a724b825378d1231 : SUCCESS in 18m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/76802180272342a085217d6aa7516167 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e497af3e819a465a87a991f08ec66f8a : SUCCESS in 15m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d71c56e0e999440eb1c00e1a597f2c96 : SUCCESS in 15m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fdaa6ebff3ac4ddd835dfe161da059a2 : FAILURE in 4m 57s (non-voting)\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df3b092d34d047f299ccb4355e9039f2 : SUCCESS in 14m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d0a42b2b0b34720bb1b86c9d977466b : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ede2d382a9149508a616909603c1ce7 : SUCCESS in 45m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1da7bfb6b972496fa90a5b00bb150abc : SUCCESS in 40m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b98150a799a49c6aa4b32ace6837f68 : SUCCESS in 37m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5baa714e3be24ed7ae88b9d3774b1f78 : SUCCESS in 15m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3b9d385a04242e8be2bd480b185a079 : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06d3e009fd1849ebb5b5cfcfe4d03273 : SUCCESS in 26m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73c5f362185445f7bd3685a9696449d3 : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":2},{"id":"860c8572797ccf61724c949da2cea3797ad0e91b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 21:05:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d38c16f359b44b30b8760effc1bda833\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/776b1ee06eaa48c383374af0d37b61f0 : SUCCESS in 1h 33m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e6eed262ad15c1f83f7964002573930fdd9e1367","revisions":{"5338310a4ae1bccadc2fdc46353ef3143f9b8361":{"kind":"REWORK","_number":1,"created":"2024-11-20 16:27:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/08/935808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/935808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/935808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/935808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/935808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/935808/1"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:26:57.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:26:57.000000000","tz":-360},"subject":"DNM: trying to zuul a little","message":"DNM: trying to zuul a little\n\nChange-Id: I9d626db7e600ac468c99dabc89893719070d6996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5338310a4ae1bccadc2fdc46353ef3143f9b8361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5338310a4ae1bccadc2fdc46353ef3143f9b8361"}]},"branch":"refs/heads/master"},"e6eed262ad15c1f83f7964002573930fdd9e1367":{"kind":"REWORK","_number":2,"created":"2024-11-20 19:29:01.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/08/935808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/935808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/935808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/935808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/935808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/935808/2"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:26:57.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 19:28:43.000000000","tz":-360},"subject":"DNM: trying to zuul a little","message":"DNM: trying to zuul a little\n\nChange-Id: I9d626db7e600ac468c99dabc89893719070d6996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6eed262ad15c1f83f7964002573930fdd9e1367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6eed262ad15c1f83f7964002573930fdd9e1367"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
