)]}'
{"id":"openstack%2Fswift~935809","triplet_id":"openstack%2Fswift~master~I56d82eb7467f5162a11427c8517833ff5f16c8d0","project":"openstack/swift","branch":"master","topic":"fix-cors-hostname","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-03-07 21:54:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I56d82eb7467f5162a11427c8517833ff5f16c8d0","subject":"cors: use same host for swift/s3api","status":"ABANDONED","created":"2024-11-20 16:29:50.000000000","updated":"2025-03-10 18:57:09.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5997543d0a326b63c1e4f2c6cb4c949e58a1481a","_number":935809,"virtual_id_number":935809,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-20 20:18:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-12-02 18:49:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-20 17:29:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-02 18:49:14.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"064e4c203fa00c8d566f47286cfc8ab78cb42f57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-20 16:29:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d76864bb6fe64f5c0105b2cdf40a85adb63f0c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 17:29:40.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc1d29513514456d820da3e4470c6434\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c22ed9a183a4624a01a8081ec34ee30 : SUCCESS in 8m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/076202ce98d843eca4f9d61e0c051b6c : SUCCESS in 59m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a9d886e5ee2f4af295a905a942899c18 : SUCCESS in 5m 18s (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b217aa4b3e834fd5b51a1d665a4c672d : FAILURE in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b399ad3201460091ac76be8cffe1d4 : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1bedf8a3ccb9a930e33939444dfeabfa5825e845","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 19:23:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80fddb5c9a4240708b1e365955362d37\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba94bc9149cf4246bee3ec28612bfb8d : SUCCESS in 2h 23m 11s","accounts_in_message":[],"_revision_number":1},{"id":"cc484260347b5544d36936d893c5c4c3a243507d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-20 19:27:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"714ee45759aa968dc5c4f01fe743f00523ec29e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:18:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e52d7bad1213482e90605a31adb1ecef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9e454f4beea4b18a1e15659f3f34341 : SUCCESS in 7m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/908cac5827fa49efbf2d5fafbe8b1c6e : SUCCESS in 49m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ddbebbbf85ad433c99fc8558ded5f59b : SUCCESS in 5m 13s (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/18c78690184047fbb5702e3360468a47 : FAILURE in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/722142d24e5646eda02b906900b11d36 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2536f6bafe926d49dac078cc6b055cb3bd383743","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 21:08:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a54f1a6e6d534b9095987de5d5513d9b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/83d0c7877edd4f3d8c57d69aa3e27607 : SUCCESS in 1h 37m 47s","accounts_in_message":[],"_revision_number":2},{"id":"cc5fc798426e572be870980d114bbaa170a1511c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-02 18:49:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"89911505326e7dbdbabc88b0fb32b7eed9f72615","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-07 01:10:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5e35dc86ecf461e8a9139c6fb11962976b10460","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-07 21:54:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"5997543d0a326b63c1e4f2c6cb4c949e58a1481a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-10 18:57:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"487ac41d6d9f3ae938fc3d1fa3e14c282ff03411","revisions":{"087c7b1d48062b4d591dc51b58d6ee3214d28393":{"kind":"REWORK","_number":1,"created":"2024-11-20 16:29:50.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/09/935809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/935809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/935809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/935809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/935809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/935809/1"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:29:25.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:29:32.000000000","tz":-360},"subject":"cors: use same host for swift/s3api","message":"cors: use same host for swift/s3api\n\nChange-Id: I56d82eb7467f5162a11427c8517833ff5f16c8d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/087c7b1d48062b4d591dc51b58d6ee3214d28393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/087c7b1d48062b4d591dc51b58d6ee3214d28393"}]},"branch":"refs/heads/master"},"487ac41d6d9f3ae938fc3d1fa3e14c282ff03411":{"kind":"REWORK","_number":2,"created":"2024-11-20 19:27:54.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/09/935809/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/935809/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/935809/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/935809/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/935809/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/935809/2"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:29:25.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 19:27:35.000000000","tz":-360},"subject":"cors: use same host for swift/s3api","message":"cors: use same host for swift/s3api\n\nChange-Id: I56d82eb7467f5162a11427c8517833ff5f16c8d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/487ac41d6d9f3ae938fc3d1fa3e14c282ff03411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/487ac41d6d9f3ae938fc3d1fa3e14c282ff03411"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
