)]}'
{"id":"openstack%2Fswift~935814","triplet_id":"openstack%2Fswift~master~Idc4cffee7dc15a84706a292bb3738e091db87412","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"37271":{"account":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"last_update":"2024-11-20 19:22:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idc4cffee7dc15a84706a292bb3738e091db87412","subject":"Make OldestAsyncPendingTracker timestamp float","status":"ABANDONED","created":"2024-11-20 17:14:45.000000000","updated":"2024-11-20 19:22:06.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e3b3f562855081980f56833fd8c6c4134cc1ba04","_number":935814,"virtual_id_number":935814,"owner":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-20 18:26:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":-1,"date":"2024-11-20 18:07:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-20 18:07:08.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2024-11-20 18:26:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2daa2e3518379e5a4aa8f9c28c0c0cb26a1f0f69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-11-20 17:14:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f49af86d40e202b5455d4bfc9a4a4921754e05e8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-20 18:07:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5457926eab0bca2236d80c470baee0c9d83f9e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 18:26:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be91dc88d8cd4d498c36d6a6a8915246\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e38103442ab4fe3835a191dd1961286 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/7cadad8066c34cd1898f3d4336db1ed9 : SUCCESS in 35m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8dd7d4c8721b45a0b881c359d10b2cc5 : SUCCESS in 42m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/453ff9d4698b42a9a6c23e00c731ee7a : SUCCESS in 50m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd0b3ecf2ab741aca6d7f8b1269a12ae : SUCCESS in 1h 03m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40aa3a4db5084e319857accd2937a778 : SUCCESS in 12m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ae32165317c48ed8c8ef331657afdfe : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb74c82e7e6a4f8aad1048d924dae876 : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1da86d27ce8747bf9ee1e2e32e7f0be7 : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f39db2ac6b264dd9b22b0d663c7ee03d : SUCCESS in 9m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df2ac9082c704f82820059700d98d5dd : SUCCESS in 18m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1ee68b7f6d7745449ccf1a2b334201be : SUCCESS in 17m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2183de67b1164e7c80e6978c550e9cfc : SUCCESS in 24m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9405698707764debb61764e36ecaf5c5 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1b7eaf08ee884e3e8e0457f48cf451e0 : SUCCESS in 15m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e878ccc2d79143b2a65c1ace7bc9434a : SUCCESS in 10m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fdc40acc2c404854bfd395e7e523f0ab : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e8109a603ef44b82801931eba125cd42 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/467296c0982c482d8f39951da26d6271 : SUCCESS in 13m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/245ee1b88cd844b78553009f3495d08b : FAILURE in 9m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/60c1811eb88a47b091efe0c3ff6f5c57 : SUCCESS in 12m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4be3575635c74aad8003cc1e7a47c028 : SUCCESS in 6m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e44aa4bb6e0c4e5882add4c8425f6687 : SUCCESS in 49m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50547da7191144899bdabe6569ea6754 : SUCCESS in 53m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5840a5487a98483f9ba9abc182954eb6 : SUCCESS in 53m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbe338e0f5474dc0849f22fba50023f7 : SUCCESS in 20m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa03e01cf40e4c67aecc8d88f249b09d : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03aa9891722741ccb83a62fe911d1b7c : SUCCESS in 18m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58004d7d9aff438caf3c6571b6490692 : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e3b3f562855081980f56833fd8c6c4134cc1ba04","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-11-20 19:22:06.000000000","message":"Abandoned\n\nMade copy here (doesn\u0027t use main branch (https://review.opendev.org/c/openstack/swift/+/935822))","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6319d5f63cc4b6f902f9e65bbd3c38a718661765","revisions":{"6319d5f63cc4b6f902f9e65bbd3c38a718661765":{"kind":"REWORK","_number":1,"created":"2024-11-20 17:14:45.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/14/935814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/935814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/935814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/935814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/935814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/935814/1"}}},"commit":{"parents":[{"commit":"0a5348eb480e0b9ff0abc9acea183d94a97d700e","subject":"Add oldest failed async pending tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a5348eb480e0b9ff0abc9acea183d94a97d700e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-11-20 17:14:29.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-11-20 17:14:29.000000000","tz":-480},"subject":"Make OldestAsyncPendingTracker timestamp float","message":"Make OldestAsyncPendingTracker timestamp float\n\nCurrently the OldestAsyncPendingTracker timestamp is a string, it should be a float.\n\nChange-Id: Idc4cffee7dc15a84706a292bb3738e091db87412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6319d5f63cc4b6f902f9e65bbd3c38a718661765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6319d5f63cc4b6f902f9e65bbd3c38a718661765"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
