)]}'
{"id":"openstack%2Fswift~935887","triplet_id":"openstack%2Fswift~master~I10083aa121db5ef86e9e07fe371ba69e5b64a5dc","project":"openstack/swift","branch":"master","topic":"fix-head-content-type","hashtags":[],"change_id":"I10083aa121db5ef86e9e07fe371ba69e5b64a5dc","subject":"proxy_logging: unit test for content-length header insertion","status":"NEW","created":"2024-11-21 12:48:27.000000000","updated":"2024-11-21 17:15:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"da50083179c26bf2d592fac80d2d20dd21a25888","_number":935887,"virtual_id_number":935887,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-21 14:03:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-21 14:03:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01258a4783d06459a0546443bf081d6f88e946b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 12:48:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d15e0f3023cf5e502da1a3ef82799ef2b1fc5fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 14:03:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05f0f964715345daaecc973d13b7ca42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845a3dc317be4db48da71a571f80a3d9 : SUCCESS in 9m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/502f7f2b59ad4eb5aee2b06b25c4ca1f : SUCCESS in 1h 06m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fcf0c99f79944e2ab77a3b7bece4474f : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d560ef0f9bab4c7cbeb9bf085fc6010d : SUCCESS in 25m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9191f37f2be24ec087c2c7adaf19d0d1 : SUCCESS in 16m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4741ee80dbb432c89299354e4361e49 : SUCCESS in 7m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1aaad544bda6409eb250cd1b378f4837 : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/619b18a824f649fe8c31ecdeb8d21b73 : SUCCESS in 18m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/13826c1d20db4a9a830fb8e7fec30dcc : SUCCESS in 20m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/25df738adec14c67a415ff78af0ebd1b : SUCCESS in 34m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/73d346e1e11f440383033e1c21cd5b74 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4f461e13aede43bdba8dcba4edca9c8d : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe05f9752f034efca72166833d9bf0d9 : SUCCESS in 19m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4823ff81b6674a1a8d44072f8f45b96f : SUCCESS in 12m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/063ac0b59ad84312ba9fe41ffcd9e636 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e3269ac2343f4981b9b62dc88cef2cfe : SUCCESS in 18m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b6c01b9ae9f45c8a992200fe639bd5c : SUCCESS in 51m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b786d82d22024a5ba265917953ea2180 : SUCCESS in 41m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fec425eec60e4b5fb974631efd1f4db7 : SUCCESS in 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cdf843b09194a66aa1599d3b7a92ac7 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/290e0c4180bf47c8bdbbc11411d2372e : SUCCESS in 18m 19s","accounts_in_message":[],"_revision_number":1},{"id":"da50083179c26bf2d592fac80d2d20dd21a25888","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 17:15:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9d442e958004c9e956b9a54bdf6108b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e06cd0f014a64f24a69ff062272ce42f : SUCCESS in 2h 27m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c05195cffc2030d4501d4da10b35bdb3469af43","revisions":{"3c05195cffc2030d4501d4da10b35bdb3469af43":{"kind":"REWORK","_number":1,"created":"2024-11-21 12:48:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/87/935887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/935887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/935887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/935887/1"}}},"commit":{"parents":[{"commit":"074ea7fb589ce55ff22c565de62058e0632841ce","subject":"sq? more tests for listing-formats HEAD handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/074ea7fb589ce55ff22c565de62058e0632841ce"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-21 12:41:12.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-21 12:46:17.000000000","tz":0},"subject":"proxy_logging: unit test for content-length header insertion","message":"proxy_logging: unit test for content-length header insertion\n\nChange-Id: I10083aa121db5ef86e9e07fe371ba69e5b64a5dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c05195cffc2030d4501d4da10b35bdb3469af43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c05195cffc2030d4501d4da10b35bdb3469af43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
