)]}'
{"id":"openstack%2Fswift~935888","triplet_id":"openstack%2Fswift~master~If1ef7b046bfe2c4ac6910992485e10b6045872cb","project":"openstack/swift","branch":"master","topic":"fix-head-content-type","hashtags":[],"change_id":"If1ef7b046bfe2c4ac6910992485e10b6045872cb","subject":"DNM: demonstrate how proxy-logging fixes container HEAD response","status":"ABANDONED","created":"2024-11-21 12:48:27.000000000","updated":"2024-11-22 12:31:18.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"42ed95cebe1a4eedc6a8bce9ba693d17831105fc","_number":935888,"virtual_id_number":935888,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-21 14:11:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2024-11-21 15:02:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-21 14:11:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d48c4c3b1ce4d480c9c98d8f02b6710a4efedf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 12:48:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8481cbe9b212b3613d481d427734ffc968e43659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 14:11:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e62475db4314029a7d1b266aa1a87c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921539cf76d3405388e2c76b6d36f803 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/9a42a38b7f1e4f5a9e8fe5c72c25c879 : SUCCESS in 50m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8999fffeede457ab19e6cadec6526f1 : SUCCESS in 52m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0014873dd6fd4bc588b336037640336d : SUCCESS in 54m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d80de4be09c47f2a62dcbcedc9e34b1 : SUCCESS in 1h 02m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e164aa03bfb46a394c309c348da9d87 : SUCCESS in 21m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0130c37a4fc74a968da2e752330a5322 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd8724bdeffa4bbcbbfa090cf56f5f6e : SUCCESS in 18m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e1ae224eb0146198b28b76cc2c1de00 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/07a8976ca3bf4255bbc8bbe4150e60a0 : SUCCESS in 20m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/650927addd444e5e991101c575d4da43 : SUCCESS in 19m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bb51ef6a80904d93836d4bcc2f0108a0 : SUCCESS in 21m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/94d2206412b14803a3175d5b5fb5ae26 : SUCCESS in 30m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bb23a1ebb1c4448b8b70168820c58659 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ec7c0f0d7044479bb044a350710679ba : SUCCESS in 17m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/90561073b8b1435ba8fa8122e2a64ac0 : SUCCESS in 17m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c823c733f41440da97aae5fae4ae56a3 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c108418274c746ae97c8ceff61c3a146 : SUCCESS in 14m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0904519f57b14db7baec59a196ab88a4 : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6050465d0ddb44d982f74101d7dea953 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8dc1cd1d18194b95bf0bdb14f7b8d09c : SUCCESS in 13m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7862163c2306446789998656cde4ff5f : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f30b1be263c4187b2b7f6c47d3ea302 : SUCCESS in 51m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0670715ca7f44255a5b799340faabcf1 : SUCCESS in 1h 01m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89682baa7abe4afebf33706302546d44 : SUCCESS in 1h 00m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2a019511b8740a69d23beebf9b1bada : SUCCESS in 22m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59fae0ed27544426aebf336343b0c4dc : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a144f806f9214648b65eda393ec6c810 : SUCCESS in 19m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e9c24bd0c214c719a93f079b5cb7bbc : SUCCESS in 1h 08m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bd82dbd19f50952a4b9ab53469dc1174b4d03bca","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-21 15:02:40.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f83154c702e9e4f2f43cf3297bbdf931151ca299","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 16:44:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a3f156d516647789dac59392c91da95\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c349d375879b41f88fefc0577d544304 : SUCCESS in 1h 55m 47s","accounts_in_message":[],"_revision_number":1},{"id":"42ed95cebe1a4eedc6a8bce9ba693d17831105fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-22 12:31:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98acd73f0b1dd33b8ee52db4bc8cde4328c132e2","revisions":{"98acd73f0b1dd33b8ee52db4bc8cde4328c132e2":{"kind":"REWORK","_number":1,"created":"2024-11-21 12:48:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/88/935888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/935888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/935888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/935888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/935888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/935888/1"}}},"commit":{"parents":[{"commit":"3c05195cffc2030d4501d4da10b35bdb3469af43","subject":"proxy_logging: unit test for content-length header insertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c05195cffc2030d4501d4da10b35bdb3469af43"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-21 12:43:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-21 12:46:17.000000000","tz":0},"subject":"DNM: demonstrate how proxy-logging fixes container HEAD response","message":"DNM: demonstrate how proxy-logging fixes container HEAD response\n\nmodify proxy-logging to show how it is weirdly responsible for a\ncontainer HEAD response having a content-length header.\n\nChange-Id: If1ef7b046bfe2c4ac6910992485e10b6045872cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98acd73f0b1dd33b8ee52db4bc8cde4328c132e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98acd73f0b1dd33b8ee52db4bc8cde4328c132e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
