)]}'
{"id":"openstack%2Fswift~935998","triplet_id":"openstack%2Fswift~master~I17e34957a225dd976d5a52c960ec23f0a66bd384","project":"openstack/swift","branch":"master","topic":"fix-head-content-type","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-11-22 21:29:53.000000000","reason":"Change was abandoned"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-22 21:29:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I17e34957a225dd976d5a52c960ec23f0a66bd384","subject":"sq? Cleanup title code by using HeadKeyDict","status":"ABANDONED","created":"2024-11-22 07:08:07.000000000","updated":"2024-11-22 21:29:53.000000000","total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"caff8234d40c91a814f397a90374053c48c70b73","_number":935998,"virtual_id_number":935998,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-22 13:17:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-11-22 12:28:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2024-11-22 18:49:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-22 08:18:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-22 12:17:03.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-11-22 12:28:39.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-11-22 18:49:31.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"7e138295c71719efdd95dc72db723c8cdc2ea060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-11-22 07:08:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d89229e7a8d7bb60e21719cf43e162cef33325c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 08:18:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e54497796bd44919c8c3f5d4d00c5a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1116f0aaf5464bb3b231c9fdb0ca68db : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/797527a14b0d4a18880b194e40861cc3 : SUCCESS in 1h 01m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b01bf7abdc404018a4cb8ef22c616720 : SUCCESS in 1h 04m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1cc367aee62349e5921db8f9c92b6727 : SUCCESS in 37m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d32b3fe724b345d1a2a8b5f2a53ca71c : SUCCESS in 48m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f84067e61784056b7f69a1142220e20 : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8812d5c58a724c6e824a955831c84bed : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b619393f7dbc42bfac9e961f9d27819c : SUCCESS in 18m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cd0695b7ad2d4ff4802e2f6b0fc5ca24 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/70b8f7ae088144ada21779d72e5e2e0a : SUCCESS in 15m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/99ff4d133e89423990ffe7abe04ab4cd : SUCCESS in 18m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5ad21a4c09934e6b8834bd6549f80662 : SUCCESS in 16m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ff9a4381a5704de0b369298a8f7153a2 : SUCCESS in 30m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0f5888d103be46989b27da44cc0e47a4 : SUCCESS in 13m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2a50a74fdd1b4ae3803ff53a68ceac86 : SUCCESS in 17m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89fdd2527126400cb57423a14af134c5 : SUCCESS in 19m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5e253edd1eab4dc8baf1b32dda9f95ff : SUCCESS in 8m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/661777b8058c4588b7812f4a5d12d071 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8723b5a5f554471b93b7bff604c8d061 : SUCCESS in 14m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e6bd3f98ee14b93b823f2d2b91d9cbe : SUCCESS in 4m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9f61259b1151470499c57968bfd75da9 : SUCCESS in 14m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/46c1002216bf4afeb3b55a8dfd680deb : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/84aeee2899c141d6b622dd09cc182ad2 : SUCCESS in 56m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/981bec3ad452497897b4eac6dff3d12c : SUCCESS in 52m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a97535d44c8d405dbfb43752ebbfb0dd : SUCCESS in 28m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/896f041d68f1426a8b20c50b7d79c8b8 : SUCCESS in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a26b3bf727c450d9e6ddb177d5ba91e : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ade677d55b34736958eecf530450b9a : SUCCESS in 21m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44f97a3b606a4c528502f08d996a90ad : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":1},{"id":"66dedd0d20b9e2a626cc85039c022e8de0b7ddc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 10:05:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/293c8f77af3348129f35756373fb9ae6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/24794876536745baae3894ef38ef0eee : SUCCESS in 2h 03m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c2039a8aa0cdef657e20548ef4a1242f3ec13689","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-22 12:16:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"84b2d2db196651383b5c26b83ab8dfc8a92125c9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-22 12:17:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb74127c38136d9fe9052e8ce39d5ce6ecb03905","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-22 12:28:39.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"91175286616e1bda666452eb6cc5886e110acc29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 13:17:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/859a9435c8674cc7a1b73d3015490a95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a88be202cdc141ebbbf4e02140c8abbb : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/eeb55f087d9a4e3e8e774d70e08d01f0 : SUCCESS in 53m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b8ca7225443409bb2e08ea5ed442531 : SUCCESS in 54m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/57a1170fc5b14c219388aac90da55981 : SUCCESS in 34m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f3f62905a534438992a3949a5af43ad : FAILURE in 36m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b8005bbed0d45c09d1149cf93f4c2e9 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2bd4dfc4f99243499e2099fbbdb54a45 : SUCCESS in 21m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/751e319660ad49b08dd2528dd1b8f36d : SUCCESS in 12m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e76c493121d64b7c9ce0659448f9fce8 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a89efdaf6fa4de6a3518b7c39f18bf3 : SUCCESS in 21m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8eb9d556bbd84081ae5063d6b1889acc : SUCCESS in 19m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/712c20e721f94c24b0970578deb0a2d1 : SUCCESS in 21m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/80228bd9b96241ff927b981bdf78bc25 : SUCCESS in 29m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/416d7e95e78846f7b46c7d450b7c5aaf : SUCCESS in 20m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bb10a78ceafb4ee68ae55d6a24014259 : SUCCESS in 17m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7a108b372f0646bc8524c4b02e0132c4 : SUCCESS in 11m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dbd6e8e77fb045e4b95fc6f0ccd68769 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/79b19a02a1fb4492aa5b77c54739a698 : SUCCESS in 9m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c78fef2f7e7a40bc80d6a1283400f75b : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/745df7ed9dcc4678af471ba823c72d94 : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6c35210a690849a89e9d66a6118f1aef : SUCCESS in 9m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e4566aef1f304137adb630dd50671cc1 : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5aa3e7a43f9c4c08bbbd993dcd9a06ca : SUCCESS in 54m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a4d52de624b74485ab7f81222832efe1 : SUCCESS in 52m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/88839944afad4774bd0991e882a78839 : SUCCESS in 55m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3d15d516d414499a6bf585196bb71e4 : SUCCESS in 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bd1d9c9831f4172bc30ae608b0c47ab : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7aeb8862e8b84b69942b37296e18cc5d : SUCCESS in 17m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f57c81256e7d4193826cfcfdeb0092bd : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b2830f4fc2cc88554423e1a6bd17b5c94e6acca3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 13:48:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/957288ae9c99420a982c6bda7ecab4c8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4ef0d95e46c34fb48f2a0f13085a3116 : SUCCESS in 1h 30m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ae0a0792f0111a7ddfc89565bdde7d49a58db138","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-22 18:49:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"caff8234d40c91a814f397a90374053c48c70b73","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-22 21:29:53.000000000","message":"Abandoned\n\nsquashed into 935817: Ensure correct content-type in container HEAD response | https://review.opendev.org/c/openstack/swift/+/935817","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9008e860f35f3bf2ce273002b1b0cf142101660d","revisions":{"eed46a6d68f0cf5aa569eeb73bd22b877e5de810":{"kind":"REWORK","_number":1,"created":"2024-11-22 07:08:07.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/98/935998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/935998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/935998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/935998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/935998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/935998/1"}}},"commit":{"parents":[{"commit":"68c6437ec085ec95e5a2c94395090ac1a4609fa8","subject":"Ensure correct content-type in container HEAD response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68c6437ec085ec95e5a2c94395090ac1a4609fa8"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-11-22 07:07:20.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-11-22 07:07:20.000000000","tz":660},"subject":"sq? Cleanup title code by using HeadKeyDict","message":"sq? Cleanup title code by using HeadKeyDict\n\nChange-Id: I17e34957a225dd976d5a52c960ec23f0a66bd384\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eed46a6d68f0cf5aa569eeb73bd22b877e5de810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eed46a6d68f0cf5aa569eeb73bd22b877e5de810"}]},"branch":"refs/heads/master"},"9008e860f35f3bf2ce273002b1b0cf142101660d":{"kind":"REWORK","_number":2,"created":"2024-11-22 12:16:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/98/935998/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/935998/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/935998/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/935998/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/935998/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/935998/2"}}},"commit":{"parents":[{"commit":"d3c4edbb106c2aecf5b79efb644d1a5915550e66","subject":"Ensure correct content-type in container HEAD response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3c4edbb106c2aecf5b79efb644d1a5915550e66"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-11-22 07:07:20.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-22 12:15:57.000000000","tz":0},"subject":"sq? Cleanup title code by using HeadKeyDict","message":"sq? Cleanup title code by using HeadKeyDict\n\nChange-Id: I17e34957a225dd976d5a52c960ec23f0a66bd384\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9008e860f35f3bf2ce273002b1b0cf142101660d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9008e860f35f3bf2ce273002b1b0cf142101660d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
