)]}'
{"id":"openstack%2Fswift~936029","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ifc2e7d7698ceb971c61bc2935879266706407dbe","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-cross-compat-tests","hashtags":[],"change_id":"Ifc2e7d7698ceb971c61bc2935879266706407dbe","subject":"mpu: return 404 early for list parts on aborted mpu","status":"MERGED","created":"2024-11-22 15:44:40.000000000","updated":"2024-11-25 18:19:45.000000000","submitted":"2024-11-25 18:18:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936029-p-mpu-cross-compat-tests","meta_rev_id":"66f40cf44de5209c372d869c83e624cc1805db74","_number":936029,"virtual_id_number":936029,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-25 18:18:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-11-25 15:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-11-25 15:50:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-22 17:00:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da71def5b6fbb2727de69a4aa7290ef902e91ddf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-22 15:44:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b2a909f9b70a09c00c56728facb9625091e8aa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 17:00:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a84b95190eda4fa08846d00a7f39cab8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c977945744b41d9b50a1bb050de6baf : SUCCESS in 10m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0614ffe7ff6499ca0b8837340302e11 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c77d3f5703c473e8f56c9bdea8d7978 : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d59d02c664241009bd14fcce5899b8e : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7bc73dfdc8d040a2aed62d12e702a99e : FAILURE in 3m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/09829dd79a934ba2b6af9c52cab46eaf : SUCCESS in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1b988dcc3b1049e89d4593a24fa177e4 : SUCCESS in 18m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1eb896f822e64d5d812bc20c5cda648e : SUCCESS in 23m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9597b51b9a444ecd9243219eccc1e9fe : SUCCESS in 30m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e885167c942f4866bed44c7561e52431 : SUCCESS in 10m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b2918fdeaf59451d86cd0dc605fcc5be : SUCCESS in 20m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/215414d807ed4671b18b5da2d9c5fdb7 : SUCCESS in 17m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/affdce7a25a5466a804700bb083ffa17 : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ffd7115159fa4372b44713a3e4c500aa : SUCCESS in 9m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/476c025a5d2e4639ac0f845231fbd768 : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4f519c9f8ef54f7ab48cc348c0f51930 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4460cc7edefd49ffb07c9ec7bd108c95 : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a6d2c08d93444b1ac06c534ad7dbc19 : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d8d00219c98423fb3182bc01e6d49df : SUCCESS in 1h 01m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63b31bff5bb648bebf3b6d8947a65c88 : SUCCESS in 25m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b72d38694063425a9acbf5dae46a66e0 : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62223afe8e3d483fb77b7a74a91b245d : SUCCESS in 24m 29s","accounts_in_message":[],"_revision_number":1},{"id":"75ba478e22a5cecfb592bea57e050d1c19fc1337","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 21:51:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6b3b768337141bcb061a3efecb8157c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/681f456920374d56835f8e51dbad80c8 : SUCCESS in 1h 50m 58s","accounts_in_message":[],"_revision_number":1},{"id":"20776157c271b5c6c27377b9c5e7d22afa89d5b1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-25 15:50:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e10d3f23c8b82f58be92f88ccd86153ee12b6909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 15:50:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0902064cf56366dde561a560df1273023a40541a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 18:18:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4b525b5eeb14844bd7dd20875739b7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2802ad2396ee46e7a9187f7056552f85 : SUCCESS in 10m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3a93f94d31c14eeca8fe6ed006581ce5 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5d23a41ec3744aa2b4e337a92b0a5c25 : SUCCESS in 21m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/11b607d71e134ff8aef8c944038f0713 : SUCCESS in 10m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e8ef2df655ef474eb15abace0235e5ce : SUCCESS in 21m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c892bbea7b314f29a66af55d53b2b6a2 : SUCCESS in 29m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/288059ea5426454b9860850a680556f0 : SUCCESS in 10m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b2fa0e5dfe6a4b4f859efad55e13cffc : SUCCESS in 19m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cff800a60b8c4338b2f8d5969bfb1561 : SUCCESS in 11m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1c613c736a20432ba00efcf85a2ab02f : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e1735431e3ac440fa6b355d3c6fff015 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dabf75d5760b43a5a9e2ac37c73ddb49 : SUCCESS in 17m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b537e9dfd2a4f43be530fb88b93f30e : SUCCESS in 8m 53s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/70d044bac814449da58f0caed037537a : SUCCESS in 5m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c29a2ce80f4441b49a7c9f9f2bcc0846 : SUCCESS in 57m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bf146a5b0604450bb8dd4f126bb8c68 : SUCCESS in 18m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33e3aec04280449c9648dd03d2f61fad : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8e2db269215b4e92a3d7d0b7aa0a36b9 : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7b63e3748a6d08bbedd41347af91549f34066df3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 18:18:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"66f40cf44de5209c372d869c83e624cc1805db74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 18:19:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08a9bd8d47164a5e9a9ec379e71e9f28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcfc340c1df34a11a89d1b821af545a3 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/81559f416e7943c6aee0ff79a52b0e3e : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b7f427d6e3158cb118e73836b65d1a7f2c5dfe8","revisions":{"1b7f427d6e3158cb118e73836b65d1a7f2c5dfe8":{"kind":"REWORK","_number":1,"created":"2024-11-22 15:44:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/936029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/936029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/936029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/936029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/936029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/936029/1"}}},"commit":{"parents":[{"commit":"f35234193bafd24e414093589666f45ec31922ee","subject":"mpu: copy sysmeta from create request to manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f35234193bafd24e414093589666f45ec31922ee"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-22 15:33:11.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-22 15:34:22.000000000","tz":0},"subject":"mpu: return 404 early for list parts on aborted mpu","message":"mpu: return 404 early for list parts on aborted mpu\n\nChange-Id: Ifc2e7d7698ceb971c61bc2935879266706407dbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b7f427d6e3158cb118e73836b65d1a7f2c5dfe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b7f427d6e3158cb118e73836b65d1a7f2c5dfe8"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
