)]}'
{"id":"openstack%2Fswift~936035","triplet_id":"openstack%2Fswift~master~Ia03af1680c6230658473c0c8d444efb5bb805f58","project":"openstack/swift","branch":"master","topic":"p-s3api-mpu-compat-tests","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-03 18:09:55.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-03 16:58:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ia03af1680c6230658473c0c8d444efb5bb805f58","subject":"s3api: add more MPU cross-compat tests","status":"MERGED","created":"2024-11-22 16:01:01.000000000","updated":"2024-12-03 19:22:42.000000000","submitted":"2024-12-03 18:09:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"936035-p-s3api-mpu-compat-tests","meta_rev_id":"1254a659b9ba26deadfae3697291ab53b4c30bb8","_number":936035,"virtual_id_number":936035,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-03 18:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-03 16:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-03 16:58:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-22 17:05:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-02 22:58:36.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-12-02 22:58:36.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-12-02 23:33:38.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"65478848aa13e8b3cab4aa55ab6a8be7a7ec2b5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-22 16:01:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75151f9b3acc4e8f3fa72e394079e69653403339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 17:05:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8a6701378cc404ca0b550aba35b759f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1bd7b3da964cec93562a12d01e2f65 : SUCCESS in 8m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/be4baced848a48889af248a7d564552b : SUCCESS in 52m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/22593241d2714af380fd44a562b5ae39 : SUCCESS in 6m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b6d147817c534bdf924bd8c38c291f4b : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb20e641d3844a88980c777a55febdb2 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0a2e1c29f11e7316b39ec5d5685b255cb04fbc3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 21:58:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f88b24b825d453597b741d46e64111b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/edaab38f2c0749d383c9e8f5214936ff : SUCCESS in 1h 34m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c076c9c0befb8e161ae2997b5654196bea5bc013","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-02 23:33:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7d7075b50ee56856ade334b4ef5cd8950a5ecc93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 12:50:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"76f8da328e676fabb8024503cdbc2eb0b449e823","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 12:50:57.000000000","message":"Patch Set 1:\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f9d812cfef62af692da72866acfa6925f879a288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 12:55:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1e8cfda160afaa103d61cb8c2c0b7575a1b0d0b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 13:29:04.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76489e3fc1b34925b0e2abbc5069100e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85c38170845f4adba6feff8c33e16a37 : SUCCESS in 5m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2b843e63ef04304a43337d6973b9f35 : SUCCESS in 32m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3d21e2ef40d94985823b56f648b76525 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0d3357d165c1434ab8dd1e1276a171d6 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be443d450e684166b2ceae803ed19d5f : FAILURE in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"6461897901223832463dce07786beb30234ae3cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 14:09:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3d7395948a74b459a5a80e1586687d7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53e456a32ca644749f33291df5cea8df : SUCCESS in 1h 12m 24s","accounts_in_message":[],"_revision_number":3},{"id":"6414ce1d1db3ac8fc29733fc9e2cad61f461acf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 14:28:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"57d0d5b9de02a634a0cf0b7b50ccbcf75fce2634","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 15:34:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bd16d599ba148e39091a6876ebb790d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b2cd44297ad64e2083ac8fd9b18857c6 : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":4},{"id":"800b8f334bba7e9aa588e219c20c8041e4b95076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 15:44:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfb5dfb10de24877a8b14f795299f173\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35031b56cbb4bcebcb149c4b2a8f956 : SUCCESS in 4m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/50e774f8dfa94802baa7508ebbd5600f : SUCCESS in 1h 05m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49e3e403867a45e193e7a23201b3c960 : SUCCESS in 4m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d49cc1eeeb024c40af433e923b18296b : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f5f0119588487592ad73adf09c92d2 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":4},{"id":"99c7b977beefa086bde5d102807f6fe1c6113028","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-03 16:58:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d022df1676a1122a2c58702f03aff776dfc022f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 16:59:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"04b6560ffe21de71ee1ca8e7e71f4ff273bda8cf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-03 17:12:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"50c6ea35ccac5a889efd4178185c4ad1745a513f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:09:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9b57563634142249738ba910f6031be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d200a5bbabd4424c8edaa6aeccff3924 : SUCCESS in 5m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85da5a4538aa4155bb2ce1ca402ba9b3 : SUCCESS in 47m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9d815ffaf2ba452b830e3b1658f4c698 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/73c6a2aa43c445ea9f7a17e26162d0c6 : SUCCESS in 16m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4c58025449441e087840d4ec1a76e29 : SUCCESS in 15m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/53123a330524415e84be6c0eeb1990f5 : SUCCESS in 19m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/19e2deb108a340c38f841ff80eef876e : SUCCESS in 21m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d7ee412a02cc4e6e8a6a5a0690cb78ad : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aff3f2e62f21406a86287c11dd2ce26d : SUCCESS in 17m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/441beaa7dcb74f36a3d31078e0ad4ddc : SUCCESS in 18m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b50b4c97af68469caa0d35043c9f5c70 : SUCCESS in 12m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a50f6cf1d19b4484a2500b5caeed0d46 : SUCCESS in 14m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/05c2832dc8fb46c6bd62f1622f6cc646 : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c646594786ea47beb306af463745be19 : SUCCESS in 7m 30s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9d98af3b250442bbb3256fa5a303bb31 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b9036e61ab8492f9e24eab85ba8c306 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":4},{"id":"9ee0a35af21ec75524bba8c742b382f02cbe20d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:09:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fe76a324dac1c5496074fe01f0db475d1cd20008","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:12:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9dcf565c9184c37b21f3296f77f1934\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb6bc90e10e42beb3e14570188b2652 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dfc12bc72ca44c57ade17109ce6b184b : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1254a659b9ba26deadfae3697291ab53b4c30bb8","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-03 19:22:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"eac4ffd7a9ce571a865e70b5b79f24921757507b","revisions":{"df602d326dfe1ee656e638d055fe4516f0bdcd3f":{"kind":"REWORK","_number":1,"created":"2024-11-22 16:01:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/936035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/936035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/936035/1"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-05 11:05:43.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-22 16:00:48.000000000","tz":0},"subject":"s3api: add more MPU cross-compat tests","message":"s3api: add more MPU cross-compat tests\n\nChange-Id: Ia03af1680c6230658473c0c8d444efb5bb805f58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df602d326dfe1ee656e638d055fe4516f0bdcd3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df602d326dfe1ee656e638d055fe4516f0bdcd3f"}]},"branch":"refs/heads/master"},"e06c357f8c56ac6a3c565e9c6e4a5d6feb074d3c":{"kind":"REWORK","_number":2,"created":"2024-12-03 12:50:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/936035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/936035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/936035/2"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-05 11:05:43.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-03 12:50:36.000000000","tz":0},"subject":"s3api: add more MPU cross-compat tests","message":"s3api: add more MPU cross-compat tests\n\nChange-Id: Ia03af1680c6230658473c0c8d444efb5bb805f58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e06c357f8c56ac6a3c565e9c6e4a5d6feb074d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e06c357f8c56ac6a3c565e9c6e4a5d6feb074d3c"}]},"branch":"refs/heads/master"},"1357640cfa3d2a562d456a9d9d4b629f519ba2e9":{"kind":"REWORK","_number":3,"created":"2024-12-03 12:55:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/936035/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/936035/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/936035/3"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-05 11:05:43.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-03 12:55:26.000000000","tz":0},"subject":"s3api: add more MPU cross-compat tests","message":"s3api: add more MPU cross-compat tests\n\nChange-Id: Ia03af1680c6230658473c0c8d444efb5bb805f58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1357640cfa3d2a562d456a9d9d4b629f519ba2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1357640cfa3d2a562d456a9d9d4b629f519ba2e9"}]},"branch":"refs/heads/master"},"eac4ffd7a9ce571a865e70b5b79f24921757507b":{"kind":"REWORK","_number":4,"created":"2024-12-03 14:28:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/936035/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/936035/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/936035/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/936035/4"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-05 11:05:43.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-03 14:28:30.000000000","tz":0},"subject":"s3api: add more MPU cross-compat tests","message":"s3api: add more MPU cross-compat tests\n\nChange-Id: Ia03af1680c6230658473c0c8d444efb5bb805f58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eac4ffd7a9ce571a865e70b5b79f24921757507b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eac4ffd7a9ce571a865e70b5b79f24921757507b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
