)]}'
{"id":"openstack%2Fswift~936053","triplet_id":"openstack%2Fswift~master~I8b590512502ec71532dfea9af17ee6785436e325","project":"openstack/swift","branch":"master","topic":"fix-head-content-type","hashtags":[],"change_id":"I8b590512502ec71532dfea9af17ee6785436e325","subject":"sq? test for title case","status":"ABANDONED","created":"2024-11-22 18:12:21.000000000","updated":"2024-11-22 21:29:42.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"5224c8b8df9b807b7f7ba48cf1f978a1d6f5a5a6","_number":936053,"virtual_id_number":936053,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-22 19:28:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-22 19:28:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"244508b27f76d9caae59765186d2dd57aa285e67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-22 18:12:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee96badfed805f122dce4c183e47faf4abfee86c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-22 18:21:41.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ee56b2e7bf0e81dc6c09df34bbbcee88e8e56450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 19:28:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14fdec7900db4968924c7eef7f489832\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37abdb6d190549e888759ed81c46c6a9 : SUCCESS in 7m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b0c6f834ea044619e1e8b9f9a838e7a : SUCCESS in 59m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79c727a8bbc244ecbb36960a0fecf31a : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32ab2f87f2ad4725a0a02781ddd4b141 : SUCCESS in 21m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca8c43d5b5774db9aa356815fa9e44cb : SUCCESS in 22m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/df2057f2ad6745e7ac37ad8493230414 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/516b867f543f4e5181c33b409e18be3b : SUCCESS in 17m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6fde1cd8c78548c5b9090dc10aed480f : SUCCESS in 20m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a32034444c654b4b883571eb89f30522 : SUCCESS in 23m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a7f677da7aa14985806c3cc4cc67eb23 : SUCCESS in 21m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d3a604d22766440fb7a0c9df0053babd : SUCCESS in 21m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/467ea60660264de6b3ac0effce1aaec8 : SUCCESS in 22m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/05b05398189d4d70bdbb5ccec95eab68 : SUCCESS in 19m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2b4707011d2e4b4aafe1e0b9bcbf12c5 : SUCCESS in 18m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/076204e948004b11b88a70c85b964810 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d3df4c75388c43b28a0b2f455c603360 : SUCCESS in 10m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/501c6b463cad4bcbab2722109d5a9796 : SUCCESS in 52m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79a7009bc76c4499bb4f53ccca4841cb : SUCCESS in 51m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c64faf91ad9f4d169b712fba9dcbb034 : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02e0a0b859714dd384dc58eeb5de8922 : SUCCESS in 3m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d837cd3ee804ee29a4e971948b863eb : SUCCESS in 21m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5224c8b8df9b807b7f7ba48cf1f978a1d6f5a5a6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-22 21:29:42.000000000","message":"Abandoned\n\nsquashed into 935817: Ensure correct content-type in container HEAD response | https://review.opendev.org/c/openstack/swift/+/935817","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07993ac50fd3c8bd6c4550473f4ced79977cf3dc","revisions":{"07993ac50fd3c8bd6c4550473f4ced79977cf3dc":{"kind":"REWORK","_number":1,"created":"2024-11-22 18:12:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/53/936053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/936053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/936053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/936053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/936053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/936053/1"}}},"commit":{"parents":[{"commit":"d3c4edbb106c2aecf5b79efb644d1a5915550e66","subject":"Ensure correct content-type in container HEAD response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3c4edbb106c2aecf5b79efb644d1a5915550e66"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-22 18:12:11.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-22 18:12:11.000000000","tz":0},"subject":"sq? test for title case","message":"sq? test for title case\n\nChange-Id: I8b590512502ec71532dfea9af17ee6785436e325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07993ac50fd3c8bd6c4550473f4ced79977cf3dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07993ac50fd3c8bd6c4550473f4ced79977cf3dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
