)]}'
{"id":"openstack%2Fswift~936066","triplet_id":"openstack%2Fswift~master~Idacc59c5f43367926eff5221ee7fc417a9bc2d50","project":"openstack/swift","branch":"master","topic":"base-server-gate-fix","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-11-25 18:35:45.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-25 18:35:45.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-11-25 17:24:54.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"Idacc59c5f43367926eff5221ee7fc417a9bc2d50","subject":"Ensure Content-Length in backend container/account HEAD response","status":"MERGED","created":"2024-11-22 20:39:49.000000000","updated":"2024-11-25 18:36:48.000000000","submitted":"2024-11-25 18:35:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"936066-base-server-gate-fix","meta_rev_id":"351976d274f8fd94d4e9d8e3c9fc5388ed72007b","_number":936066,"virtual_id_number":936066,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-25 18:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":1,"date":"2024-11-25 11:35:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2024-11-25 17:24:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2024-11-25 17:24:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-22 21:43:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-22 23:47:52.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2024-11-25 11:13:26.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"3f41dee8e493dfe09d38e40ba0ad4e11e7ed056b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-22 20:39:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fa2d2177d85ab5fbe1a117f5f89db70ff543ca9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-22 21:10:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"052d3246284c11135b5638ed38bbe91aa4ae2892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 21:43:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89c7d2f3db15404c90435f4ff30e9d25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe1f7ffb4c14335bd49ea1b340a0e78 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/5de6c953fa8d445a8e8b530b93cd9b74 : SUCCESS in 1h 02m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b5f0cdb5407f49bba4e93961e83d9e8e : SUCCESS in 43m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb8dade88c434eb6a2c02ad6fd08c7f6 : SUCCESS in 47m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed3c0944317e4696be79293920889028 : SUCCESS in 48m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bf794973570401eb97be2f9b30baa8b : SUCCESS in 12m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/539bac2009254f1aabc11dfac46bad1b : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9207e120eb54a27bc09593b6a20cd1e : SUCCESS in 12m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1013539504a547adb43a547fd4f335dd : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/60e681125b0a429cb2d355575dcd0055 : SUCCESS in 19m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/702e83d7773f4d3da8a97fefd10e8681 : SUCCESS in 22m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e2287a4d8b6340c3be79b55abc4f96e1 : SUCCESS in 19m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/604e5e372bbf409f81250932c094c350 : SUCCESS in 30m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ab84916b33c9497885dbb0d217125f58 : SUCCESS in 21m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a1f775fcfbb94b70af9d22222d7757aa : SUCCESS in 17m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/821f1801637e4f04bd15bbe46dcb00c8 : SUCCESS in 18m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/69110b16154845fa94b79c8b403566dc : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f6e0302549224cdd893484967aeee280 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e51829498b574d908327ccf9382533c3 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c5d1780e054845ab97eef961cb908dfd : SUCCESS in 4m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/825fb6b633e5443d92aceb1bd64dd629 : SUCCESS in 14m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7fb3adf258e24e1aa7936af1ea413af0 : SUCCESS in 7m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c56577e46e1d410aabc8f04e4b993a28 : SUCCESS in 55m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/515fc7b31b2d42cab434e25f45a097b6 : SUCCESS in 42m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e4495dfef8324f89ae473fbae740d4ce : SUCCESS in 52m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c71d963f22f4d1c90fcce185d5a9e01 : SUCCESS in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f84e17aa84d44e9bd24f747e9966725 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74e2f725e2894617823eae466bdf864a : SUCCESS in 19m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9f5323f52204d9ea3568676c932753b : SUCCESS in 47m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c2a0e9b53112c66e5c0c9f728354cdd8d541de60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 22:58:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5bde6046c1f4a52996b331758059a1d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56bb766be131435d8b7c6bb7b5a64969 : SUCCESS in 1h 10m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ba677f8828632d6c228f1b053b96456b809bdad3","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-22 23:47:52.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"eef1ecbdfa6b86f1e36127c8ec3e9c51b43ac41c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-25 11:13:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9ff91e6063a7ca34ad47ad58c94adebd2ad812ac","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-25 11:16:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b6fd9c67cfb6e256957a8df5a79ff54bbbfb8b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-25 11:26:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f12143ad5af046b9720a05f49894ecdcb3333dfb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-25 11:35:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6bdb56b9998538ffa2303c5f93a7b0f0031bea9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 12:51:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a5d2f4532644999a32bb3f93e78ad76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1341101c0a47f1836f207aba86470c : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/6cc93b07510a4bb9b8c53936adef8528 : SUCCESS in 56m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/35986c05869f4cfdaa2f8962006d8987 : SUCCESS in 53m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e0027b790c14e3997863e3a9844f985 : SUCCESS in 54m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc8b3082a6d340b3ab576e552f24ad8c : SUCCESS in 1h 06m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/753a4ccb432444588ac8bd2efde4521e : SUCCESS in 21m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2f4f9d096934b3496d935d4277c7917 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7520cbd343a64461a4b6c113920b89ca : SUCCESS in 20m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba88fdddd64347e08cd87bf24c2e649b : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5afc1283d32a4e9396b203657640e675 : SUCCESS in 19m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fdf9b3ec66db4c89b94efe1688c2f99a : SUCCESS in 17m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6e5ce4a34f8e4a4ba893f5181733860e : SUCCESS in 22m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/17d32f6dd02544458d250f2507a0ed4e : SUCCESS in 29m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c10310bf05f6497cbb00340a1914435a : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1fd4d96528944d2fb5fff4049df2b016 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4d363f8a99124c26b43cbf74819c8158 : SUCCESS in 16m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7ad731c0650341559c55673fddb7c9c0 : SUCCESS in 13m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a36fdc214df447a2bfafe9a33eb33100 : SUCCESS in 16m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1cfe2d5da04a46f0b79aaeb811f4560d : SUCCESS in 14m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1232bcc282e241138b09b7a1ad9bc203 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4e113b9e0b6044adb737bde710dc0fd8 : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e0505eeda1bc424a842d8697266cdca7 : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e4fada876fd4d928b989a8b7716895c : SUCCESS in 48m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5074b9fb8eec4300b8f9dc5eb6bd8114 : SUCCESS in 1h 02m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ca80da632c6e4a30b7ca549e092d805b : SUCCESS in 52m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4211ffc03f974592adca6f0c8fef6070 : SUCCESS in 18m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57bc3b19889743d391eb10fe0623017b : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9aad5f8a667d40598c51eab7f11147a9 : SUCCESS in 16m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec9e050e6b394c72aa5e9ce3fb9550a9 : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":2},{"id":"6fbd0675df9bd02c3b63ba5877cf9704d52702a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 14:01:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/588efcdd6bdd4871aa03481b7ad0fb9b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/31bb2924e97a4ddba23278c1506339d0 : SUCCESS in 1h 31m 06s","accounts_in_message":[],"_revision_number":2},{"id":"25856f226fe68bc35440c9585ffb46509f45f32d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-25 17:24:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e98f0688bfc2caa9026476550da5b90b3f12be02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 17:25:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ede964b97c9a5c6694da0f4b90ec007e39321879","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 18:35:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35deb5941da1473fb129afa10c6fbce8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f0078100604aea8773920dd69e18f7 : SUCCESS in 5m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/eede2aee0d3e485e90074dc5a29268d6 : SUCCESS in 48m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/378f2978625843cd92ec2950d42f5551 : SUCCESS in 55m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0e0f3ec45764064ade92bec6fdfe787 : SUCCESS in 49m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d09fefb06f224513b2964e494becdd89 : SUCCESS in 50m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/159cb5fd7d29412cb890ef90942f5ab1 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a28e2cce6c0c4bbe85001323843f2fe6 : SUCCESS in 17m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/310e78e9d51d4d17bc5bbf28f998de11 : SUCCESS in 16m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/688f68044b2b4835b6cf243b0fcf71a5 : SUCCESS in 25m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c4a98e18bf074775894c514f352bd524 : SUCCESS in 25m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6ea3d536e8bc456faa99fa58e9cc4da3 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b9dd209f8c2f4010abaa912c29435ad6 : SUCCESS in 19m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee4721f3cb554a2bbf6a9b5d0cb46a3e : SUCCESS in 16m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2fb1c45cfdaa4d7e9eb7b0a029a4468b : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/209bd32fb99f40b5b6f71080f7b96396 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c1544b3b5fdc4e9087d0857792578ad8 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b4c9cd06da44e3083db9627c53f1e06 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c6f2ebf9982e4ae9a2dfb16b9440bad1 : SUCCESS in 9m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/913d3903663a4866888274a1d01e30eb : SUCCESS in 56m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1fe625757dde42b29ddbebd1c0693b07 : SUCCESS in 53m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e91577d63b0547538871bac853c89bfb : SUCCESS in 58m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b335b4162b17405eb0575ca7bd59d061 : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33232e6db69d43caaf4aa6dc261d1406 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dda45af6dd25429181b5ecb751941697 : SUCCESS in 25m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c20885af2a684190a8d8cad5bde52010 : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":2},{"id":"a09eec9a7b87c458ca72986b152c59830dc81332","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 18:35:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"351976d274f8fd94d4e9d8e3c9fc5388ed72007b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 18:36:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/503aa02e56be4fa1bb983ca43e64866e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d34488380bfc47189434fa1e6bb3a873 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bb1c5553c01d470a99394695d7d51b90 : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4aadb54025085d6bf6c3ca71320004fd50667849","revisions":{"1f0844f417f1b66f3dbc5d46689e00d6a3a0f8d9":{"kind":"REWORK","_number":1,"created":"2024-11-22 20:39:49.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/66/936066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/936066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/936066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/936066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/936066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/936066/1"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:59:43.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-22 20:29:25.000000000","tz":-360},"subject":"Ensure correct content-type in container HEAD response","message":"Ensure correct content-type in container HEAD response\n\nA failing CORS test in the gate discovered that we were responding\napplication/json to ?format\u003dtxt requests (which is maybe not even a\nvalid value for that qs param?), but only when running with\neventlet\u003d\u003d0.38.0\n\nThis directly fixes the problem of backend account and container server\nHEADs no longer having \u0027Content-Length: 0\u0027 by adding it explicitly\n(which is ok for 204)\n\nChange-Id: Idacc59c5f43367926eff5221ee7fc417a9bc2d50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f0844f417f1b66f3dbc5d46689e00d6a3a0f8d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f0844f417f1b66f3dbc5d46689e00d6a3a0f8d9"}]},"branch":"refs/heads/master"},"4aadb54025085d6bf6c3ca71320004fd50667849":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-11-25 11:26:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/66/936066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/936066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/936066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/936066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/936066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/936066/2"}}},"commit":{"parents":[{"commit":"fa889358acf7675efba2de095777886b4c1ff7f8","subject":"Ensure correct content-type in container HEAD response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa889358acf7675efba2de095777886b4c1ff7f8"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:59:43.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-25 11:13:42.000000000","tz":0},"subject":"Ensure Content-Length in backend container/account HEAD response","message":"Ensure Content-Length in backend container/account HEAD response\n\nA failing CORS test in the gate discovered that, when running with\neventlet\u003d\u003d0.38.0, container and account HEAD requests returned\nContent-Type of application/json to clients regardless of the requested\nformat. This was due to the backend HEAD response no longer having a\nContent-Length header, causing the listing_formats middleware to not\nmodify the returned Content-Type (see Related-Change).\n\nThe Related-Change fixed the client facing issue by making\nlisting_formats middleware ensure the correct Content-Type is returned\nto clients even when Content-Length is absent in the backend\nresponse. The Related-Change also ensured that the 204 response to\nclients always has a Content-Length header.\n\nThis patch directly fixes the problem of backend account and container\nserver HEADs no longer having \u0027Content-Length: 0\u0027 by adding it\nexplicitly. This violates the RFC prohibition of a 204 response having\na Content-Length header [1], but preserves Swift\u0027s historic behavior\nand is consistent with the proxy-server\u0027s 204 response to clients.\n\n[1] https://httpwg.org/specs/rfc7230.html#header.content-length\nRelated-Change: If724485e1425d1481d10b9255436301e346f07e8\n\nChange-Id: Idacc59c5f43367926eff5221ee7fc417a9bc2d50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4aadb54025085d6bf6c3ca71320004fd50667849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4aadb54025085d6bf6c3ca71320004fd50667849"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
