)]}'
{"id":"openstack%2Fswift~936137","triplet_id":"openstack%2Fswift~unmaintained%2F2023.1~I64bd376d40f42ca9e330c406bac4ded77c20e4eb","project":"openstack/swift","branch":"unmaintained/2023.1","topic":"create-2023.1","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-10 14:03:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I64bd376d40f42ca9e330c406bac4ded77c20e4eb","subject":"Update .gitreview for unmaintained/2023.1","status":"MERGED","created":"2024-11-25 10:49:04.000000000","updated":"2024-12-10 14:05:28.000000000","submitted":"2024-12-10 14:03:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936137-create-2023.1","meta_rev_id":"c533cad276963255b64d8bb7b9ae449c18b5e80a","_number":936137,"virtual_id_number":936137,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 14:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-11-25 17:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-11-25 17:59:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-25 11:17:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-25 17:59:43.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-12-02 17:29:30.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"106f63663587af3ef1b1e4ebddea95af46d17462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-11-25 10:49:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd24514261ecdb3de39fd14780a73074e00956e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 11:17:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c83c961ca8849029446334e939491bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00dc882b5dad481dbe00eb852022a992 : SUCCESS in 8m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b9fcf73dbb6453184d3a2ef7fadf854 : SUCCESS in 12m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dfde6efe842d4f57bf6490347a184b40 : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/285ab49a7cf64b90903f8fcc5f052198 : SUCCESS in 17m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e3040c3216714365bdd0abe9a593285f : SUCCESS in 7m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f583fac0bb264c14adc6958d1b2d30b0 : SUCCESS in 14m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6e150f6523c446a5afe134c83ad96f3f : SUCCESS in 17m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/391193b1fe6a4fa3a01ce328772662b9 : SUCCESS in 18m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c073127af249d49aa06fd25d72c189 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"36c8cc426834af92d11f890ece57c7fbeadb63cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 12:49:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c959e5fd2ff4eeebf4a0a4f3e0834b5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/59c920c974e949e8992f14de1f263369 : SUCCESS in 1h 15m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c145ce8f5ca548f09445409332a025a6 : SUCCESS in 58m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/62df79bf9e6b492ba10a0ee963097460 : SUCCESS in 1h 04m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d23895b40e164e73809332b1d533004b : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":1},{"id":"74a1f6e861ffcab69aeefb4b606dcd39f2f2489b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-11-25 17:59:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b08ee6747a1d1312d4e5378fc852a9b0ee787f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 18:00:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a303edd64e8fe305eae27a248b2b79c3e33a5ca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 19:35:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e9eade312a3491b8864e1060e20352e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69f0f7f91a04d95b12abc6f5685ae2a : SUCCESS in 8m 09s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/851440429a334c80a4094f1a42c7e94d : SUCCESS in 12m 02s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/78558ae082ce4d7fb0770764ac801b4a : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7f69188587594827930dab344fc4dd22 : SUCCESS in 8m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0f5b96d25df1428390725170ddd6385d : SUCCESS in 8m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/74ce630e5a134af99aa5805bcd27a76f : SUCCESS in 16m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b8861f8247bb432890de4fdf9f40f7dd : SUCCESS in 16m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b4b5da3c3cf94bbfb61ecef02d67f829 : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce2db6ab9b1c454f812a6604eef8cf07 : SUCCESS in 10m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a1d0600d70664a589a05ffedb143daa1 : SUCCESS in 15m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/69a0d5361e2b4d91a08fa9bf6481a747 : SUCCESS in 20m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/20d7be5a75c64bf49f94d91af39f7e1d : SUCCESS in 14m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7f70e77ead134110875d1ace8fc79a78 : SUCCESS in 15m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6b74c2588a9345ba87f0e7c3df2b230e : FAILURE in 6m 24s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c967cbcf46f941e9b37ea60cbcc9ee9f : SUCCESS in 6m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cab1784efe624fc2941ba0858332ba19 : SUCCESS in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81cf25d527474610bdb502bb19359223 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4e2bdd32c7ca237c81b80ce5761f7ee9baf86c7f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-11-26 11:41:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efcc65b41431988b3e1ae30e071b9a3b2da6e94f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 12:04:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/631338b3f4df4954b940139856c7a5c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b39bf7736e48909501b2b616cfa8d1 : SUCCESS in 7m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49f2c8b62bfd4ebd85847127060a9825 : FAILURE in 5m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aefe94e2700c4262a9e73d78f50d77df : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94143357a86c43f9b3ed27d37c2c0351 : SUCCESS in 19m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19e986cb901041e79500d5d4093599e2 : SUCCESS in 14m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e2a95f5284e24db1a96ecbcf2e4c1f37 : SUCCESS in 14m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6562e958b9064467a28767706deb9ec2 : SUCCESS in 18m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf0efe3fea7d46dfb5ec040ec6cc5863 : SUCCESS in 17m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c16c12cb804bd08d45a5086d553907 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"eb6f4eb1de2d970e61f07951a1b15471aa2a825f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 12:05:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"521611472998596e8c1a458a1753d63626f3cbe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 13:08:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4ffe86878d544f9b9123117e686ac21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3406856f4db04eeab6649b920a96a153 : SUCCESS in 8m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a90fe7fa3afe4c3dba0fb98d0a344d6a : SUCCESS in 13m 01s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e49583116993441d9aa5240b4a741ddd : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/90ba06e28a2448c7856ed52a16c1ecae : SUCCESS in 17m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2236079fe7a74b599bdf7e5504ac0bd5 : SUCCESS in 8m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/334e2df90de94f9da13ab6149d883483 : SUCCESS in 15m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1d4f37a68dd742939fa1071254282f2f : SUCCESS in 15m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/000a102fa4cf4fb38372a583b83814d9 : SUCCESS in 10m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b4c32b6d3c2e4534a74fe127a6d738a3 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4683d79e94d54141a7ea96856836f0cf : SUCCESS in 17m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/43bd915a228142aaada3dd1a119f9a5d : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ba3d7ae59ea1441fbe7bf0a1788d185e : SUCCESS in 16m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/02d7c59de0d64b72aecfd9eafccf3790 : SUCCESS in 8m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a9c13057c6e74ca284cb623d3e94f72d : FAILURE in 7m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3cf6e0bcf6f24577a7f17176cf54c3de : SUCCESS in 9m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/377c36aa4362477e804a7028ab714383 : SUCCESS in 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7413b5b211414e0fb14e5e184e3dfd0d : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"89476b34d476060bc78a1e01b019058d40ee206f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 14:20:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39e3f9b471b744ff8e3a1a1aabcaf8a6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f8cdbd38c184d8da1bf65c02ca1a35b : TIMED_OUT in 1h 30m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cadb5cc07fb74a078a346f8ab4aa9e1e : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b2202e0d9f848db935d030727ce86dc : SUCCESS in 1h 13m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4465299832c4d20a93a0b841d3f5dd6 : SUCCESS in 1h 20m 11s","accounts_in_message":[],"_revision_number":1},{"id":"402cd3777caa02ed2d09f22af0acbac72b7d3620","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-11-26 14:48:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf1ffe2a33c783e0158da84cedb099c979629e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 15:41:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dccfad8cab1d415bb5929105e2e7e67c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7942da333e624bb8beee7db95cc0f384 : SUCCESS in 7m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/176dec62162444cf8e4ce478d810dc7f : SUCCESS in 10m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f045379cbc6d4d8ca412db3dfb3fa570 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/252e568f7c0b4e8189d193920c7dc98e : SUCCESS in 20m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84fcfd497ad445f3a9ec9ae9b32dbf6c : SUCCESS in 13m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aac7efadb9db4173bcc8d46f7632eb3c : SUCCESS in 16m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/08bb31b4131a4478a24f618d2f698b50 : SUCCESS in 11m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b3ae711de624c58a57bc1ddda5a204b : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c943124bb9148ff9e7ef887fd0b6e34 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e0544963e46a2db556c055d9dd3c3882c2bac832","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 15:42:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5be64bb47d7c240be55c91aa7a232a7cc9f38c49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:22:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a363c649d8954667a13f9f8a34aa328a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e22a6ae58a0418b8a2cc03db294cd09 : TIMED_OUT in 1h 31m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/228fc3c63f3f41eebc01831e5e975a6e : SUCCESS in 1h 11m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/334194d9541143e78bcca8898d20fcdb : TIMED_OUT in 1h 31m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b8a328cfa6944ec8cf817221998866d : SUCCESS in 1h 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e8644d06afc487f9b9ad21d02780265958e22e40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 18:16:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b01691be6264c2bb179956f2c9f9681\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a607c6280bd54fd6913b5563a5a5ccfb : SUCCESS in 7m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/fe2dced2a1a54e189386510fd5276075 : SUCCESS in 11m 59s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6b32e4d90b6945f6b9837b4277f5462d : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7924dbd885fd4a66a41e4d30cf31f563 : SUCCESS in 14m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c34395a4299c4494a9b30a7d6a2581af : SUCCESS in 19m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a8f2c3f2bd774d2280377c937571de4e : SUCCESS in 13m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8e9194cde78b42bebe9b32b01ab99d4d : SUCCESS in 17m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e78cc33cbf914787a8c0526fef7764eb : SUCCESS in 20m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a81f6a221a20441ab0dc17440a9125ed : SUCCESS in 17m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e94dd270db0444cb362ca2ecc0f8c05 : SUCCESS in 19m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d20cf118df2c4c34a04e58f8dabf7f84 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4105efb24a0c4feea9987c5d80c23472 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c2a02322ce9e4f01a6688a90640dfc0f : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fd69177c18b1457d9171af0e0e532344 : FAILURE in 8m 18s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f9b2f89053ee4c09acc4a2a8e74bbda7 : SUCCESS in 9m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7863cb6e41940469139f635b113b4da : SUCCESS in 17m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/289ff313366d49448dcc0d8f180cbc71 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"03cee055faca043f75daef97870905e1f3cbe166","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-02 17:29:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0fb3293e7fd6ea3c38462c07eda2b57c41c08cd2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-10 12:18:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"784fe0214439b22be722d510da9f3cd027b4b06b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-10 12:18:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c8d3000f18fe42a1f6980e2f55dbe9c642883c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 12:40:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf6f7be7224f4fcd9e10442b0dab5d86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c47375ded94ad691d4dd5ca1f54e94 : SUCCESS in 9m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/57ee21e3724a4a138e4b7b2b7acdd604 : SUCCESS in 7m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/be1feeae582e48d483499f7b61640d76 : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0496170cf1174021897c003ac14eb94a : SUCCESS in 15m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d3538bb42e5422fa76a7173f2bdf89e : SUCCESS in 19m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c63de2da2ce74a39829ca0098ee79ea6 : SUCCESS in 15m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/139c442a419741499a1e5b72fbde499b : SUCCESS in 15m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efe76e03cda94387a66466746849d976 : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d14b4c30776464eb6763f55e121c54b : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4543f1cb2200e845978af80dc1c50285b889d4e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 12:43:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"366d600f337555d476d0b1e108d3f94387db59f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 13:55:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eadef76bd4b4937b041aef67c9e96c2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc9fa7b91e6c4657890d8df8e1a4d50e : TIMED_OUT in 1h 30m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4db0abe906a547859c39a4736ae6f0d2 : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8964da12d33b495cab40bbf0bdacccd4 : SUCCESS in 1h 23m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b99a9241ca3454f8d0219bd49609ecd : TIMED_OUT in 1h 31m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f45e9819ad8a556a6eefb80f01916a5796a45e05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 14:03:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13832b2fe77c43ebb577b60e425c22d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9500459ea86a4603945ae17e1cfef4dc : SUCCESS in 9m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/723edae47341449e825cab7f9cab74a4 : SUCCESS in 13m 47s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fe638338c9564e989b1685117342f8ac : SUCCESS in 8m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4cbda5374b9d423f9c829e739212d890 : SUCCESS in 12m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/13105c20c6904c1a9eff94d6f5942418 : SUCCESS in 16m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e71b1fa7234245a4803d49014cf54fc3 : SUCCESS in 15m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/038a258e02d747dfb53c1c27eb063041 : SUCCESS in 13m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/91870da3b3c44243853ece838d4416bc : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5093d3dddeb14bda94bae275fa0b9e21 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/174f6908e2cb4d7a9011c30c7f81c197 : SUCCESS in 17m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/01e143dfaed74e90b7ae3ab1c9423684 : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8f85bf205ecd46e48750a61817a97bf0 : SUCCESS in 10m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/55f13981878342a4a397a19952faa911 : SUCCESS in 16m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/021398f5d7504d99b8f676c3721cb5af : SUCCESS in 6m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8096168ea03e4ff2b53ce45d68ce768b : SUCCESS in 6m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e63cedb17ba45548820bf8cf176fb2f : SUCCESS in 24m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e3d7cd317c45089641d981c3c24113 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d687a769b7c2a18639d2f552c0e6372b6ab808b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 14:03:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c533cad276963255b64d8bb7b9ae449c18b5e80a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 14:05:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42406917915642f38b7d66424c854f51\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba45135d19644adaf9da2c62ad9a868 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/193ec007cb4d4478b8b36acac0a26ed2 : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e99b5a10cde34577afaea7f24c336b98 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53d3bb86c9c849b75a61f43ba8c2fef91d105901","revisions":{"53d3bb86c9c849b75a61f43ba8c2fef91d105901":{"kind":"REWORK","_number":1,"created":"2024-11-25 10:49:04.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/37/936137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/936137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/936137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/936137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/936137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/936137/1"}}},"commit":{"parents":[{"commit":"27d66a63f3bb85df5a110e86ae6a286770d902eb","subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27d66a63f3bb85df5a110e86ae6a286770d902eb"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-11-25 10:49:02.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-11-25 10:49:02.000000000","tz":0},"subject":"Update .gitreview for unmaintained/2023.1","message":"Update .gitreview for unmaintained/2023.1\n\nChange-Id: I64bd376d40f42ca9e330c406bac4ded77c20e4eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53d3bb86c9c849b75a61f43ba8c2fef91d105901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53d3bb86c9c849b75a61f43ba8c2fef91d105901"}]},"branch":"refs/heads/unmaintained/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
