)]}'
{"id":"openstack%2Fswift~936208","triplet_id":"openstack%2Fswift~master~I85b5302c2416de1793599385b791fcd3ec3b4da0","project":"openstack/swift","branch":"master","topic":"port-formpost-to-wsgicontext","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-12-11 08:56:20.000000000","reason":"Change was submitted"},"37337":{"account":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"last_update":"2024-12-11 08:56:20.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-11 06:37:17.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I85b5302c2416de1793599385b791fcd3ec3b4da0","subject":"Refactor FormPost to use WSGIContext","status":"MERGED","created":"2024-11-25 23:28:26.000000000","updated":"2024-12-11 09:15:15.000000000","submitted":"2024-12-11 08:56:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"936208-port-formpost-to-wsgicontext","meta_rev_id":"31b330aea105d07b69e246a44fe5939f44a56222","_number":936208,"virtual_id_number":936208,"owner":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-11 08:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-11 06:37:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-11 06:37:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 00:45:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-26 23:41:12.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-12-10 00:39:07.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"bd7ae7b111b28a5dfc42acfccbc688232d263a06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2024-11-25 23:28:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b00f8d29dc8f9a17c24f3a81009893a9b5299e50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 00:45:11.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd3e37486963418a81befcaa85d99a1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec819470e0974d4d94410f1ac608e921 : SUCCESS in 6m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/3477894fbb5943cea0deab5de8b6b0dc : SUCCESS in 1h 08m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cac7d4d341274e97b51f5ea03b4fb431 : SUCCESS in 1h 09m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/914650acfba6406580eb95b06588f2fa : SUCCESS in 53m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9169bea0296b45a08ab06ce4bd726d6a : SUCCESS in 1h 02m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b84aa471b834abe8f1eb1d3f6480627 : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca93bd74f1534a09bb23241fdad0f669 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6c45287aa974f8eabcf71e7a1bfb33d : SUCCESS in 20m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b619f83b6d5546f0b29db25208bfe28f : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/54a13a59eda54097a55dfef1e0df9562 : SUCCESS in 22m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ac8ca5e054249118bbbad67dddb6d3a : SUCCESS in 17m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3d160adf7ac24864bf383a856284762e : SUCCESS in 19m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dd4c81b4f26f4bf6a129e2f9e387731b : SUCCESS in 12m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/367601b3f9f74c15842c855937c4d11c : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/134d2742732b4a09b3b2e97760a9b08f : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/12426f5ce2ef43fd90530d87228626d0 : SUCCESS in 9m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/128279f7beff446ea8caee16e138bb66 : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4d03eb19225b49c5bb3040e47fe0af66 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/feaa7cea5f654becac1ed438f8a4288e : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d7600b57ae4b40b8ac7effaa18ea1682 : SUCCESS in 4m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c41a2df184534e3490185aeffc5b925c : SUCCESS in 16m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0753009f486145379b916bd3a24464be : SUCCESS in 8m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9ab454a61ed42ef83062af96da1535a : SUCCESS in 32m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2061659edf684ff78becefe22eb7a8a5 : SUCCESS in 55m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5321619520084bf9b776d24434c8404b : SUCCESS in 27m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/076d939bd47a499cb721cf8803bb7287 : SUCCESS in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a740c67b1fca4794ba61f52bbcca6900 : FAILURE in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d14acb040874effa313185f1532ad85 : SUCCESS in 18m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b683aefc28d4d2481708e08d5992c2a : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":1},{"id":"bfb2d5ccb6cf8293f48f71db4de8c0faf1528092","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 02:36:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcb59aefc3db4dc9a8e96e97f969a118\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/266888081857447cad6895107bd05c9d : SUCCESS in 2h 45m 35s","accounts_in_message":[],"_revision_number":1},{"id":"44dd1da7aafeddce4d16edccabdc5403073afcd8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-11-26 23:41:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d32834bb62a8e9e18eaad572648246278a57e04c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2024-11-27 01:35:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e656260430791e7a53202dfafdea03fd4b3d66fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 02:51:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8e09f866db34dd9aaf36dc6d0084332\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e4700b89cd4798b2609202cf2ef578 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/6816ffa5da7342e995a7b9df86b5a157 : SUCCESS in 1h 10m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/991d1fda0960433aad3edae2520df9ee : SUCCESS in 1h 04m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/56de09b199004733844f64592851b8d9 : SUCCESS in 49m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e6252ba9d70499a86b5ec0440ddc6d4 : SUCCESS in 1h 10m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c1bd238d2b74a13918493ce4800352f : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b779d94fd7934b09885ebafa7fbd3166 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27a5206ac3b24588b5c20971c6d5bcf1 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/24cc6a1b4d7a421da198f19c520a56a3 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a7f4e5e19f5a402c95a786e2beb2b602 : SUCCESS in 21m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/296950d26f934bd98ba3e72dc37368e3 : SUCCESS in 18m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/94f9db9dd58b40bdbc24d90d901f462c : SUCCESS in 9m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/405c34d01c8042db912b03083b079db5 : SUCCESS in 17m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba245751c0a346ecb3c3156c8e03e4c0 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d6d8b53f1dbd498a83c250630b6edb07 : SUCCESS in 9m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e002b25ca7464f0196f551bc6f6c4146 : SUCCESS in 18m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c9349cfcc3f64d9591659e0653a0702b : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6f8f4d86513544408c3cf66ea7fe572b : SUCCESS in 14m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6e050e175c4848fcbcac616a5ed17ed4 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c2226fb8d4346049cdbc01c7c06db74 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/493a02cca44c4adfb3042a836ad29cb0 : SUCCESS in 14m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/111953dbf17744d4a8aaf0adab33fc1e : SUCCESS in 6m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5801301f158f402494a280c7018e577e : SUCCESS in 56m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3332a264935f4766bf4c275586735224 : SUCCESS in 27m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9699e7fc955b45e5aed81a17194c4b2d : SUCCESS in 50m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec558fee583241799a1688005a9329fa : SUCCESS in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84aff138845d4978a92d75b3e5e57989 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e8b9548910f46de92d8d7419cb3008c : SUCCESS in 16m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/916dd484e0194c14a2416e5e1013b0a2 : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":2},{"id":"c9c026daae5613978e75c71b5fdf52c84f834f79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 03:07:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff86e9073178459782b3903a1fbb6792\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce091fc8078e4efdbf39ef4dc8252fa7 : SUCCESS in 1h 30m 12s","accounts_in_message":[],"_revision_number":2},{"id":"615d9748247468d27435e9b5ab749ad10a79c0e3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-10 00:39:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cf431424a27346776b2a4fa595f355bd2cb04ad2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 00:39:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eff44ba2c7d47b561174a0724e72435dda37a960","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-10 00:41:23.000000000","message":"Patch Set 2: -Workflow\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8980f04dd390f8d59a9b5875ab22d92f19ef685d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 00:41:36.000000000","message":"Patch Set 2: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":2},{"id":"79928a851f2f7ea66e0d4834136e4ef533c545e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2024-12-10 01:12:34.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n","accounts_in_message":[],"_revision_number":3},{"id":"03c0456450b7df2e06d82f2ac702be525faaa007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 02:36:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed0c9ccb6ef444dd969648d8ec684b13\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad2a030f00a044688b99d2244f644f2d : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/5e32ffe4ac2148dc9a62db6957d291af : SUCCESS in 45m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b31b2ee470c24555a3e43c35ce30038d : SUCCESS in 41m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3cf7dccf1a1e4320a7e3f3c2eac3ead6 : SUCCESS in 49m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed1b79afd4ec456a89eb8043b10c41de : SUCCESS in 54m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df394ba1ea7c43ee9dfe94005bdcb3c6 : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37e023599ce14611a500a30c798d5dc6 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4340d5cea9474637a52aee5ff71ecc79 : SUCCESS in 15m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f0a3a0543dc452d9c6c2bb9a50b6f2a : FAILURE in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6ac4a1b619f84302b4ad781c0b0ecc87 : SUCCESS in 21m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee58d618353f499ea6fe05fe709971bc : SUCCESS in 18m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/65b64a87a4634eceaf1e52f2c033acf7 : SUCCESS in 20m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bcdd512cb30e4bf6887bb3d16f5bdfa0 : SUCCESS in 29m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fa68df1efbb6432b9d319c1303506671 : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/92e56632c0cf4ee7be0c4fc0c1f62b56 : SUCCESS in 9m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b980bdc51fcd4848bbc908e7eb65c246 : SUCCESS in 18m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5e3453be2d3b4c308901400a592085aa : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ab1a657fd65d4bb6a3522f703c54826c : SUCCESS in 13m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/949f9c93d93244d09ea37881a231f008 : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/98a98b0e77b54521b41d2249fba1c742 : SUCCESS in 7m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d598d6aff1d146a48383a998c19eeafb : SUCCESS in 15m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0d3d80e6f214462f81051009286502f7 : SUCCESS in 6m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/68b58c55345b4412ac440d686c52a47f : SUCCESS in 39m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17193e41335640cbbaa028f824cee573 : SUCCESS in 29m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77717ec1c0e948feb8d9d9bf73f870dc : SUCCESS in 31m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dec773b4613745f69264beca5fbfea57 : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eccf763922342dd894b5585ac6078a8 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5b7f49abe35a4626a8d16f8b71061719 : SUCCESS in 17m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d401db8572f4646921c13f26bda7fba : SUCCESS in 49m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69a0d5289dda40a6b291aced9122cd77 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":3},{"id":"535facfe5ec8bb1aed999febc20e73e1cdd0bb23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 03:34:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce068ca808eb4940880603626d2552db\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc3904507b2b4e2bb5451183418dffd3 : SUCCESS in 2h 18m 54s","accounts_in_message":[],"_revision_number":3},{"id":"1f31ff25aa0e4e8dff5c0124c78d14ee75eb603b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-11 06:37:11.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f00c648666b92d16dd168a3af021b8841b6a009f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-11 06:37:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f1a48099796b6acef01df130dcd2a09c9fd26366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 07:51:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2342b3380332452293835cb601c12e6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c8681584214eef8a9ce3250584eb36 : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/5cd5ff2c65fe43d880fa31a34c40254a : SUCCESS in 50m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/93e127ab70a743d5aad109d013b93105 : SUCCESS in 32m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84313be5cf28484f816c2f10a38300e5 : SUCCESS in 54m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/732a21cbff8644cb854b8d879014aea5 : SUCCESS in 56m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67b0a84464e04846ba2057d82196816a : SUCCESS in 22m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2e2990e4ae24eeeb7be8de294c4f8b4 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14a0706b7bc94e72869c606b9f76b1c3 : SUCCESS in 19m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9ab627de8a054e8e82c7c81a705e170e : FAILURE in 4m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/293259416dfd45debbb04f25b6223280 : SUCCESS in 19m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b0c793b19d7746b6a510ff11ca17d3ef : SUCCESS in 20m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/51507709e1a44d4f945feefb22b1a098 : SUCCESS in 19m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c5c604daf8fd4a0eb7acee6fdbecb875 : SUCCESS in 26m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71227df8dc3443a9921f4f0b5aa96774 : SUCCESS in 19m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef270a7ce55f4f9384da592f2c7ca29e : SUCCESS in 16m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d85c0d4d93844fe95baac3b3d5af40f : SUCCESS in 18m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8260e231021549c6b9f835bbcaff3f5c : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/75f93ab3b9e44939972968d559a14025 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8e3a248efcda407ba6905ed594128300 : SUCCESS in 14m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5cce6b7f9629410ea5d915c2709e81de : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/12028e4285ea490d900f1b9a84270962 : SUCCESS in 14m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6c66b3da024a4788b38484d473695446 : SUCCESS in 6m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/90a905130df144f6a494599275cf9ee9 : SUCCESS in 1h 03m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9394aad22ef7411caf8c43b868ddd99d : SUCCESS in 43m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a9c2437e9d374defab8be775b7500fe1 : SUCCESS in 43m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d631c7395cf43aeaa5fd776c294eb66 : SUCCESS in 22m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9db9e6340d647a3987358b67892c578 : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1cb20c7ba924d56a9e576d4f257f7d0 : SUCCESS in 18m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b2d4beba46240b1865faa9082bfc28e : SUCCESS in 48m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/005028b8e92a4d969a41eb7b538818a8 : SUCCESS in 49m 41s","accounts_in_message":[],"_revision_number":4},{"id":"c0db89dbbd199c843d78c3fd0591b202015081c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 07:52:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"17b12d1080d539e542a30b7eddcbc5a19f3a4f9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 08:56:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f91ff61d2884ada8c91c79b7d738da7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37569e4c42e4d7f96f866d5c5c272f8 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/3d1f9c50dc26468890145be6461ae8dd : SUCCESS in 59m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/874f4e01d5ef480396300a459faaf268 : SUCCESS in 49m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f55f7d6c7b74950b45160891f7738ee : SUCCESS in 53m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2be4ae4c88084300a21a2721f9999b1f : SUCCESS in 39m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/291aede81e084b82ad2795ef8b78cb1f : SUCCESS in 4m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b8cada2c97db4e4c8fb06bb69fa819da : SUCCESS in 9m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62c8d61418dd408c9dae47c0db3e65e3 : SUCCESS in 8m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/357847d9787549f8b7d6762efd59ee30 : SUCCESS in 9m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/01cf25888c184202bfa2754e8d29c360 : SUCCESS in 27m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/13dddbdafe6b43f9b50b28a64db70731 : SUCCESS in 9m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1fbd8c7d236b4e3fb4560fe137de9574 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd6a5db4305b4aa8aeb3c7b3b60562ed : SUCCESS in 17m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/45853e80ef51469f8b30d537e9a1f2b6 : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4862d353994b4a54ad37b1de9e8c9618 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dc0e40e23d5d413f97522710072796c9 : SUCCESS in 14m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f916baf9562b43c38fa1dec9ac4361ad : SUCCESS in 7m 30s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3d94635ca1e749b39f2e3d07bef4b6fb : SUCCESS in 7m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b458cc26819846748874bbc9ca7fa2b8 : SUCCESS in 56m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d6ef93fa0ce479d98201bc2b3cdce17 : SUCCESS in 47m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/52a8d32f244b46a38e2f2936a4ee363f : SUCCESS in 25m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ab4181cf3c547e1a992852b9221b2dd : SUCCESS in 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf4ed122d0de45b78716a8ec9d2d98cb : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d171521a5c3d4275a5db55a804d62ec5 : SUCCESS in 17m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3269c7746a124ddc97b320199133a8e2 : SUCCESS in 45m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ff56c3c6f6d44ea848056b645f7ec2f : SUCCESS in 35m 53s","accounts_in_message":[],"_revision_number":4},{"id":"86429db905397f3509b0bf109fe4ed668ff695a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 08:56:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a83affe64da791520d586e43785a2ed0f9bd3532","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 08:57:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b05296145d5a47358f5b564683c869fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5dd81f13a547a891a04ec25a108720 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/caa2bc92cfaa400cacfd5f0b95f5dd24 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"31b330aea105d07b69e246a44fe5939f44a56222","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 09:15:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f78677eec02041909781b82da3cac173\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d0e3d49d06ab41be8f953b5c82f3509d : SUCCESS in 1h 16m 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"404edeb7fa68730fd95682cb96e0f39ee51051d0","revisions":{"d41a971fb509cc2e649874030f08756c2228bdc4":{"kind":"REWORK","_number":1,"created":"2024-11-25 23:28:26.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/08/936208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/936208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/936208/1"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"nathang15","email":"nguyennathan1502@gmail.com","date":"2024-11-25 23:28:17.000000000","tz":-300},"committer":{"name":"nathang15","email":"nguyennathan1502@gmail.com","date":"2024-11-25 23:28:17.000000000","tz":-300},"subject":"Porting FormPost to use WSGIContext instead of self-handling subrequest","message":"Porting FormPost to use WSGIContext instead of self-handling subrequest\n\nChange-Id: I85b5302c2416de1793599385b791fcd3ec3b4da0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d41a971fb509cc2e649874030f08756c2228bdc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d41a971fb509cc2e649874030f08756c2228bdc4"}]},"branch":"refs/heads/master"},"2473f289675342ea4456c150850ae42b8af46170":{"kind":"REWORK","_number":2,"created":"2024-11-27 01:35:42.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/08/936208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/936208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/936208/2"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"nathang15","email":"nguyennathan1502@gmail.com","date":"2024-11-25 23:28:17.000000000","tz":-300},"committer":{"name":"nathang15","email":"nguyennathan1502@gmail.com","date":"2024-11-27 01:35:09.000000000","tz":-300},"subject":"Porting FormPost to use WSGIContext instead of self-handling subrequest","message":"Porting FormPost to use WSGIContext instead of self-handling subrequest\n\nChange-Id: I85b5302c2416de1793599385b791fcd3ec3b4da0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2473f289675342ea4456c150850ae42b8af46170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2473f289675342ea4456c150850ae42b8af46170"}]},"branch":"refs/heads/master"},"548b17c583a0a08f9ba4a909e6f4a1fd6a6ae141":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-12-10 01:12:34.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/08/936208/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/936208/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/936208/3"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"nathang15","email":"nguyennathan1502@gmail.com","date":"2024-11-25 23:28:17.000000000","tz":-300},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2024-12-10 01:12:34.000000000","tz":0},"subject":"Porting FormPost to use WSGIContext instead of self-handling a subrequest","message":"Porting FormPost to use WSGIContext instead of self-handling a subrequest\n\nChange-Id: I85b5302c2416de1793599385b791fcd3ec3b4da0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/548b17c583a0a08f9ba4a909e6f4a1fd6a6ae141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/548b17c583a0a08f9ba4a909e6f4a1fd6a6ae141"}]},"branch":"refs/heads/master","description":"Edit commit message"},"404edeb7fa68730fd95682cb96e0f39ee51051d0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-12-11 06:37:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/08/936208/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/936208/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/936208/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/936208/4"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"nathang15","email":"nguyennathan1502@gmail.com","date":"2024-11-25 23:28:17.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-12-11 06:37:10.000000000","tz":0},"subject":"Refactor FormPost to use WSGIContext","message":"Refactor FormPost to use WSGIContext\n\n... instead of self-handling subrequests manually.\n\nCloses-Bug: #1523401\nChange-Id: I85b5302c2416de1793599385b791fcd3ec3b4da0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/404edeb7fa68730fd95682cb96e0f39ee51051d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/404edeb7fa68730fd95682cb96e0f39ee51051d0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
