)]}'
{"id":"openstack%2Fswift~936257","triplet_id":"openstack%2Fswift~master~I019447614ebadbb9e2cc8a18c0369bc16a89c0d9","project":"openstack/swift","branch":"master","topic":"p-probe-test-reconciler-versioning","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-27 19:43:25.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-11-27 17:13:39.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I019447614ebadbb9e2cc8a18c0369bc16a89c0d9","subject":"Add probe test for reconciler with object versioning","status":"MERGED","created":"2024-11-26 12:57:08.000000000","updated":"2024-11-27 19:44:30.000000000","submitted":"2024-11-27 19:43:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"936257-p-probe-test-reconciler-versioning","meta_rev_id":"bb677f2f8d94c6e8a93839a861882da260259991","_number":936257,"virtual_id_number":936257,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-27 19:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-11-27 17:13:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-11-27 17:13:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 14:00:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-26 16:13:48.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"3f984d361e5f2bdc1847311e38569afa3ac7a0a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-26 12:57:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f3e942082507d53138fd15c5e691cbf6c092b07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 14:00:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71aa3b8537ad4cc2ab2c6229e7e81b38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78b087a211f341c2a8efbf496efc6e1d : SUCCESS in 8m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/91a3176cc2b540b3a3dc6538e0766347 : SUCCESS in 53m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c44e6e29374a47d9a2ffe2f353f19a9b : FAILURE in 57m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a67a161d1db34a9f91b9739378bfb557 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f85f5bccccf3100e5e213f64730b60f9f4cdbae7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 14:59:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fda9feab9f044730a5d2045030bf6a6d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/693fce3065ae4f23a28fd5e7f0d36b63 : SUCCESS in 1h 25m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a3ac088ebe616bb07c1d97f1ad6f1189b2e59e16","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-26 16:13:48.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7cfef52b3750ef897edbd12bfdccd3b6d0053df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-26 18:27:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6e9eff04cd74fb37eca683c522f743f95cde89c1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-26 18:30:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23d2491270ffc098bb034683bc602052e2c3651f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:33:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bac29aac8f649048f1fe0ac61ea1c96\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7ea5a5e43c0b4776a077d79294a2d389 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ea40e4566942beea6015b0bad7c1c88b05ffc84e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:40:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f54af6edb594046b47e323f9d9bcded\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea00b4ae17084760bf6f434280276741 : SUCCESS in 9m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b42cfe44a5304f12beec48b6d2fcc111 : SUCCESS in 1h 04m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2d519ced9b641759b34ee5441a6c0ef : SUCCESS in 56m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6710765779654c169a3c3d9515dfb2e9 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"6f8e5a942151ac591a81d01374af1c4b62808007","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-27 17:13:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7ea6680eaf8b403f379fea8ee2ea13a0ea54207","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 17:14:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ba3aa56988d55259fb66ce058ec516071eba659","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 19:43:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40fff9d7d90240d7a7b13387ee14596e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343440ffb7134206b2e443539906a53a : SUCCESS in 9m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fbdcf69c4d764745a79513a1caf39ab7 : SUCCESS in 47m 23s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c8f67643bec1487a81e1dfd12c3d4a39 : SUCCESS in 22m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c5c5ce187ea0464a87f81d260ea92a02 : SUCCESS in 9m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c641747974bd4c4c92436d127e784132 : SUCCESS in 22m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/48294674e02941cdaf2c4913eed757d8 : SUCCESS in 14m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e821931bb7b4cabadca32faeabe9069 : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2ebe0780ecfa4baa913090fd587f29c9 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a67bad6578094537ba93f719abf41821 : SUCCESS in 16m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/119e788b4be94b4f944b193c2a52be00 : SUCCESS in 9m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4481827941f7478ebf9934f355c784d7 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0b2553a3c8334c9298d7a9798816d0fc : SUCCESS in 13m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d821c4a6e2f945d98475215f81e4c47c : SUCCESS in 6m 31s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a17325da5ab14ff8a58bc3e43616b621 : SUCCESS in 7m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/16043b89e8bd469b9ac7db9d0844cfe3 : SUCCESS in 50m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b04031f9a3d45fba81b16faac551a49 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":2},{"id":"a7a92d716da08666a4ae9eb5566c71e4a3df997d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 19:43:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bb677f2f8d94c6e8a93839a861882da260259991","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 19:44:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78e9976662734272925b64ad8aad35b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488041a35a864ca886766b76727f8448 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/13423195b92c4ccc9674e8307da7db75 : FAILURE in 38s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e751ccfb270436a3a0d728701b61d7460352ab41","revisions":{"68aa790e5c56f5906ef077d08b2183f3abe2747b":{"kind":"REWORK","_number":1,"created":"2024-11-26 12:57:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/936257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/936257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/936257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/936257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/936257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/936257/1"}}},"commit":{"parents":[{"commit":"fa889358acf7675efba2de095777886b4c1ff7f8","subject":"Ensure correct content-type in container HEAD response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa889358acf7675efba2de095777886b4c1ff7f8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 12:55:19.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 12:55:19.000000000","tz":0},"subject":"Add probe test for reconciler with object versioning","message":"Add probe test for reconciler with object versioning\n\nChange-Id: I019447614ebadbb9e2cc8a18c0369bc16a89c0d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68aa790e5c56f5906ef077d08b2183f3abe2747b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68aa790e5c56f5906ef077d08b2183f3abe2747b"}]},"branch":"refs/heads/master"},"e751ccfb270436a3a0d728701b61d7460352ab41":{"kind":"REWORK","_number":2,"created":"2024-11-26 18:27:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/936257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/936257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/936257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/936257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/936257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/936257/2"}}},"commit":{"parents":[{"commit":"fa889358acf7675efba2de095777886b4c1ff7f8","subject":"Ensure correct content-type in container HEAD response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa889358acf7675efba2de095777886b4c1ff7f8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 12:55:19.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 18:09:02.000000000","tz":0},"subject":"Add probe test for reconciler with object versioning","message":"Add probe test for reconciler with object versioning\n\nVerify that the reconciler is able to move objects to a container that\nhas, or will have, versioning enabled, and versions of the moved object\ncan subsequently be created.\n\nChange-Id: I019447614ebadbb9e2cc8a18c0369bc16a89c0d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e751ccfb270436a3a0d728701b61d7460352ab41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e751ccfb270436a3a0d728701b61d7460352ab41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
