)]}'
{"id":"openstack%2Fswift~936295","triplet_id":"openstack%2Fswift~master~Ie880e401170105272dbaccd617a1830fb928bd05","project":"openstack/swift","branch":"master","topic":"p-probe-test-reconciler-versioning","hashtags":[],"change_id":"Ie880e401170105272dbaccd617a1830fb928bd05","subject":"probe test showing 4 versions from 3 PUTs !!","status":"ABANDONED","created":"2024-11-26 18:27:39.000000000","updated":"2024-11-27 12:32:23.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"89a08c0b6b5c4cd3c582356cebf5a227348381ae","_number":936295,"virtual_id_number":936295,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-26 19:25:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2024-11-26 18:30:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 19:25:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b02fb200894a1b2d30fb015fb17f533b50e1e337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-26 18:27:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09f0edba590ffc37b7bbe23aba4d9bbff9cc245c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-26 18:30:36.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a460c51f473b6118f2a0a18b74a7f24d387d0fe8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-26 18:38:58.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"061c960f6c27f81610396c18d689beb388f0c32e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:25:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d1356ae5174668b401495586cfb666\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8bd5d4abd5d44fc82ec998981a7769e : SUCCESS in 9m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c860c0c977b041feaaa1e925bf425574 : SUCCESS in 54m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4feb614e6134ef08c028c8abd7d1e76 : SUCCESS in 51m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/948a1c33c69049faa63a0d07715767c5 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d779788156e555f1569ba67c549ae607978a1503","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 20:02:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aca600fe3d624b3793f817b8d0d40a17\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/325702f32cd14a65bc1c4919088ae195 : SUCCESS in 1h 32m 23s","accounts_in_message":[],"_revision_number":1},{"id":"89a08c0b6b5c4cd3c582356cebf5a227348381ae","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 12:32:23.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/936296","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c612b034fce66c6411468dbfc280ed6f24f0de16","revisions":{"c612b034fce66c6411468dbfc280ed6f24f0de16":{"kind":"REWORK","_number":1,"created":"2024-11-26 18:27:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/95/936295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/936295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/936295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/936295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/936295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/936295/1"}}},"commit":{"parents":[{"commit":"e751ccfb270436a3a0d728701b61d7460352ab41","subject":"Add probe test for reconciler with object versioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e751ccfb270436a3a0d728701b61d7460352ab41"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 17:39:10.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 18:24:17.000000000","tz":0},"subject":"probe test showing 4 versions from 3 PUTs !!","message":"probe test showing 4 versions from 3 PUTs !!\n\nThe probe test illustrates some inconsistency issues with object\nversioning.\n\nobject_versioning creates a fresh version of potentially the same\n.data file every time it finds it in the user namespace, so when\noverwrites only partially succeed there can end up being multiple\nversions of the same .data file copied into the versions container.\n\nFurthermore, object_versioning doesn\u0027t attempt to preserve the time\nprecedence of PUTs and POSTs; in the case of partially successful\nPOSTs, versions can end up with either stale metadata or metadata that\nbelongs to a later version.\n\nChange-Id: Ie880e401170105272dbaccd617a1830fb928bd05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c612b034fce66c6411468dbfc280ed6f24f0de16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c612b034fce66c6411468dbfc280ed6f24f0de16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
