)]}'
{"id":"openstack%2Fswift~936296","triplet_id":"openstack%2Fswift~master~I15a59a35d857791140b2475cb70a6514e0f95e28","project":"openstack/swift","branch":"master","topic":"p-probe-test-reconciler-versioning","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-01-03 16:14:15.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-01-03 10:30:33.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-01-03 10:30:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I15a59a35d857791140b2475cb70a6514e0f95e28","subject":"object_versioning: only create one version of each .data file","status":"ABANDONED","created":"2024-11-26 18:27:39.000000000","updated":"2025-01-03 16:14:15.000000000","total_comment_count":31,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"1077cd339c6362a501e9b386724f3ba89278501c","_number":936296,"virtual_id_number":936296,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-19 11:24:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 19:48:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-18 15:41:26.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-12-18 23:56:06.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"c783dba19f925a16872ac15ec3ec3406575a1fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-26 18:27:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"709bffc31b1cdac79639fc4c3139e5aa2a394b51","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-26 18:39:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eaa5adfe02228d83155a53ea7afd2de1ee91fc3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:48:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1efae0b379864a94a4de910da532bf44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece1cf8e1d784baaacfe2f613cab5e7e : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/c99cd295d7814daba8337cdab39bd7e4 : SUCCESS in 59m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1ad5b8b65ae407bbb21cdb676fafaa3 : SUCCESS in 1h 11m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f2f02ed528d74f9dba1067fa6ec0fbb1 : SUCCESS in 53m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01c033bf911a47b09703d463aa5df510 : SUCCESS in 41m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa59af93eda244a29a52daa36ed9faa9 : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20549b78367c474bbf4da6963c60f0e1 : SUCCESS in 22m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f04746ffea64b9b906999a20f440a86 : SUCCESS in 18m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ccf1ce99c9034b598d01f1ab4d4abe95 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24d03323c18d461f9aac698ae4dc1513 : SUCCESS in 18m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d01fd71deeea4a63842637be553e3196 : SUCCESS in 15m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b8c9961f675a4c6c8ae25a4765bae0ff : SUCCESS in 11m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4f14ea58fc5a4a94acf67be66b2215fa : SUCCESS in 20m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d9615b32637945fcb0ebf29ee1defaf3 : SUCCESS in 21m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e4b9a781417943b4bb55df4c25ce411c : SUCCESS in 16m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a74ed5a942fb4dcb8c173184e19419f8 : SUCCESS in 16m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3d2a67e50a254842986f799fc2b2a1be : SUCCESS in 13m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f85d8c443f9b4e67b3d4f7c6d78bee26 : SUCCESS in 13m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4bc35cfba46f455cb39d120fad6f3970 : SUCCESS in 18m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9eef0408cf5240479d89fa162cb69557 : SUCCESS in 7m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c75e0170ef454b66b3951d51ce5aace3 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a8f2a49e96494c70b7b78cf4c8e5c627 : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae39b1d899ad4d768ab8293f1065b24b : SUCCESS in 55m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/56a0e560e48748f29668f54dad3222b8 : SUCCESS in 43m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/025829e5a3964177b2d9ec5227fff51b : SUCCESS in 1h 07m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4473a50a07cf479b8945f6d66bc59bae : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bead89313194f099fa18e8c7654cabe : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e8c3b861f9349e4a0dcd7081cc8ddb2 : SUCCESS in 18m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6414115e5cb424088386fdce7bb838f : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":1},{"id":"01c12fdcdc33e13ecd96d048d22a868ba66db218","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:56:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccfe654c7a364c50811a57dd882067f3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de6d858ff3d94c80a7e413ac0a8738be : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d9e3818dd63da03b90be04c7158840f58aaaab56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 12:31:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"633ce44699523f4779368fecb132a074999398c9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 12:32:37.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"715fddf49daf05deebd79506894aa5379db92e31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 13:49:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7076f8ffbcd14b92bc61897a1ae87531\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88eea380003849a3a894180b7c5ce5ef : SUCCESS in 5m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/df317aa65cbf4613b4e49b9259592234 : SUCCESS in 1h 08m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abb44bbe636d46528e0bcc7b452f6590 : SUCCESS in 1h 04m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63258666b3244ecd83ce41730c9c251b : SUCCESS in 55m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9aa195ccab9f446cbfe1dffe8da53de2 : SUCCESS in 1h 00m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b18cbae9e114fd8ba1f7f9400a01b30 : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8caed37d15b4465ba1cf073a1467bc95 : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e368bcacdaa42a398cce5d327622269 : SUCCESS in 16m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b4c15e157c3943ecb9b2f8bab22ca6fc : FAILURE in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a22018493e1a494aaff1a983f0388466 : SUCCESS in 21m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef8b0c746d854639b149fb4d81cab4f2 : SUCCESS in 19m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cee66d8980ee4e0b85293638dfae6de5 : SUCCESS in 22m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e02b968cbae74c0b9e61aa7bb63d8bf5 : SUCCESS in 26m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3c0355ec3564454ba899e4137575c45a : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/622816e45d3a4f5089c517bb224e74b4 : SUCCESS in 18m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/596b7acf08364da3a052724ff200fca0 : SUCCESS in 16m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/68036acbe6e34359b26df52744d4f874 : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9d680d1c7c4a47a28d3112bc2a2b9078 : SUCCESS in 12m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1099066fef904dc5a8deb1b5d9d3bfd6 : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83679e516f5d4c94937e124d68fa92b7 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ae2197c990d4eb99ebc537e5e5a2b37 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/efe6e67fcf8145938d102fd30f27851e : SUCCESS in 6m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c4a2137f3fa43fda1f5b35ca71f6c23 : SUCCESS in 42m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24a478017b3c430b93f7c915825f423a : SUCCESS in 52m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5b8914cc213431bb1d6683b5b29820e : SUCCESS in 51m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90313f374d55420daff383ff8bb60f6a : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f81d5f4d4aa64c71bc1d5acc7ddb074c : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b0211bb980d64ac8a16065b355d23835 : SUCCESS in 19m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f816ecf0b00147e1aac4a40cb71f529f : SUCCESS in 52m 36s","accounts_in_message":[],"_revision_number":2},{"id":"53722714b39b4d80935908f6976708e154f52fd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 15:12:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331b6f7ecba245b69cf586438550252d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf0678d768a34526a5cfe41a468900a4 : SUCCESS in 2h 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"74b66ae1168043e48fe447e51ea5772cfc643fdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-28 16:34:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7f988d36c4883c6f8fb2de4f06d74944a0846f5a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-28 16:35:06.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"afa3f18c42ed8082e37a730173127ed348a443a1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-28 16:35:19.000000000","message":"Patch Set 3: Code-Review-2","accounts_in_message":[],"_revision_number":3},{"id":"cd7cfb114f1211ae8fe7e34306fdcc5f79471fd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 17:45:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b40d18ca35f4d0b91dcb6616c9273b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb4506deaf14bbea52d66aa2ebc8a05 : SUCCESS in 6m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/1f294695deae477aadc42535a11a186c : SUCCESS in 54m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/697f0a5878544cc0bdf3dd39364e320e : SUCCESS in 1h 00m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a336709cd12403bb724ac1de489c958 : SUCCESS in 53m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4488f3377f84a9fa1ae1b26e49adf4c : SUCCESS in 1h 05m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7a468583a224a7bb4998064a0659d51 : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15cf8e0b6b0f4cd793f7244ea498da0f : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/736b926fdba043e587a69b4090ee0311 : SUCCESS in 18m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/996cfefeff954d45844b81c13bd481c5 : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/df902c0f12824e4aa8851c4ed492be92 : SUCCESS in 23m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3adf40323c9e48aa866aa692084ab56c : SUCCESS in 10m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/66f30a9c08c24b9a9584ffbb7cf8b194 : SUCCESS in 20m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a4e42d61b2fc49dba8a7eda1a074b10a : SUCCESS in 12m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2a8ebc64dc2c48ad812521c91774ef8b : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b6ab177e7153476fb7a7b35d5f84c5a9 : SUCCESS in 9m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3187212076bf4b9194b307ef49468055 : SUCCESS in 19m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b72d12b2e4ac46669aeff6b85f1f2f99 : SUCCESS in 10m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ab808179bb1e445bb9fe21e85201d867 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8699a67123cd4ad0857a4ca57f34bbd6 : SUCCESS in 17m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/24f5b04ab2164e67b15adbf6f321a99f : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1825299217124be586b11efad374feb3 : SUCCESS in 12m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0b1b68f58898442aababecd074e8dfc8 : SUCCESS in 8m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/419f146187064755b8658794bc385a7f : SUCCESS in 54m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85d0f96db7b745d2b327f23606bd4ac6 : SUCCESS in 49m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c892174c0494ea6a36a159a719389a7 : SUCCESS in 28m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66c35e0671dc4f7099e092380f50f092 : SUCCESS in 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533b8b293ed84acbaf2c77bb082331ab : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5719c1e333bd40dd99a0f2a7c51b9fa5 : SUCCESS in 17m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0417de383204174a303a45a1794365b : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":3},{"id":"4269b2d2007fae3343ccae018217376f891d60b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 19:43:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e69eef864fd448e48a880a716f91c58f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/553bf6efdabc48748f31c8ee8a7eae4c : SUCCESS in 2h 06m 31s","accounts_in_message":[],"_revision_number":3},{"id":"ae2c4622aeaed13c2c9c3901084274304db90eda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-18 14:04:31.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9abd5ad4cbdfe3102039dd1fba78e728820ecf9d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-18 14:04:43.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"6ea49138370ca9ba7f1a858232e53daf23daa179","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-18 14:08:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c3f11322838ddea8aec02c08eac9fa62fac0f9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 15:19:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f53f0601c36d445a8263b5bf3f7708b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7868c4c90fe34a3e959e71d31ca675ee : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/32f5abf528884e3180c617700a8e3a6a : SUCCESS in 39m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/87925b6eafa44389a93d1a7ce0d4497e : SUCCESS in 55m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92c03f4ac7c84aa6a538a99223c8b6ba : SUCCESS in 46m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb11f225aab14ca08499b7f0272f06df : SUCCESS in 43m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cfe19c7a3e34b769ffad23bcb34e2b0 : SUCCESS in 24m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ae799b90b1d4824b564c73e7dd1233e : SUCCESS in 25m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39b037bb9769450faf7482f1af199f50 : SUCCESS in 22m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec1438eb852942b88a938c7a1e612e0c : SUCCESS in 10m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/260daad78f7b4cb792d4e321df57f7b6 : SUCCESS in 23m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/209b3b81fe5043ca9e8ff929be141da8 : SUCCESS in 23m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9f2def6ea05e4ba0b29a76cecf467d2b : SUCCESS in 26m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ba353dd830b7434f91c83ac457a31afb : SUCCESS in 34m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9f3c63a3032d4e1690d84bba0d080c39 : SUCCESS in 10m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/185fa6d2bb574f7a9e0baae16893f015 : SUCCESS in 23m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4dd51206df914c808ddd5905088b7334 : SUCCESS in 19m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/03725d69a880433cb5d414ccd594b98a : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/465f83d186384555a4cc5c13edb7f81f : SUCCESS in 15m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fe6ebccb9fbf492488e2322d9c694061 : SUCCESS in 17m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f523a20321454c19854b4fad12517944 : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b9aa7ed2ba0646358481985cf0a043b7 : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/10f0ac786c7844d09fd4ed2d7d1d2ea7 : SUCCESS in 9m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f3c512172dc48809c46ddb819bdcbb5 : SUCCESS in 1h 00m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/27c06029e19f4771b35ab78f54c4ca65 : SUCCESS in 29m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6d11f4683604bbd9bac8759eef0bd51 : SUCCESS in 40m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e310c06d82f54ed3952b9bf6bc36d0f4 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d41d74a0725451b94b70e07443676aa : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71faf427ba314097ad373cf62e908a32 : SUCCESS in 18m 12s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b8c490fd4454bcab07b5fc61c1af4b3 : SUCCESS in 36m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ebb344dad5b4c1993574cc7dc6de11a : SUCCESS in 42m 06s","accounts_in_message":[],"_revision_number":4},{"id":"edb5b166b9f8d4f1a75fd996848b6f4cdc731c38","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-18 15:41:26.000000000","message":"Patch Set 4: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"125523b4b6ca918bd709806ec6f2e3779af3c07b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-18 17:12:31.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e47a7814b002ace7bea5033e3fe0c7297733876f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 18:15:57.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5e36009c525475087ed1ba621040565\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8bb193e5eb9f4ad982dbc026a91b4e78 : TIMED_OUT in 3h 00m 55s","accounts_in_message":[],"_revision_number":4},{"id":"b99e6483cad9abb6486bbf219f21967c18eaa372","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-18 23:56:06.000000000","message":"Patch Set 4: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3f497f4e2b25e467085fdcc6a087b8c0ab17a1fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-19 10:15:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"22091139c84dbbbb6bf4446e2117aacac4f2ec34","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-19 10:19:17.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ff0df455fc65dbc50efa91ae0fe4ccc1ef107193","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 11:24:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a515ce08c5a543b68bc8373e03d35b68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49cef7e000fe48beb6fcdfda201b6cf3 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/7129dd2b8a234ff18482db87f970592e : SUCCESS in 1h 02m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7a29980472954b4b9102b8a7f41491e1 : SUCCESS in 45m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/13ddca8570d64542973889b40a6191d5 : SUCCESS in 45m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05e795de3588418492daf1b90ed93ddc : SUCCESS in 55m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f2dd02e6a59419f94ce8e04304dcaa9 : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81258f0ca15e46d2a8043d55fd0d4319 : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc73d1862a654c9e8702754322c92d32 : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/894e280785ae409e86a3a87ca4df2365 : FAILURE in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6798cf5348d44835a3f65b09981b69c7 : SUCCESS in 19m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5fc36172019d497d85263333f13eabef : SUCCESS in 15m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4a5613365a574659b1fc4d56416b922f : SUCCESS in 16m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6caf979d72914f36848f952c29458cf7 : SUCCESS in 25m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/00cdb59e1b0049d8b68a62f014a25ab6 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa58d17977d944b4a955319cac8225ba : SUCCESS in 19m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aa25e0178fe148c6bbe4299c71b9b6b3 : SUCCESS in 15m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/51826648a1d84ace9c754a8f7bd21f48 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/377bc1604e8f4a99911fa59c92ed9696 : SUCCESS in 13m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a1d21cf9c2c14a1eb009c011635cf491 : SUCCESS in 16m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/981f2a1e1e234fdca4aeb7218897e11d : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7382ecbaa1984cff8ddea8a370eede9c : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9503810a910f41a8ba51da2ad279bc8c : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/be5898ea89e54b43b980ffef16e856d6 : SUCCESS in 47m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00fbaf3b3bb5414fb0f18b1857984c0f : SUCCESS in 50m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9fd897cec31f4e44b26c56626fd5d0dd : SUCCESS in 35m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cbd8d06beb3418c8e9687021e32608a : SUCCESS in 19m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f7792c2564b4d6da89cf864c4f5a1c5 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f19d915de69a47829ad5c2a093f1cc4a : SUCCESS in 18m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00a86851cd2244038807b29f36a9fc86 : SUCCESS in 39m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9558f7a0416b467bb148d458134720f9 : SUCCESS in 50m 44s","accounts_in_message":[],"_revision_number":5},{"id":"c1c9dc82391e50cdecbd9727c9beb5c817356ffb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 12:05:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e80b72c2fd2441d1971927a37124e0be\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e8e50711748448aeb264e6ffc1b59faf : SUCCESS in 1h 39m 36s","accounts_in_message":[],"_revision_number":5},{"id":"8e911a6089cc43179da191b4b61a6653da98c0ae","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-02 23:28:21.000000000","message":"Patch Set 5:\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"939457822e89fcf3eee1273cb7de229042fb533a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-03 10:29:24.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c9557b63dcab0346202e8b0c8620bbbf6c91c707","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-03 10:30:33.000000000","message":"Abandoned\n\ntaking this back to develop on feature/mpu https://review.opendev.org/c/openstack/swift/+/936379","accounts_in_message":[],"_revision_number":5},{"id":"1077cd339c6362a501e9b386724f3ba89278501c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-03 16:14:15.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"85d6f8a984dd0d7f075318a9d42f6d78e6bcc337","revisions":{"f8aca839e04eb64ba4985275ff75c217da011f57":{"kind":"REWORK","_number":1,"created":"2024-11-26 18:27:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/96/936296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/936296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/936296/1"}}},"commit":{"parents":[{"commit":"c612b034fce66c6411468dbfc280ed6f24f0de16","subject":"probe test showing 4 versions from 3 PUTs !!","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c612b034fce66c6411468dbfc280ed6f24f0de16"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 18:09:34.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 18:27:27.000000000","tz":0},"subject":"WIP object_versioning: use data file timestamp to name versions","message":"WIP object_versioning: use data file timestamp to name versions\n\nAvoid multiple versions of the same .data file by using the data file\ntimestamp to create the version-id that is used to uniquely name\nversions.\n\nChange-Id: I15a59a35d857791140b2475cb70a6514e0f95e28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8aca839e04eb64ba4985275ff75c217da011f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8aca839e04eb64ba4985275ff75c217da011f57"}]},"branch":"refs/heads/master"},"2b1f506cdf48d28fc26c68e57929504b115a2820":{"kind":"REWORK","_number":2,"created":"2024-11-27 12:31:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/96/936296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/936296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/936296/2"}}},"commit":{"parents":[{"commit":"e751ccfb270436a3a0d728701b61d7460352ab41","subject":"Add probe test for reconciler with object versioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e751ccfb270436a3a0d728701b61d7460352ab41"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 17:39:10.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-27 12:30:36.000000000","tz":0},"subject":"WIP object_versioning: only create one version of each .data file","message":"WIP object_versioning: only create one version of each .data file\n\nTODO: add unit tests for the change\n\nAvoid multiple versions of the same .data file by using the\nx-backend-data-timestamp to create the version-id that is used to\nuniquely name versions.\n\nPrior to versioning being enabled, object data is written to the user\nnamespace. After versioning is enabled, any such data (let\u0027s call it\nD1) that is overwritten is copied to the versions container before\nbeing replaced by a symlink to the new data (let\u0027s call it D2). The\nversion name is based on the x-timestamp of the copied data D1.\n\nWhen overwrites only partially succeed, a replica of the overwritten\nD1 data can be left in the user namespace. A subsequent overwrite may\nthen copy this replica of D1 to the versions container *again*. Each\ntime D1 data is copied to the versions container in this manner, a\nversion is created with a name based on the current x-timestamp of the\ncopied D1 replica. The x-timestamp may have advanced since the last\ncopy if a POST has been made. This can result in multiple differently\nnamed versions of the same D1 data.\n\nFurthermore, object_versioning doesn\u0027t attempt to preserve the time\nprecedence of PUTs and POSTs. If the PUT of D2 is only partially\nsuccessful, and then followed by a POST of metadata M2, the D1 replica\nremaining in user namespace may temporarily have the M2 .meta\nfile. This is eventually rectified in the user namespace by the\nobject-replicator (which replaces D1 with D2), but if another\noverwrite occurs before then, the copy of D1 to the versions container\nwill create another uniquely named version of D1 that erroneously\nincludes the M2 metadata.\n\nThis patch modifies object versioning middleware to use the\nx-backend-data-timestamp to create the version-id that is used to\nuniquely name versions. This means that each repeated copy of the same\ndata to the versions container is written to the same version object;\nthere will only be one version created for D1.\n\nThis change does NOT fix the possibility of erroneous metadata being\nincluded when data replicas are copied to the versions container;\nmultiple copy events may still occur (albeit now to the same version\nobject) and may include subsequently POSTed metadata.\n\nChange-Id: I15a59a35d857791140b2475cb70a6514e0f95e28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b1f506cdf48d28fc26c68e57929504b115a2820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b1f506cdf48d28fc26c68e57929504b115a2820"}]},"branch":"refs/heads/master"},"771c95d34e4679b79824ff4bb55e5709bd87fa8b":{"kind":"REWORK","_number":3,"created":"2024-11-28 16:34:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/96/936296/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/936296/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/936296/3"}}},"commit":{"parents":[{"commit":"e751ccfb270436a3a0d728701b61d7460352ab41","subject":"Add probe test for reconciler with object versioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e751ccfb270436a3a0d728701b61d7460352ab41"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 17:39:10.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-28 16:28:19.000000000","tz":0},"subject":"WIP object_versioning: only create one version of each .data file","message":"WIP object_versioning: only create one version of each .data file\n\nAvoid multiple versions of the same .data file by using the\nx-backend-data-timestamp to create the version-id that is used to\nuniquely name the version of any existing object that is copied out of\nthe user namespace to the versions container.\n\nPrior to versioning being enabled, object data is written to the user\nnamespace. After versioning is enabled, any such data (let\u0027s call it\nD1) that is overwritten is copied to the versions container before\nbeing replaced by a symlink to the new data (let\u0027s call it D2). The\nversion name is based on the x-timestamp of the copied data D1.\n\nWhen overwrites only partially succeed, a replica of the overwritten\nD1 data can be left in the user namespace. A subsequent overwrite may\nthen copy this replica of D1 to the versions container *again*. Each\ntime D1 data is copied to the versions container in this manner, a\nversion is created with a name based on the current x-timestamp of the\ncopied D1 replica. The x-timestamp may have advanced since the last\ncopy if a POST has been made. This can result in multiple differently\nnamed versions of the same D1 data.\n\nFurthermore, object_versioning doesn\u0027t attempt to preserve the time\nprecedence of PUTs and POSTs. If the PUT of D2 is only partially\nsuccessful, and then followed by a POST of metadata M2, the D1 replica\nremaining in user namespace may temporarily have the M2 .meta\nfile. This is eventually rectified in the user namespace by the\nobject-replicator (which replaces D1 with D2), but if another\noverwrite occurs before then, the copy of D1 to the versions container\nwill create another uniquely named version of D1 that erroneously\nincludes the M2 metadata.\n\nThis patch modifies object versioning middleware to use the\nx-backend-data-timestamp to create the version-id that is used to\nuniquely name versions. This means that each repeated copy of the same\ndata to the versions container is written to the same version object;\nthere will only be one version created for D1.\n\nUnlink the X-Timestamp or Last-Modified time, it is possible that the\nX-Backend-Data-Timestamp has an offset component. This is stripped off\nwhen making the version-id since ~Timestamp does not support\ntimestamps with offsets. The author is not aware of scenarios where\nthe offset component of the timestamp would be significant for the\npurposes of identifying unique versions.\n\nThis change does NOT fix the possibility of erroneous metadata being\nincluded when data replicas are copied to the versions container;\nmultiple copy events may still occur (albeit now to the same version\nobject) and may include subsequently POSTed metadata.\n\nChange-Id: I15a59a35d857791140b2475cb70a6514e0f95e28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/771c95d34e4679b79824ff4bb55e5709bd87fa8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/771c95d34e4679b79824ff4bb55e5709bd87fa8b"}]},"branch":"refs/heads/master"},"e2e3360f7dc748cbbcf084a8dd84f555652e4f91":{"kind":"REWORK","_number":4,"created":"2024-12-18 14:04:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/96/936296/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/936296/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/936296/4"}}},"commit":{"parents":[{"commit":"0b534d58465f9fdad9e6bedb96c9ecc77107b074","subject":"Merge \"Up-rev hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b534d58465f9fdad9e6bedb96c9ecc77107b074"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 17:39:10.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-18 14:02:38.000000000","tz":0},"subject":"object_versioning: only create one version of each .data file","message":"object_versioning: only create one version of each .data file\n\nAvoid multiple versions of the same .data file by using the\nx-backend-data-timestamp to create the version-id that is used to\nuniquely name the version of any existing object that is copied out of\nthe user namespace to the versions container.\n\nPrior to versioning being enabled, object data is written to the user\nnamespace. After versioning is enabled, any such data (let\u0027s call it\nD1) that is overwritten is copied to the versions container before\nbeing replaced by a symlink to the new data (let\u0027s call it D2). The\nversion name is based on the x-timestamp of the copied data D1.\n\nWhen overwrites only partially succeed, a replica of the overwritten\nD1 data can be left in the user namespace. A subsequent overwrite may\nthen copy this replica of D1 to the versions container *again*. Each\ntime D1 data is copied to the versions container in this manner, a\nversion is created with a name that is currently based on the current\nx-timestamp of the copied D1 replica. The x-timestamp may have\nadvanced since the last copy if a POST has been made. This can result\nin multiple differently named versions of the same D1 data.\n\nFurthermore, object_versioning doesn\u0027t attempt to preserve the time\nprecedence of PUTs and POSTs. If the PUT of D2 is only partially\nsuccessful, and then followed by a POST of metadata M2, the D1 replica\nremaining in user namespace may temporarily have the M2 .meta\nfile. This is eventually rectified in the user namespace by the\nobject-replicator (which replaces D1 with D2), but if another\noverwrite occurs before then, the copy of D1 to the versions container\nwill create another uniquely named version of D1 that erroneously\nincludes the M2 metadata.\n\nThis patch modifies object versioning middleware to use the\nx-backend-data-timestamp to create the version-id that is used to\nuniquely name versions. This means that each repeated copy of the same\ndata to the versions container is written to the same version object;\nthere will only be one version created for D1.\n\nUnlink the X-Timestamp or Last-Modified time, it is possible that the\nX-Backend-Data-Timestamp has an offset component. This is stripped off\nwhen making the version-id since ~Timestamp does not support\ntimestamps with offsets. The author is not aware of scenarios where\nthe offset component of the timestamp would be significant for the\npurposes of identifying unique versions.\n\nThis change does NOT fix the possibility of erroneous metadata being\nincluded when data replicas are copied to the versions container;\nmultiple copy events may still occur (albeit now to the same version\nobject) and may include subsequently POSTed metadata.\n\nChange-Id: I15a59a35d857791140b2475cb70a6514e0f95e28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2e3360f7dc748cbbcf084a8dd84f555652e4f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2e3360f7dc748cbbcf084a8dd84f555652e4f91"}]},"branch":"refs/heads/master"},"85d6f8a984dd0d7f075318a9d42f6d78e6bcc337":{"kind":"REWORK","_number":5,"created":"2024-12-19 10:15:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/96/936296/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/936296/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/936296/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/936296/5"}}},"commit":{"parents":[{"commit":"0b534d58465f9fdad9e6bedb96c9ecc77107b074","subject":"Merge \"Up-rev hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b534d58465f9fdad9e6bedb96c9ecc77107b074"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-26 17:39:10.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-19 10:13:19.000000000","tz":0},"subject":"object_versioning: only create one version of each .data file","message":"object_versioning: only create one version of each .data file\n\nAvoid multiple versions of the same .data file by using the\nx-backend-data-timestamp to create the version-id that is used to\nuniquely name the version of any existing object that is copied out of\nthe user namespace to the versions container.\n\nPrior to versioning being enabled, object data is written to the user\nnamespace. After versioning is enabled, any such data (let\u0027s call it\nD1) that is overwritten is copied to the versions container before\nbeing replaced by a symlink to the new data (let\u0027s call it D2). The\nversion name is based on the x-timestamp of the copied data D1.\n\nWhen overwrites only partially succeed, a replica of the overwritten\nD1 data can be left in the user namespace. A subsequent overwrite may\nthen copy this replica of D1 to the versions container *again*. Each\ntime D1 data is copied to the versions container in this manner, a\nversion is created with a name that is currently based on the current\nx-timestamp of the copied D1 replica. The x-timestamp may have\nadvanced since the last copy if a POST has been made. This can result\nin multiple differently named versions of the same D1 data.\n\nFurthermore, object_versioning doesn\u0027t attempt to preserve the time\nprecedence of PUTs and POSTs. If the PUT of D2 is only partially\nsuccessful, and then followed by a POST of metadata M2, the D1 replica\nremaining in user namespace may temporarily have the M2 .meta\nfile. This is eventually rectified in the user namespace by the\nobject-replicator (which replaces D1 with D2), but if another\noverwrite occurs before then, the copy of D1 to the versions container\nwill create another uniquely named version of D1 that erroneously\nincludes the M2 metadata.\n\nThis patch modifies object versioning middleware to use the\nx-backend-data-timestamp to create the version-id that is used to\nuniquely name versions. This means that each repeated copy of the same\ndata to the versions container is written to the same version object;\nthere will only be one version created for D1.\n\nUnlink the X-Timestamp or Last-Modified time, it is possible that the\nX-Backend-Data-Timestamp has an offset component. This is stripped off\nwhen making the version-id since ~Timestamp does not support\ntimestamps with offsets. The author is not aware of scenarios where\nthe offset component of the timestamp would be significant for the\npurposes of identifying unique versions.\n\nThis change does NOT fix the possibility of erroneous metadata being\nincluded when data replicas are copied to the versions container;\nmultiple copy events may still occur (albeit now to the same version\nobject) and may include subsequently POSTed metadata.\n\nChange-Id: I15a59a35d857791140b2475cb70a6514e0f95e28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85d6f8a984dd0d7f075318a9d42f6d78e6bcc337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85d6f8a984dd0d7f075318a9d42f6d78e6bcc337"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
