)]}'
{"id":"openstack%2Fswift~936387","triplet_id":"openstack%2Fswift~master~Ic6179dea20ca9c441ea2df3955bafdd159b1b659","project":"openstack/swift","branch":"master","topic":"patchFacade","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-01-07 17:18:15.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-01-07 17:18:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic6179dea20ca9c441ea2df3955bafdd159b1b659","subject":"sq? test for weird statsd_client_source robustness","status":"ABANDONED","created":"2024-11-27 16:23:09.000000000","updated":"2025-01-07 17:18:15.000000000","total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"12532517cd5ca8fba9655ae4d64565c7a0b2d857","_number":936387,"virtual_id_number":936387,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-27 20:17:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-27 18:13:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-27 18:58:38.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-01-07 15:06:48.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"CC"}],"messages":[{"id":"5af51f7208d75c54ad50d2975fed88422e6e9f94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-27 16:23:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cf1d0a672390e45d0f0e9c5f9ba9c3abfbb0f53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 18:13:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78614c095d6d4e90a936a7d7dc793ae2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72aeb1ee41294d379b7858c281501132 : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/ae2afaf0a8344bb2a9f7b8a7b18d6b0b : SUCCESS in 1h 08m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cda47453d0b14f7ba23c85f4728f700c : SUCCESS in 58m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/df3c70ca83804e71b1e564b7108c56b1 : SUCCESS in 52m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60c06fc88f344796af787661b0fefb3e : SUCCESS in 1h 08m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/343651e32df044a7bb4961b6af27b217 : SUCCESS in 21m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fe89afb7099456e9e5a7c895b908321 : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/869d828d249343008df235aa3a525091 : SUCCESS in 20m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b3fd7da25a74ac799f636297b2f34d0 : FAILURE in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ed5332ad861c49799a458f26bcc5c043 : SUCCESS in 19m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/44e6bab8c6254a648112cc11e7fda3f9 : SUCCESS in 17m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/64005c84405e4c679d57b1ca175bb363 : SUCCESS in 21m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6a31f637f50649a0a9e3ce32a08d37fb : SUCCESS in 27m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/341313dd5f5549659765ad4826163091 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c5c7ec9923984eb4adeff233f9b3c3f8 : SUCCESS in 20m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a072990de8e74fe5834c93bd2a001539 : SUCCESS in 18m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a9ecc72a5e0844349947eb523e1d25a2 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6c49c85c4f0244e98d0095c2b3d0abb3 : SUCCESS in 14m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/21d07c4000d242a7931b6cd616f16b84 : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/137c5979f2e645c89e6f4468e47c0b54 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/59d658d8e60d42dd859befc6ce0165b1 : SUCCESS in 14m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b97ec2efa604b1198c972fab817a471 : SUCCESS in 6m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e804612c6b3f46f3a49091dbcc846b71 : SUCCESS in 56m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bb219806321841a98b6da078e9a5653b : SUCCESS in 48m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ecf20ca8593f41c3b736ae53d1469b36 : SUCCESS in 52m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8de46baec7654d9ca40d3a7d8170f6dd : SUCCESS in 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca67893395e146c38e187351ff3083f4 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62733c73b50d4319ace502fbb9ca36b3 : SUCCESS in 23m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3dc91327d8c54eff8ada325aba5fd834 : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":1},{"id":"34f71123a1080a712f2bc3d3a0ef172439c4db0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 18:56:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6d80eb64d4e5e68c8353b226d9d5e4a0f284fdaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 18:58:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d6aef95ccac1611a7f9c7c923776ea030c7d51fe","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 18:58:38.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8025d2a4512398814cfe22a8d5ba7e4282c95483","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 19:00:40.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1ba15e70039e29831e3bf9469cdbf120b294d1d1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 19:04:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"21728f68b279cb420ec013262252418186504f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 20:17:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86beb0c22e0e4f98b3bc3658f460b98d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4572b129940c4e4ab59f320a638778e6 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/b24174f6371b4c29aa0d395d079a360c : SUCCESS in 1h 02m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/27539bc5769343c581ec56d0c3dcd71e : SUCCESS in 1h 01m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bafe968e0ebc42a3a1ee45fb305f8398 : SUCCESS in 53m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbf0a27302b5404bb4322eeb1ee553fe : SUCCESS in 1h 04m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28f655bcf30844198baa8ad7c10c2040 : SUCCESS in 11m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/442ffdd426eb4f3099c0005f3e6fbee0 : SUCCESS in 23m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8013b03f0e3046558d10be325e56e2b7 : SUCCESS in 12m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dbdc842f08d34520a868ffba200a7e13 : FAILURE in 3m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a44445718944f39bb633f36dc5bb1da : SUCCESS in 15m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/153368853fe54355bfc3723b7cf62472 : SUCCESS in 16m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a326e7d24aff469d8055810c72310f4b : SUCCESS in 20m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d035d52946b643fa8995b9fb7b854da8 : SUCCESS in 13m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/792482d126fb4cdd8f18d9a08fce6a9b : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/46585265d507480aa500296356e9df23 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e0f546b888d74759a10013ece1c155ae : SUCCESS in 18m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5e638bf6470142d1ab6601124100f12b : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f6e7a57b602b4e47b1d954e62c47a2ec : SUCCESS in 15m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5eb6473703cb4b17a7d2e0d72f01f6f9 : SUCCESS in 15m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b26358b9946e4d0abcef231eed194533 : SUCCESS in 5m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/32e587af1f04479ea3118a69ab5999f6 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/23dd94ad53d34d68a52c5983dc1a236c : SUCCESS in 7m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/705d869c6834413792bd496d0bb7b168 : SUCCESS in 56m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95640365da8d41dc98c943fa28bf20cb : SUCCESS in 55m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f99fe34b52044b34912b33f8901d1f8e : SUCCESS in 55m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abd0502047184cbe9db0314449c368f4 : SUCCESS in 20m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a94d140854f4075b744789562015b63 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7319ef944e114d67b897e7fb76555c05 : SUCCESS in 17m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16f671b559ba428b86918fa6eb9a7968 : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":3},{"id":"2d037dfa2e7078609f921c787cf5a37485984d13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 23:50:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af485e71a7b641daa0f7a3904ab3d5dc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0fdc6cf5be4f4642b70df085370124d2 : SUCCESS in 2h 36m 58s","accounts_in_message":[],"_revision_number":3},{"id":"240ecf6f32896d6922772d86d5e4553738ffbe2c","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-01-07 15:06:48.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"12532517cd5ca8fba9655ae4d64565c7a0b2d857","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-07 17:18:15.000000000","message":"Abandoned\n\nbetter in 936406: Improve get_logger tests re. statsd prefix | https://review.opendev.org/c/openstack/swift/+/936406","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"76db0cdb25e5d2c99680ee9abcb422b24c5de160","revisions":{"c834aa3dce51e932807987ca9561068fd4aa0950":{"kind":"REWORK","_number":1,"created":"2024-11-27 16:23:09.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/87/936387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/936387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/936387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/936387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/936387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/936387/1"}}},"commit":{"parents":[{"commit":"8699af83c9ee1cbc5691ed12a6f57fa1134aab41","subject":"Remove statds from the logs module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8699af83c9ee1cbc5691ed12a6f57fa1134aab41"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-27 15:59:51.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-27 16:09:04.000000000","tz":-360},"subject":"sq? test for weird statsd_client_source robustness","message":"sq? test for weird statsd_client_source robustness\n\nDrive-By: let get_swift_logger pick the default name\n\nChange-Id: Ic6179dea20ca9c441ea2df3955bafdd159b1b659\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c834aa3dce51e932807987ca9561068fd4aa0950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c834aa3dce51e932807987ca9561068fd4aa0950"}]},"branch":"refs/heads/master"},"80173916a380ed3f32147a3ab7bffa950124a74d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-27 18:56:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/87/936387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/936387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/936387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/936387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/936387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/936387/2"}}},"commit":{"parents":[{"commit":"af0c0608487bdf2fbad3f4f2e0c654b39fd2d68f","subject":"Improve get_logger tests re. statsd prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0c0608487bdf2fbad3f4f2e0c654b39fd2d68f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-27 15:59:51.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-27 18:55:53.000000000","tz":0},"subject":"sq? test for weird statsd_client_source robustness","message":"sq? test for weird statsd_client_source robustness\n\nDrive-By: let get_swift_logger pick the default name\n\nChange-Id: Ic6179dea20ca9c441ea2df3955bafdd159b1b659\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80173916a380ed3f32147a3ab7bffa950124a74d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80173916a380ed3f32147a3ab7bffa950124a74d"}]},"branch":"refs/heads/master"},"76db0cdb25e5d2c99680ee9abcb422b24c5de160":{"kind":"REWORK","_number":3,"created":"2024-11-27 18:58:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/87/936387/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/936387/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/936387/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/936387/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/936387/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/936387/3"}}},"commit":{"parents":[{"commit":"af0c0608487bdf2fbad3f4f2e0c654b39fd2d68f","subject":"Improve get_logger tests re. statsd prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0c0608487bdf2fbad3f4f2e0c654b39fd2d68f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-27 15:59:51.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-27 18:58:16.000000000","tz":0},"subject":"sq? test for weird statsd_client_source robustness","message":"sq? test for weird statsd_client_source robustness\n\nDrive-By: let get_swift_logger pick the default name\n\nChange-Id: Ic6179dea20ca9c441ea2df3955bafdd159b1b659\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76db0cdb25e5d2c99680ee9abcb422b24c5de160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76db0cdb25e5d2c99680ee9abcb422b24c5de160"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
