)]}'
{"id":"openstack%2Fswift~936866","triplet_id":"openstack%2Fswift~stable%2F2023.2~If724485e1425d1481d10b9255436301e346f07e8","project":"openstack/swift","branch":"stable/2023.2","topic":"fix-head-content-type-stable/2023.2","hashtags":[],"change_id":"If724485e1425d1481d10b9255436301e346f07e8","subject":"Ensure correct content-type in container HEAD response","status":"MERGED","created":"2024-12-02 17:27:08.000000000","updated":"2024-12-09 21:30:35.000000000","submitted":"2024-12-09 21:28:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936866-fix-head-content-type-stable/2023.2","cherry_pick_of_change":935817,"cherry_pick_of_patch_set":6,"meta_rev_id":"6d6930680f2521a3673c3b59c547876536594144","_number":936866,"virtual_id_number":936866,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-09 21:28:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-09 18:32:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-09 18:32:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-02 17:27:08.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-12-02 19:38:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35af5a70a3fc313d0b7d1f8fde565df2373c5ce4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-02 17:27:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fa6fdc1b63d077e67e701e4301b9e51e3462ae94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 19:38:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a3d4a4492f2416f9a585a32308f995d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b44ca51fd224d69b1dbc7d7b5a7ec5f : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/c70be9d1e49c4ce78bd8d5d0ccd1e3cc : SUCCESS in 59m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8064053c27e41fa88e42645d7841f37 : SUCCESS in 53m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6058b333cca945eb8a06fba177d7f385 : SUCCESS in 57m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1776050279ca4fbea8fdd217fd728faa : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed35131ae1f7457e95736fd526ebe9ab : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3299f621f6840329407cc9d9c9b9e22 : SUCCESS in 17m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a69913916fd44530b49de3470bd661c6 : SUCCESS in 10m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/038c0b0a08bc4ea9a75e41ed1b8c2dc3 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/07286ded53b4412fb94db6d82c264079 : SUCCESS in 17m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b9977fe39eac413bb2bc3740b7a2a2f5 : SUCCESS in 18m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/54871e9d28264ab782c4f86da534f49c : SUCCESS in 15m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bc4f910c52dd45639bd52863baf1031c : SUCCESS in 15m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ee4ddc0a0c6d4aa9b22aa393cc21bd63 : SUCCESS in 14m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6385be2ecb4d476a9e7bcf13e9f01326 : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/447cdd46114945b3a7bb6251646816f2 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/79bbea236893434a926bc0f771fa986e : SUCCESS in 16m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b7917fea3534404c9a6fb63800e359b2 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0e18afd35c084ebba00b6aba1b85b226 : SUCCESS in 14m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/59c87cd106db4a7bb7c39d7386b0be35 : SUCCESS in 17m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45dbd73c93c047578e47130d0540284f : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0090d46f57894ed3aea496c335d33aee : FAILURE in 13m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d803c6fc63aa4a22be4f7f39206b8919 : SUCCESS in 6m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/de8b2d9515564c9c9ec10438a129b357 : SUCCESS in 57m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/788c64d2adf345bc843ff22c40e92c40 : SUCCESS in 33m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3078d64236d748db92ab1f0427ea6470 : SUCCESS in 37m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6202fb67c9141408ec5120bf79dadc1 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3a359a63b4400ebaa094289a6035ff : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c202621ff0854aa896d57a2620d8df92 : TIMED_OUT in 1h 32m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19559b7797dc47028375142c85dc2f58 : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f04eff733b1f3159ba2d991517904201117ae58e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 19:45:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eb7e1408f254d398af4184b26692531\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db445953b7a84f608a64c0545cee62d3 : TIMED_OUT in 1h 30m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0df68c2b59b646978f18a8bbee65056f : SUCCESS in 1h 08m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc28aa524a7f417ca2e6c58011eed1db : SUCCESS in 1h 15m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2544e09c9e4e4077936acdec258f79ab : TIMED_OUT in 1h 32m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8fcab4ae45ad88fbfac8ef17f616c63b524164f1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-09 18:32:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e744d90333fe3e23e91e9638cdacef8a4fe8ef0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 18:50:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9bbe42f5cc03060bf87a04979ebb46b3631b374c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 21:28:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2b3503a0c8446ffa6dcaef8457f60ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c3fa9ffdd904f20b4b5da8ba737ab30 : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/1330b6975d93499994a86ff3009070d4 : SUCCESS in 52m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3d7ff727a7b45edac08223e5572dc83 : SUCCESS in 48m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27edf9664b764a4b81fad007624a1451 : SUCCESS in 1h 02m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a02dbad0e9564a6db9d6d865e05e10d8 : SUCCESS in 13m 34s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ce6ed32898384de3b3030bd180c75504 : SUCCESS in 7m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28e12e5a49964f85bc472f3616981634 : SUCCESS in 21m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef163e6f5a9842a5b4be3a8b06de65aa : SUCCESS in 16m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7614fe2caef948f9b540167b8a95c4f2 : SUCCESS in 16m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a1975c1165d5429a8aff5d6312bf4a28 : SUCCESS in 16m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b14b5b072c9d477ca0cd95a503ad50ae : SUCCESS in 17m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/29bff97c38454e28a2860d3699ed0cfa : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fe428f813d394e8f89918dcc05ec89fa : SUCCESS in 19m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c5f625de8e7d473f9bdf8933b1f5b2d6 : SUCCESS in 18m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a1b472d6049a476ea41bae65e1a764e3 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/21f3f631a214447b80ea07ba12fa9bdb : SUCCESS in 15m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f96ded481d434e8bb807e90493066fdf : SUCCESS in 17m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a7771d76ef64edaa14d3cb7c27e2aeb : SUCCESS in 7m 15s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c6bdde696ef143f78690f4538f340a3b : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/930c8662b60547238a9e6ce555d074bd : SUCCESS in 52m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d7369057981448769aec8b006f005d3f : SUCCESS in 33m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89c1ce3fb6884d1fa7290219fb508730 : SUCCESS in 47m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dc182a152d6470384553120dccd0f40 : SUCCESS in 19m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d5a3d5285748e2a91f89b5ab1eae45 : SUCCESS in 7m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a68707f7e8c1420eb80dfa32a098abac : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d7946b49dfb94baa713c56818d573c910066ae2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 21:28:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6d6930680f2521a3673c3b59c547876536594144","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 21:30:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fff7e3a47daa4e83950508389ba0a7ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80699bd4d3c7422e990594e4065799a7 : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/40ea2b6ada5b47788e2f17fe1039ceb6 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/927188fa7dd74812ae426447644228c0 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b47d0cf53e6151354168f007f7d8255dd65eb854","revisions":{"b47d0cf53e6151354168f007f7d8255dd65eb854":{"kind":"REWORK","_number":1,"created":"2024-12-02 17:27:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/936866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/936866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/936866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/936866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/936866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/936866/1"}}},"commit":{"parents":[{"commit":"0de976cf5636e6572660181ea196c3f1da5f6454","subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0de976cf5636e6572660181ea196c3f1da5f6454"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:59:43.000000000","tz":-360},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-12-02 17:27:08.000000000","tz":0},"subject":"Ensure correct content-type in container HEAD response","message":"Ensure correct content-type in container HEAD response\n\nA failing CORS test in the gate discovered that we were responding\napplication/json to ?format\u003dtxt requests (which is maybe not even a\nvalid value for that qs param?), but only when running with\neventlet\u003d\u003d0.38.0\n\nThis avoids the problem of backend container server HEADs no longer\nhaving \u0027Content-Length: 0\u0027 by fixing the client HEAD resp headers before\nwe check for chunked-transfer resp.\n\nDrive-By: refactor listing_formats to use HeaderKeyDict and always set\nContent-Length explicitly\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: If724485e1425d1481d10b9255436301e346f07e8\n(cherry picked from commit fa889358acf7675efba2de095777886b4c1ff7f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b47d0cf53e6151354168f007f7d8255dd65eb854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b47d0cf53e6151354168f007f7d8255dd65eb854"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
