)]}'
{"id":"openstack%2Fswift~936882","triplet_id":"openstack%2Fswift~master~I6e119cc9c7297d8aade9736fa1d6f4a105466d77","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-03 10:49:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e119cc9c7297d8aade9736fa1d6f4a105466d77","subject":"tests: Use format\u003dplain instead of format\u003dtxt","status":"MERGED","created":"2024-12-02 20:26:00.000000000","updated":"2024-12-03 10:50:45.000000000","submitted":"2024-12-03 10:49:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936882","meta_rev_id":"52f5322510735046711d9cbf20fde455f8a477f2","_number":936882,"virtual_id_number":936882,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-03 10:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-12-03 09:20:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-12-03 09:20:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-02 21:29:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-03 09:20:19.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"6da40926e8f9a247a476581fcea7f69a02e45286","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-02 20:26:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12bac80e655f1b556138c26948598cd7fa013e2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 21:29:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e9b5a5ecc914c0fa172a0d20a8a48e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b5d437c83a4bc5a9bf171fd1bf9cad : SUCCESS in 10m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5dae4ee8299048949bed6e6814d8d2bc : SUCCESS in 37m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd1e6cc7cd14451a87f2ae7e2ce29bba : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4b2ee710d2544f4b87b07ce43db0f61 : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76ed6f762dec4161921953c270eef52e : SUCCESS in 16m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e54b741b7a004ddb84a11bd5531042ce : FAILURE in 3m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f7f1e3ce7c304025a7362ab2f3ec3f9d : SUCCESS in 13m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dce305d1b54541c89e9fbad67c2a2b2a : SUCCESS in 17m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ccb0cc59f5547a492b7d46f2fb2351c : SUCCESS in 19m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/444af5627b40425faa10a30274478164 : SUCCESS in 34m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e6141682fbb0467e84fcea4d09ff6012 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eddabde8c3f84fd4a75d4dcee079dbd4 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d228bc7f41274a7cbfb7eefcc8811bea : SUCCESS in 18m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4d2c4a0d2d79412d95213d321a8f17bf : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d311c5434aa441abbb700744982bc7c4 : SUCCESS in 13m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3e082e0c156d47388dc84ec2f911a99f : SUCCESS in 15m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fe7f5bc93a314aa586bae14f81c53eae : SUCCESS in 7m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f325dc6ce7b42b883602e80f3df80ac : SUCCESS in 23m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f53162f7174b43a4bc690eb382eb0b4a : SUCCESS in 53m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65d633634ae545998d67050186081b88 : SUCCESS in 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60e3bf17b7734b78b8ae23296a15ac2c : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18a0728b649e4af692f8cad7dfaa7be6 : SUCCESS in 25m 35s","accounts_in_message":[],"_revision_number":1},{"id":"121491a82948f0123ed386a1a45286503ace09ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 22:04:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaeca5c85ad54b099113747ca5fbb327\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad203849ae674ef898afc357beeaf5f2 : SUCCESS in 1h 36m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1779b4c6a7f085be7d343b2062394e551ebee4c5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 09:20:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"548cea87c61e05d059869171e66f0df1f59965f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 09:20:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0026f2c20301c8457d327eb0a34982599e598725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 10:49:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1c47c4e7f87427598687aa2045c4a9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f69b14c98a34a3d8da32a1ef6ca40c3 : SUCCESS in 9m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d797171b31b1455788e6087d0a850711 : SUCCESS in 53m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/09c6b50ecfd0407abd118f3c1b5da294 : POST_FAILURE in 7m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eee9dc197bdb4a73af7c35589ca5297a : SUCCESS in 19m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/40f58b4965764daab608b71659178ec2 : SUCCESS in 21m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/27d27320567143cea1bf3e1f9f06e926 : SUCCESS in 20m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dca90a1df4a943b99b60a4d02ca8bc92 : SUCCESS in 22m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/62c100b659db4e948de74a9d740e3458 : SUCCESS in 21m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c48e613fc96047e1ae4822683c492bfa : SUCCESS in 17m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/87cac16b6c2848139578e321c0959e8a : SUCCESS in 19m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/731e29a12c6740d58b6111f1a5314146 : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e1c77df9740e49b4b84bc44998c0c35f : SUCCESS in 18m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/90901a43f6ca48439a6e8d756161eae6 : SUCCESS in 18m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b26980345e8046ae9ef60a8c4338c291 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/472cdbeb288c434dab701f69251dae82 : SUCCESS in 7m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99dd13ee57094730905703a41255e00b : SUCCESS in 46m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e042b95d3cdc43f7bca629517fe6a041 : SUCCESS in 43m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94b70166226f4df6a7e2113a493272c6 : SUCCESS in 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42af269cfa8b4111a097026bb005d1a6 : SUCCESS in 3m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf4597cc6b154601adf28fbcfba44d2e : SUCCESS in 18m 21s","accounts_in_message":[],"_revision_number":1},{"id":"9ecaf68b3ad262004ff22a3c33390576f1be6edb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 10:49:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"52f5322510735046711d9cbf20fde455f8a477f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 10:50:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0721c9e887a146b7b276c177a6507af1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48764d78fab456488c01b5d956caa2d : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dca84267de37469fb7daa5db7a1bc399 : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ace2357c62c51ea2b684fe8f9fe1e25db4cf4bc3","revisions":{"ace2357c62c51ea2b684fe8f9fe1e25db4cf4bc3":{"kind":"REWORK","_number":1,"created":"2024-12-02 20:26:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/936882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/936882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/936882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/936882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/936882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/936882/1"}}},"commit":{"parents":[{"commit":"89815389d5376195d9bc1b0bf65f88e9fb7ce637","subject":"Merge \"probe tests: Set default timeout for subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89815389d5376195d9bc1b0bf65f88e9fb7ce637"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-02 20:21:16.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-02 20:21:20.000000000","tz":-480},"subject":"tests: Use format\u003dplain instead of format\u003dtxt","message":"tests: Use format\u003dplain instead of format\u003dtxt\n\nOur API ref says \"Valid values are json, xml, or plain. The default is\nplain.\" There\u0027s no reason our tests ought to use an invalid value; it\nwill confuse anyone looking at tests for how to do a thing.\n\nNote that tests were passing because invalid values are ignored, so\n?format\u003dtxt behaves exactly like ?format\u003dplain.\n\nChange-Id: I6e119cc9c7297d8aade9736fa1d6f4a105466d77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ace2357c62c51ea2b684fe8f9fe1e25db4cf4bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ace2357c62c51ea2b684fe8f9fe1e25db4cf4bc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
