)]}'
{"id":"openstack%2Fswift~936907","triplet_id":"openstack%2Fswift~master~Ib868258b224d97ebd1c14d5b5b98bb0718caaea9","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-03 10:49:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib868258b224d97ebd1c14d5b5b98bb0718caaea9","subject":"api-ref: Clarify eventual-consistency of DELETEs.","status":"MERGED","created":"2024-12-02 22:39:09.000000000","updated":"2024-12-03 10:51:48.000000000","submitted":"2024-12-03 10:49:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936907","meta_rev_id":"b06ee07274c9d1e232b7f43f3ad7be065bb8a76c","_number":936907,"virtual_id_number":936907,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-03 10:49:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-12-03 09:21:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-12-03 09:21:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-02 22:54:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-03 09:21:54.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"5504296e15974c9cd2aef1059b5efe16ed04d098","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-02 22:39:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc742b4156fc34e74303fcf5bd5368390d4c9945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 22:54:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7d4055a81fe4e78bdcbf5f8d68e3eb5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8fe9a20ba0004c83948ce37ac803297f : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef57016d480d484a8587bcd1514db2ba : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1220f67ebe8d7c4a176c76a3d35acb0f7636d1ed","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 09:21:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b0470efcfc3adb1cd21a0aa50af2440c02f7b9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 09:22:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed53dac47138b63d3b40b8b3c6d7e111e899b16a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 10:49:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/934d32ab9df94c33aec92c4a5813446e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/50fc052ed0f84b0a8a76717d11ae761a : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c361ccdd8742178c2fd875032e4f1f : SUCCESS in 11m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f35db0de3dba45c3be4741b1f19cd83d : SUCCESS in 5m 59s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6fa46a472d52465e871598aaebfcc898 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1},{"id":"df94f4a81341fe2ffc6aa5cc6a55e43b5815d88f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 10:49:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b06ee07274c9d1e232b7f43f3ad7be065bb8a76c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 10:51:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7581e1a91f254ceb98d95838a36fc1e8\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0628b808bd024995a72639ba143984a0 : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3f4f811b29427c88d5e7f817f9e179 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/29eaa7d3245344fbb5e76e57b1a4aac2 : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6b712dfa2a49b9bf8eef25d228056f980f1865b","revisions":{"d6b712dfa2a49b9bf8eef25d228056f980f1865b":{"kind":"REWORK","_number":1,"created":"2024-12-02 22:39:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/07/936907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/936907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/936907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/936907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/936907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/936907/1"}}},"commit":{"parents":[{"commit":"89815389d5376195d9bc1b0bf65f88e9fb7ce637","subject":"Merge \"probe tests: Set default timeout for subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89815389d5376195d9bc1b0bf65f88e9fb7ce637"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-02 22:38:37.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-02 22:38:37.000000000","tz":-480},"subject":"api-ref: Clarify eventual-consistency of DELETEs.","message":"api-ref: Clarify eventual-consistency of DELETEs.\n\nChange-Id: Ib868258b224d97ebd1c14d5b5b98bb0718caaea9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6b712dfa2a49b9bf8eef25d228056f980f1865b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6b712dfa2a49b9bf8eef25d228056f980f1865b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
