)]}'
{"id":"openstack%2Fswift~936930","triplet_id":"openstack%2Fswift~master~Ib8196fe24f8a999315af469a435bd639378c78a9","project":"openstack/swift","branch":"master","topic":"p-empty-etag","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-03 18:09:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib8196fe24f8a999315af469a435bd639378c78a9","subject":"Remove duplicate definition of empty string etag","status":"MERGED","created":"2024-12-03 10:47:18.000000000","updated":"2024-12-03 18:09:52.000000000","submitted":"2024-12-03 18:09:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936930-p-empty-etag","meta_rev_id":"433acd47d19e442a2e5252c8aa419b750d236b17","_number":936930,"virtual_id_number":936930,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-03 18:09:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-03 16:45:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-03 16:45:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-03 11:58:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-03 16:45:26.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"cf6b05185a425cdbcd4bab2c1a8ad1935d17e80a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 10:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"246b0e4f8ebe2d5b5303b53bdfc571d6057096db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 11:58:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f37840debdea44348731fcdcf058b457\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db797513307467d96191f87ef306dd3 : SUCCESS in 9m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/e86c322580c64e65b7741f7b4041b8a8 : SUCCESS in 1h 00m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/388eb2b69e824627801d52efb84ffef9 : SUCCESS in 1h 07m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/22eebc2dbf574e96b3488fa57fbe1069 : SUCCESS in 33m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df37a24513b64f26b701612f49af288f : SUCCESS in 1h 01m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5cab78550706482abc8060333fb4aa6c : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8eab0bf7cefe4277ae880f6245befd65 : SUCCESS in 12m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef7e8f53395346538416d651e72f4db4 : SUCCESS in 25m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dc562521c81545c7b66b5445ff9ead2a : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/26d00c0582bf48a9bf9fddaba289af7b : SUCCESS in 23m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c47c5083bf74f07b37c0743aad797f9 : SUCCESS in 16m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/011eb810f1ba47bcbd3c213bde2eaf43 : SUCCESS in 18m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/37205e5c442c4f54bce290a8c1bbac16 : SUCCESS in 26m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e9996f0c1a724986a554351056f085de : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2506cd02dfd744ea855d87f77d3f42a2 : SUCCESS in 18m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8f38766747cf455a81d47157a7f843af : SUCCESS in 17m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a7ff0237a2374821bf01711aa8535496 : SUCCESS in 13m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d7ee88fab94f495a8930b53c014362cf : SUCCESS in 13m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6628ef31ae3a41a49885360345ddf2f0 : SUCCESS in 12m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a0f89a201a6f47319540e59f4c9eaf6e : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e190a58bcd24a519ee178e07f1b89a8 : SUCCESS in 13m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/43c0714369ab475cb301a1d41ed5b976 : SUCCESS in 4m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/212439d445d742ceadee63ec1abf4dd3 : SUCCESS in 54m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cdef6a0758994b4a8b2b99c000f29a37 : SUCCESS in 53m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4cea037d23bb44ee86b2c938789a98f7 : SUCCESS in 30m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06ce2785052a4280a13a3c12f878d107 : SUCCESS in 19m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd56774621e4710970873a8f9e30749 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8af1c8f9c9241ee84ff5056269fb6ec : SUCCESS in 23m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/502ca173e67246dd817fd5bfe609890b : SUCCESS in 39m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13c9f3e88c73490dbd604eb6c1039575 : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0072c16b1f7cdeef261c64909e50ea1323f22761","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 12:09:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8298bf2f4a44d9795f6a087d0625f27\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/628fd8276f73407c808a714efcac152a : SUCCESS in 1h 20m 24s","accounts_in_message":[],"_revision_number":1},{"id":"565ca06d1989d8757985e62b1aafaf4c95506d75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 12:57:13.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cd7896bf3b5c6ebae17aa31c8e331b985f9d6e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 14:08:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d444c20f5dc40d9b5de0bc2ddbbf1ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a515b25453c4cdbbdf3dd16950d8e73 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/a53981d304ee41f6bd168ee19f3900ae : SUCCESS in 1h 05m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62c691aafc4743b08ef54fe2a1a66746 : SUCCESS in 48m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0afbc49b870e492bb224d99f70ef0229 : SUCCESS in 47m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36b6d69f6ebc422aad22c1c1487e1c05 : SUCCESS in 52m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6899c4b47d444a2b29f2e92fb0fae83 : SUCCESS in 11m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c895b9c977054d90b8452cbe0bf0e62a : SUCCESS in 12m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/224e22c90ad942758cfbc01d7ef1737c : SUCCESS in 12m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f27b15c704cb42fcbad3577a8a8df90b : SUCCESS in 4m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38d0c429152846d7804479e23a25268d : SUCCESS in 21m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b39fef80dc4446c4b0357e3d597e8269 : SUCCESS in 18m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e5549c233f7e4f52a5efbb442d535452 : SUCCESS in 9m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/184efefdccd645b8bd2774d62aa3d058 : SUCCESS in 30m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/126783dfcb3b41a6802bf71332bf05f5 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f6e6b6f44c804a1795777382a9b71feb : SUCCESS in 18m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/855c0e4f7d5e4f3980a7786579842212 : SUCCESS in 18m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b3863b2a586f4a74b8b7a8441018f4a5 : SUCCESS in 8m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c9c18b4cb5bf48ac8831c8cda7543e11 : SUCCESS in 17m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f5083b0f6af747cc8fbc419a6fe85120 : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2847f8d694634d8bb3906842a76bf050 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/50d2c442145d4267adea33af9544c75a : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8b22a8e78f5040c7970219f3aa0231c8 : SUCCESS in 4m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/654a35b6ca1049bca876372f94bf1e65 : SUCCESS in 1h 01m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e6451ea52c5e41feadaa3f8086c76295 : SUCCESS in 31m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7b3ecf8551943aab7f560005b304bd5 : SUCCESS in 39m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32fb3408e1444790bcf81fdf458fe356 : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d5b19e243b4c1f95f0e4498a860e31 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a9c67a8e37a4368b30ee54aef723320 : SUCCESS in 17m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/298d14b6d5db4bf28e284c0b7cf62aa7 : SUCCESS in 38m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/475849280b4e4728936665ce19de1bb9 : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":2},{"id":"79a6611b7ee56c2c9b9baef46659bcd97936845a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 14:16:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5816c7d8f6541aebe0723f9ab26719b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/602cd8ab5ee24fa4bd5683d4cbcea5d3 : SUCCESS in 1h 17m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4f19f53e10fdb077ff06211b685fc67a8ff07e40","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-03 16:45:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ea5b81dd93c92ff83896b733d60035f286c374b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 16:46:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b73b49894234a0502bf698fcd1b5a4460f01329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:09:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3adf6a95affb40b2a76c892f80f3d89c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b49ab5bee2054199a5ea99eb9ab1e5cf : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/91489434131d49c281aa5117391689fc : SUCCESS in 36m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee19d4db6dd149499fb56892658da304 : SUCCESS in 42m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fc44b089152a45d3861d7a8975e33d2f : SUCCESS in 40m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91baefcd247c4a7193fcced2fe363b86 : SUCCESS in 1h 04m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c4462e14b5d94611b3497017a77bdf75 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69dbad6c485e49c4a4081985020c1a0f : SUCCESS in 19m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/771f568b79a44e8dbd52eeaff8bd6ad0 : SUCCESS in 19m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7af60eaf7aeb4c009bab2ae3a191053f : SUCCESS in 19m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ba244d312efd4dcaafebe1f1a1aaf48d : SUCCESS in 20m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/55e730e76adf415685f3da2623fbda7d : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1cd7721a40e34eb180651e4c2ea71f00 : SUCCESS in 17m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8beb6932b8c8418c933c2f2dc6cc7cf7 : SUCCESS in 15m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c4cb6e650355463eb2b0065fe8d1518a : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/51b420c01bb040128b112b5569965da2 : SUCCESS in 13m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4f8a65c896e44edfb3420ab91755c8eb : SUCCESS in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa96171c1dfe4661b7496a32939280dc : SUCCESS in 7m 22s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/58c2c13481964203bedfa250f17c80a4 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b8122c7e69042b5ad567abd9cd188f1 : SUCCESS in 54m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d373d090e4374858b13324e7bf004a5c : SUCCESS in 40m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40cede71a08842ce8a9dbb951ec6d635 : SUCCESS in 24m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ea843b8134a4d4ab72bb6f7e6b337ae : SUCCESS in 22m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d91dcc1c9a17465cbd2e9fcabde4d3f3 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/25894a6fe999477081ce400c432399bb : SUCCESS in 22m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46f3d656527640d687de59348f705450 : SUCCESS in 52m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74b1d85264694c589efe7281c3c24570 : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":2},{"id":"433acd47d19e442a2e5252c8aa419b750d236b17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:09:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5e800e328eca15a1a42bb8da62775956391fa32f","revisions":{"0548dd76689b994e765ea13c02add6da506e2a96":{"kind":"REWORK","_number":1,"created":"2024-12-03 10:47:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/30/936930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/936930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/936930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/936930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/936930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/936930/1"}}},"commit":{"parents":[{"commit":"2dad4dcbbd71191d3ec014f93eeefd4d2013d293","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dad4dcbbd71191d3ec014f93eeefd4d2013d293"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-02 12:27:11.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-03 10:42:34.000000000","tz":0},"subject":"Use MD5_OF_EMPTY_STRING everywhere","message":"Use MD5_OF_EMPTY_STRING everywhere\n\nRemove duplicate definition of the etag of an empty string.\n\nChange-Id: Ib8196fe24f8a999315af469a435bd639378c78a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0548dd76689b994e765ea13c02add6da506e2a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0548dd76689b994e765ea13c02add6da506e2a96"}]},"branch":"refs/heads/master"},"5e800e328eca15a1a42bb8da62775956391fa32f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-12-03 12:57:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/30/936930/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/936930/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/936930/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/936930/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/936930/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/936930/2"}}},"commit":{"parents":[{"commit":"2dad4dcbbd71191d3ec014f93eeefd4d2013d293","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dad4dcbbd71191d3ec014f93eeefd4d2013d293"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-02 12:27:11.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-03 12:56:34.000000000","tz":0},"subject":"Remove duplicate definition of empty string etag","message":"Remove duplicate definition of empty string etag\n\nChange-Id: Ib8196fe24f8a999315af469a435bd639378c78a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e800e328eca15a1a42bb8da62775956391fa32f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e800e328eca15a1a42bb8da62775956391fa32f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
