)]}'
{"id":"openstack%2Fswift~936973","triplet_id":"openstack%2Fswift~feature%2Fmpu~I157e524833d868bfcfaa95478ebad61b40949ea8","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-06 11:47:28.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I157e524833d868bfcfaa95478ebad61b40949ea8","subject":"merge master to feature/mpu","status":"MERGED","created":"2024-12-03 18:16:38.000000000","updated":"2024-12-06 12:49:38.000000000","submitted":"2024-12-06 12:46:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936973-merge-master","meta_rev_id":"78819a523b57bc905a9844030dbb8f917dcfe75d","_number":936973,"virtual_id_number":936973,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-06 12:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-12-06 11:47:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-12-06 11:47:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-03 19:38:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"631d234d0044a29285dc777199c424d703d654cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 18:16:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e01e9a487bc2708714a034c81bec5495d32c831","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 18:22:05.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af1ed12bada7cc896d84e7182d07e235d744f680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 18:38:14.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4e2569e2be96129943ccfe0d650337b5e297f519","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-03 18:42:10.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"30da4327ae0861b584e658fddab726e904c1d47b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 19:38:40.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdafa22587fb4ee387d1c3466d0f451c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50b2dd0227f1403796ce5c3b4c6b4d3c : SUCCESS in 6m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c819eb0de414241b9da32c6ba824a92 : SUCCESS in 11m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20fa3b8d9d7a4e6f846cb36df7ad750f : SUCCESS in 11m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e63528165d8e428db250cda868bc9732 : SUCCESS in 12m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/00702edb4cee471aab4bd0fb82d79efd : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d398ff1bfd7845ebbfa3dbe7ed38544a : SUCCESS in 10m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/347b250da925474ca1aa8f96b5a8d6dd : SUCCESS in 18m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ad341e90204740d98dee3f1634255b75 : SUCCESS in 15m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f8f1e203d013488d9b51f4c16a755c9e : SUCCESS in 31m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2386df488e9d400f883058a156980920 : SUCCESS in 13m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/02b3863f108346ca9f9087fb978af891 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3a0deef194e34cfdb504bb3da249d190 : SUCCESS in 18m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/54c6e8ca22e54560a5203fa2453c4186 : SUCCESS in 9m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/46db3f2e5f324ad7b16658a3ff34c495 : SUCCESS in 8m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c39afd51abab4e29bdb3cda05a13b26e : SUCCESS in 17m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ef26dde222cb430b882b98d697460040 : SUCCESS in 4m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6af5fbffbd484bfb8dbf6764aa617fb7 : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3800f532010d40169eca83eb93dea9c8 : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3b14b90f3d344ff8dbdb7edcf173084 : SUCCESS in 53m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bceb89b15a540e6b3f679ce12f9c359 : SUCCESS in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8683c9a906a34b6b9a7ecdd2b52ed7d1 : FAILURE in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/37ebe085ce3949d2a85ffb15ffb9b73a : SUCCESS in 24m 43s","accounts_in_message":[],"_revision_number":2},{"id":"ad86f514a60b586dadec8378b86f68b0c08a720d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 19:58:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b394da9d1ee14f54830e769ae3118eb5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fb059c46653b49688eac9814b7d6c35c : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":2},{"id":"27e10d5d19fe1870762b5436c9e5e6a5f6f5243b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-04 10:23:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4eb7da83853e394b61961888f26f76071ec3bef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 11:32:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef2e390774064544af6b66eec7757501\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb8d76b0362645bdad98c33844d6fcce : SUCCESS in 7m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba09836b75ce46c8bf121e0e2c4b13eb : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/067dd912182c42bab2d54e6bf5844f31 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5d145bc8f67456b85fa38c2ea207cbd : SUCCESS in 20m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b1573f1653d4180873fb534bf484381 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/784681a136bb4fefa567bb28ba2589e7 : SUCCESS in 18m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/240bf94ae34e47d2a7e5d4d571816a62 : SUCCESS in 18m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c06b34a3b5244b41bac11dc4a8adebe2 : SUCCESS in 21m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/81b6dc9815114225b5a8fd71ec6acd56 : SUCCESS in 23m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/31121d91185146d88238c5bccdf31ff6 : SUCCESS in 19m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9d96888842b944ee98b0a42aa693f84a : SUCCESS in 18m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/670d8fed213f4ec581d2fcb98618df85 : SUCCESS in 18m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8099257c5350499a94ebf3f693e12fa0 : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dd3ca5625a734dda8cddc61360bf3d6a : SUCCESS in 16m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ace384c746e94f36b55d464ce17f6a14 : SUCCESS in 8m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03ca2052a868400b96c661328e36eff3 : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/57088a793c814813800859a3bde7b285 : SUCCESS in 16m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c21fc7a459164bf79fd5ac4bbd3ec60a : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fd9f68eb1324650bff7c308e516c52f : SUCCESS in 1h 03m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c73834e914fa47a39be9f344b8bce676 : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ea2c822d8ad4262abb0084b1169c41c : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9bae879840954ab3a1d3e0355886303c : SUCCESS in 18m 18s","accounts_in_message":[],"_revision_number":3},{"id":"5d8e3c0a7e9a424f5c42d8fdcff07c2735268358","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 12:03:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e18472a70ac4ca296a69e241ca1d870\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e2e32dfc303419492512d6342ff9550 : SUCCESS in 1h 36m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7448eaf19087af01132db72f429ab36229f3cc2b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-06 11:47:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1ad939e94c112a835e242d24df86c9b8a11a6ada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 11:48:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d2bec73937a36ceb07b3ec9919a4ddcaacfa11c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 12:46:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a225b9ebdfec449a8615821692138ccc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae5a1f3209134f7eae018fdda26b20e5 : SUCCESS in 8m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6ea83911e38a4ec686d136a02fd15506 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c528a42a46f4611a0b5d1a34e83ff40 : SUCCESS in 25m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ba57c3473d84da3aed52e8f95596a53 : SUCCESS in 27m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f19acf485c144328a1d796990774536 : SUCCESS in 25m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8ddeb2a7b35f419db523474eb718862f : SUCCESS in 28m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/99c8118711d34351b2aa8988d50d5f61 : SUCCESS in 11m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/65ab5dba426848cc9a66a7030f6edf13 : SUCCESS in 18m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/98a3cd7cda47403486433e5df4018b92 : SUCCESS in 17m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/10e7bd9469d84364a83a521e43e598c0 : SUCCESS in 12m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1cc794d3ced34848bc49ef481059d0da : SUCCESS in 17m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c1b9a017b2a243f397bb680d66c10d8d : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6c333584d1e149cc8fd02a1f057cf286 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a23bf3040add498eb2fed0e35c76900f : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ba2027f27874e58a621038419e3bd43 : SUCCESS in 52m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da2087c412824f5f8d98991560301638 : SUCCESS in 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82d46e47a41e41aa9d27ac6e3dfebf58 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3305542396fb4ab097e3f39d92e165fa : SUCCESS in 17m 16s","accounts_in_message":[],"_revision_number":3},{"id":"ed7f8477476916f1f553a147daeb138d60f4d723","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 12:46:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"78819a523b57bc905a9844030dbb8f917dcfe75d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 12:49:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e89af91900bd4090ac30408ad87dc16a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b378e2c5ac645c4ad9566261be9d950 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e30af994438447c4aeb6a51da080a35b : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e991fa788f317d3cb4d87d7e549bd240575b44d7","revisions":{"9c6f735e77caec5324efd3d12330bb49cfe25140":{"kind":"REWORK","_number":1,"created":"2024-12-03 18:16:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/936973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/936973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/936973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/936973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/936973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/936973/1"}}},"commit":{"parents":[{"commit":"e1588ee50ea6a58a59cac0b6f3ea6be9f15b3357","subject":"mpu: include parts bytes_used in user container bytes_used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1588ee50ea6a58a59cac0b6f3ea6be9f15b3357"}]},{"commit":"1d240aa86c8f2e637452bf7be2d04dc99897626b","subject":"Merge \"Rename probe/test_mixed_policy_upload.py to test_mpu.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d240aa86c8f2e637452bf7be2d04dc99897626b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-03 18:15:38.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-03 18:15:38.000000000","tz":0},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: I157e524833d868bfcfaa95478ebad61b40949ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c6f735e77caec5324efd3d12330bb49cfe25140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c6f735e77caec5324efd3d12330bb49cfe25140"}]},"branch":"refs/heads/feature/mpu"},"03b748ae156566b606bf2bc708795fe2050d22f1":{"kind":"REWORK","_number":2,"created":"2024-12-03 18:38:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/936973/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/936973/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/936973/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/936973/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/936973/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/936973/2"}}},"commit":{"parents":[{"commit":"e1588ee50ea6a58a59cac0b6f3ea6be9f15b3357","subject":"mpu: include parts bytes_used in user container bytes_used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1588ee50ea6a58a59cac0b6f3ea6be9f15b3357"}]},{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-03 18:32:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-03 18:35:35.000000000","tz":0},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: I157e524833d868bfcfaa95478ebad61b40949ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03b748ae156566b606bf2bc708795fe2050d22f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03b748ae156566b606bf2bc708795fe2050d22f1"}]},"branch":"refs/heads/feature/mpu"},"e991fa788f317d3cb4d87d7e549bd240575b44d7":{"kind":"REWORK","_number":3,"created":"2024-12-04 10:23:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/936973/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/936973/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/936973/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/936973/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/936973/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/936973/3"}}},"commit":{"parents":[{"commit":"e1588ee50ea6a58a59cac0b6f3ea6be9f15b3357","subject":"mpu: include parts bytes_used in user container bytes_used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1588ee50ea6a58a59cac0b6f3ea6be9f15b3357"}]},{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-03 18:32:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-04 10:23:25.000000000","tz":0},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: I157e524833d868bfcfaa95478ebad61b40949ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e991fa788f317d3cb4d87d7e549bd240575b44d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e991fa788f317d3cb4d87d7e549bd240575b44d7"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
