)]}'
{"id":"openstack%2Fswift~937027","triplet_id":"openstack%2Fswift~feature%2Fmpu~I2ad69baca1d392582f01092ceaf87ad5a8f7fdd9","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-wsgi-quote","hashtags":[],"change_id":"I2ad69baca1d392582f01092ceaf87ad5a8f7fdd9","subject":"mpu: wsgi_quote mpu symlink target","status":"MERGED","created":"2024-12-04 14:08:11.000000000","updated":"2024-12-04 17:03:26.000000000","submitted":"2024-12-04 17:01:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937027-p-mpu-wsgi-quote","meta_rev_id":"7179727f69a340fd5eca288cc250205687f3ba19","_number":937027,"virtual_id_number":937027,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-04 17:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-12-04 14:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-12-04 14:09:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-04 15:06:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d30f791a1be2984a39da836618a7773ec5e7c78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-04 14:08:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74ec199a03e45a4672710137526482b3e3d803b7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-04 14:09:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe1475d924650fb775b893c534f2578b72c56e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 15:06:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a36f12fac6443d684a9f37577dbdf25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c2dbbb731d24f9a9bca9f9a350cede4 : SUCCESS in 8m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0800506caa649afa7a361606dd9a170 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af045b5ff8c143b581061f1342bdabd2 : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3af8c1772a674503bed287f8c8f78dc0 : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b2d2e9bdbeb4212be51fc7636aa468d : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4e9b27df7b8d4c4b9332179ed994f96d : SUCCESS in 20m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4ee3549b1b541f69fee140459e1fa9b : SUCCESS in 18m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c885ce117c84a94bd6f11c685153c74 : SUCCESS in 20m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3788f59331634733b764e6d01e49a1ee : SUCCESS in 19m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e0c5607ddb545f69c3957ee279b512c : SUCCESS in 20m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c0084b22c0cd4586bf2ac69f38c2626b : SUCCESS in 19m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/130692480ca542f78c710dd4823b4def : SUCCESS in 13m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b0544d448d4a49bba5a0f7f73f8dda04 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d33ebd1dc1964fd6b112e44004ee55c1 : SUCCESS in 13m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9dbf53f3b95e4d15a0c86e2de1e3d2d1 : SUCCESS in 13m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/522055fe31b246eba51047fe317df9c8 : SUCCESS in 7m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/909991fd9a4c444f9d08b60f2faadbcd : SUCCESS in 14m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d44fd7bfe21402380aa6d88745a7f2f : SUCCESS in 7m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0911bf4dab90442b90e4877266b89e0e : SUCCESS in 50m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02172985c6ff442b88dda787f0248bc1 : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81abd1c58ee64d13bb06c7fc4b97f6d6 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32ae2a10260d41b78e3ba78aba1c04c3 : SUCCESS in 24m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7fd9d139db5a717602695312e3a877c798421ac8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 15:07:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbfb25b5b4e9625ea5b9308c5edb81c231ea1c2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 15:31:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b950df274aaf4006a170caf9c74f0d1c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/59389b8489af4f60ab60c3491846c145 : SUCCESS in 1h 21m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1a4612428bf0da765a7c60442e49460bdf4c7712","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 17:01:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b96bd74e64c8403da7ef63a2c890dbcb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccbed6367b1040b4a764d69b0044a36b : SUCCESS in 11m 23s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/445d45078b3b4df3b7eafc11fbc9e22d : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a358e3ef1bf4712b23c476ab051ff85 : SUCCESS in 17m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/027692e1af6a4cb4aa0c642afd8b0410 : SUCCESS in 18m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/240ad57bcbed4bffbabc06cf3e532990 : SUCCESS in 24m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9541bde525e54826993099aa8019c2a0 : SUCCESS in 21m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dbfc339fc77b44ceba0ad40a1add317c : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6598ac84eb30431a8385641372b16eb9 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2192291c8e004fc4bf8c79527e361eae : SUCCESS in 17m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d6b4d48f03714c60800084a7b159e263 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3bdc4d4fd419455a84ab0be144dadb93 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/72084f8f330d4b44b3a8032616aa520a : SUCCESS in 10m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/adc63b3fc67a41ab9ffb925865dfde28 : SUCCESS in 5m 25s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/842ff7b4531048ce8cfc210969f94170 : SUCCESS in 9m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0037ff7627dd4cad8830880b8d3b9ab4 : SUCCESS in 59m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afec4b8c3d0d40bd890be1b7286871af : SUCCESS in 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e66c8d9dd0b4e27a8901830b9a63250 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ac620db96fd4f3d95b7d2d59ebc2bd2 : SUCCESS in 17m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c6ad2e3a058c531a3a54431d32d78933a9b9c861","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 17:01:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7179727f69a340fd5eca288cc250205687f3ba19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 17:03:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/886ee3de02b94f45b3cb64ec0420552b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b9a42d321e54785b8ccff87ffe6ac48 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/51963b18afac4533ba465a3c943d2203 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"28c95803e29aefcffdb27fa463d6e42e6415bbbd","revisions":{"28c95803e29aefcffdb27fa463d6e42e6415bbbd":{"kind":"REWORK","_number":1,"created":"2024-12-04 14:08:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/937027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/937027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/937027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/937027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/937027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/937027/1"}}},"commit":{"parents":[{"commit":"e1588ee50ea6a58a59cac0b6f3ea6be9f15b3357","subject":"mpu: include parts bytes_used in user container bytes_used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1588ee50ea6a58a59cac0b6f3ea6be9f15b3357"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-03 13:10:39.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-04 14:06:08.000000000","tz":0},"subject":"mpu: wsgi_quote mpu symlink target","message":"mpu: wsgi_quote mpu symlink target\n\nChange-Id: I2ad69baca1d392582f01092ceaf87ad5a8f7fdd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28c95803e29aefcffdb27fa463d6e42e6415bbbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28c95803e29aefcffdb27fa463d6e42e6415bbbd"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
