)]}'
{"id":"openstack%2Fswift~937045","triplet_id":"openstack%2Fswift~master~If6cc8702e89f5110ad89ba933f55641de02550e9","project":"openstack/swift","branch":"master","topic":"use-constraints","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-12-19 16:49:42.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-19 09:06:37.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-19 16:42:13.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If6cc8702e89f5110ad89ba933f55641de02550e9","subject":"CI: Clean up deps for various doc builds","status":"MERGED","created":"2024-12-04 17:47:22.000000000","updated":"2024-12-19 16:51:01.000000000","submitted":"2024-12-19 16:49:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"937045-use-constraints","meta_rev_id":"259d3a236ce975ee9cbc746b9c188901fc6d76b4","_number":937045,"virtual_id_number":937045,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-19 16:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2024-12-19 09:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2024-12-19 09:06:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-04 19:01:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-04 19:11:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-04 21:19:05.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-12-05 10:51:59.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"7cce0c1019e1bb51e993e9d92f2ac4318fb00870","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-04 17:47:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3845b651418169f3b019bbba9bf4eff9921a94b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 19:01:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/885438fb854b4343ab279eb4d55f9b49\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3ddf6ab23d4b467b942a5df9fb9b82ae : NODE_FAILURE Node(set) request 300-0025849967 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"341c93f7fe48fd3fa3b691daa07b1b3e0e727c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 19:11:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46a2e4ed7be44e28808bb657ba45927f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/94885b574dc647c5abf0ab5c0ae39e35 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2ab2eb563fa45b693900dddf99eae10 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5badf3c9eff44e3994c94acbdff4102 : SUCCESS in 6m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/50dd4627e08e4acd8ec4a8381c788182 : SUCCESS in 35m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b4b467ad9e1d4176ae32d58180bc030b : SUCCESS in 48m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ac90c042ed14e7ba786bebb97a66ecd : SUCCESS in 57m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3cf22af97254102bc13c80d4bf47979 : SUCCESS in 1h 03m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6e3d9abef504a62b589b564c7e0bb37 : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6e0931d6013423093fe12ff75b2377a : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2948c82e2f634264ad43b66d9c42d8f7 : SUCCESS in 19m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/20d3af6a92284d46b099cd5c2fef6918 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d38c8eff24c42b5893b68a1a9d94941 : SUCCESS in 21m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cfb167662a9e428d8b0245930559c7af : SUCCESS in 19m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4478bffbf2ca4ebeb2600201ef866260 : SUCCESS in 19m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/021d122df65d4e92a42dceca91600fb6 : SUCCESS in 36m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/16cdc0c6fa8b4fd799eccc0b03f7ccce : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9ce8dae91d824707984fe723ea799eeb : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2961962853714c88a775f6c64030bf5b : SUCCESS in 17m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8ca14ce7100f4aa8b1d2ad321b5f08c1 : SUCCESS in 8m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d4f4c0d433eb4c6bb151dfb885eb376d : SUCCESS in 16m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5cd7ae68c8824c9380662f08393ed5b7 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7bac0379f2e44dee98737485560b535f : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/16cee48a9e3340adb3c4b602b14c4eae : SUCCESS in 16m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/33d82da3a9ee40a6a1db9bcea4039121 : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9aefd8d59f2e4a01a9e9197d87f8b290 : SUCCESS in 57m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6157fb4e101e40df89c8801ce4e8dd6e : SUCCESS in 32m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a9508fcfd15743c2b3eaf62364cd47c6 : SUCCESS in 55m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fabe5d35b6ca4d5c9187d4c0d0ebf642 : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a179c6c33c9642b1b5433138ec56d11b : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a806c78413144e68f7e02722f81b11a : SUCCESS in 18m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b64d0ae283a4f4abefe7499164eb577 : SUCCESS in 1h 01m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53cb7512cb0a42f28aada9060cb4f56a : SUCCESS in 1h 16m 29s","accounts_in_message":[],"_revision_number":1},{"id":"707f21c136df1fe921288198fd924bbfffc909ae","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-04 21:19:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e8f5d192a708d8ee6aa7c065a4661e0e0c624e6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-05 10:51:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"924a71ba4139944ec21b50672b58b34c05095374","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-07 00:44:54.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f2b5706bcb6d119b802b6d2528563a80f0c9c412","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-11 19:54:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45c29d01d546fbcc57273cdd99f80d74e02d8ccf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-12 23:04:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"552b54ab78faa44bd404335d57730e7d10f883be","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-12 23:12:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25d24e5cf70578e95dbedc22d85a5104f0f66d7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 00:18:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0f173dee53f4cadb0802a7aa22b44ce\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c10833bd74994ae9a45a63b2ee3ad5dc : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d5ef782252445c3bd7c70f75c4699e0 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bea26c39e4745399a8807393f2101fe : SUCCESS in 7m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/7b027234596248aeabf45cbb3351e68e : SUCCESS in 43m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/60f36ce6f7284d00beb779c1c36a140f : SUCCESS in 41m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1593e5bff4f42d789ddb5a7eb9e2d3b : SUCCESS in 44m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f99fe41100e4b3b906876e59f4e1733 : SUCCESS in 1h 08m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b806e1f8689448d7851f6b265abd183d : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fee9adae93f4a28be7df3e8cee67b05 : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b25817433dd74397800491af91895106 : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21f3058ad46644c6b8967b4c9f80033b : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aae6d4878c304aee959c4a23ceefc42c : SUCCESS in 19m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d17084126307454bba2593166d8ed722 : SUCCESS in 17m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/209bcad7275f47b689e69d45eeeb98fd : SUCCESS in 20m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bf34b4d93b8245308583b78c12fa3fca : SUCCESS in 26m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf7e01dd343b49f192555a9dcfb37dfc : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1363b87ef5d34d6eb6cb01e38c57d8ae : SUCCESS in 19m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ed3db9cc56c24e85a85afdeb8defac5e : SUCCESS in 19m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b1d15fe604dc4b8f83affe09e79b945c : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/384b7c59287f4174b1888ceaabca33ee : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d19be1e1a5ea494da3241a082bbaf881 : SUCCESS in 12m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9d9f5cc92d35431396ff30bb6376e0a2 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0c004fd957414aaf8ae560d9de22fb3c : SUCCESS in 13m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/79d3039500c24460a66be17946681afb : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d732ccc334c64f8bad124ac471b662b6 : SUCCESS in 57m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3715c63e7b0d484db5779a06bb1c096d : SUCCESS in 43m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6f9560f5c6414ad997134869b2d2daf3 : SUCCESS in 58m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0590932671d4b79b36f1f94ce8bebef : SUCCESS in 18m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a7a41e95bb4b58abc3982036a3fd75 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ebe8218ac5f44a8895e68e0ec42d774 : SUCCESS in 17m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fd3866f6dbc4f06b9755a6b23244407 : SUCCESS in 45m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a11cf894e28f4b429f8679cdc8de9f63 : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4993fa42d77df2ec1acba338e18762053de2c9ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 01:08:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a97f84462b324cea9862c877237e5975\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2f9fb2b3a9394bc6b047f58cd4c44483 : SUCCESS in 2h 01m 34s","accounts_in_message":[],"_revision_number":2},{"id":"50be8de7985be83b729d496eec9056f7ffefb9b2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-16 10:48:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c9396f380e1aa642577e487234bf450567437e3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-16 11:50:27.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"898fc526c21e7947c1094b3f5d9722a78b7442bb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-18 19:22:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9fb196f7d16072e1225bd6657fc306a4934331ab","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-19 09:06:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21cfc7bd574f08e1e3c422e7d075ff590d3c1306","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 09:07:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c87e268b929be64381fd299e839b5b796f93363","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 10:14:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8017633c3664757bfcbf8b3e81ca0d8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2ca37a14806f41cbaefe4662cd60a7be : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ce82285dd243b08e63a4e74ec65217 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce92a5c710234d61a8e0566a6716fb55 : SUCCESS in 4m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/838fdc2c730d4e82a69f0f957836971f : SUCCESS in 1h 02m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/26778ff77415432a8fa6905691df321e : SUCCESS in 39m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63f13de566054167a290291cbcec1820 : SUCCESS in 42m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b04e9aed9371457bb2d9e4ffa15dffd5 : SUCCESS in 1h 06m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/88c03044e2514b45b68a712584653430 : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/022ef63695744d59bc30375979301f10 : SUCCESS in 17m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bdfc32ffc7474c0f99b2522b5f7ebcac : SUCCESS in 18m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/760be27c725a4314aa428dd57e685864 : SUCCESS in 21m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/16680af8745e4bbe8d257d87dd22a43c : SUCCESS in 14m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3ab89a57fe524a67a266990cf988e12d : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d8be6027d15f47f9bb8b2f7b61d5e6c7 : SUCCESS in 11m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7e4754ca710841eabd36f067a580f58f : SUCCESS in 17m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/451dfc456a8647699a7dd6c9a955155e : SUCCESS in 9m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5d73047a0f7947e98d5c148aa7bbd951 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/491c6da263d141a3a7c602bef9bb8dc1 : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01770b106ebf4901bd38d30ba2b7019b : SUCCESS in 6m 18s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/14b9fde9d227444ca172a56bd1c2746c : SUCCESS in 7m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea63b26223954d50960db39aebd015d3 : FAILURE in 51m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db9584260b794600891c70a510078719 : SUCCESS in 49m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/448f03c680204834b93fcf7845a78bab : SUCCESS in 37m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b505dff6bd43491cbd00789c75d74db9 : SUCCESS in 18m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b0aa08f43e4d678028ee68592dcca6 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f9317adf62124ed4b05c996a32084ea2 : SUCCESS in 16m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a76b96dd6c3540e58c660e0350ecda93 : SUCCESS in 54m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/877f76a808d9493fa3b9b743acfe14fb : SUCCESS in 48m 21s","accounts_in_message":[],"_revision_number":2},{"id":"3b53e0e6bf8272f7de7e4caeeec8bb837886be8c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-19 14:31:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c55a7d563ff28642c86e07452a1603354af2c264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 15:37:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841533259a344d3f9aea45264f4ade7a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/84035197c06946c0baa36d3a8c5d8757 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec5f99ac77147a183ea825d7b7be1cf : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8bad8b9ef0d45528ed6285004b11be3 : SUCCESS in 6m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/06c98e2c55974f92a007852bd4b09025 : SUCCESS in 59m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3c4a15c5af9d4f3aa73c169d2cc06d1b : SUCCESS in 58m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/742bd075d77d4340acc77ed7da80045b : SUCCESS in 49m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5dd025ed09b3436b9cc8bc716539f6c7 : SUCCESS in 1h 03m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3cc8e69669d04ffd975816b52fd781d0 : SUCCESS in 24m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73bb6f444515453092befe721b8c3080 : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8371901c24954f5ba317fec62865ac22 : SUCCESS in 25m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c11e2d53b5574bb4910191a883df3347 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/22584c466eea459c97bec8e2a1376179 : SUCCESS in 24m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/726ee776ffb9402999070276b2c5e162 : SUCCESS in 20m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bde8a6791bd74ed795c312250b82f5f3 : SUCCESS in 23m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d04db597b79640dcb91625745c71c9d3 : SUCCESS in 27m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c002e32911944453833c2b4be86c3593 : SUCCESS in 10m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9c2cd91bd7594ee5ae0fc90e46bb7086 : SUCCESS in 25m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ec4f02cb653d47c8ba9cde86c7615d87 : SUCCESS in 23m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/744e2af6df42447b8064e26c54c4f35e : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/69f2c213e0964ac88048274647077fdc : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5fd581f10bb74f1c93ef1691c328cdbf : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/27e8aebdfe164c1199579ea14389be20 : SUCCESS in 7m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b8940053f2a444fea007050d0ef453c7 : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a48a519dcd11490a8b236262929b9cbc : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c45818e7a4f42ec9ae8b3d722746e9e : SUCCESS in 59m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d47f2797610e46419fb31841ce7eb8ef : SUCCESS in 34m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1e61c672aed4401ae7f05badc075b0d : SUCCESS in 48m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64eea729a3654d9aa657ae0e33afe589 : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/364596dfb4a346dc9207afc28a03604a : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/842fd2beefed4dd69190a889ea6f7618 : SUCCESS in 24m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e460816662a24ee08a076aee0c521478 : SUCCESS in 46m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c953695768f04f75ac93e4ace9ee7f42 : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":2},{"id":"0fbf44f953c78925abaa2d306ff976254fa16808","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 15:37:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36eb9561b5800f727b6582af42c0b1235da9bbef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 15:45:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c6a98ae37374c9ebc16a90d52e741c1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64cbabbf354c44b89695c767ea000962 : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":2},{"id":"6fca9795ef7b4a3c0bef6ad1383135591beea2f3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-19 16:42:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9d64d83bd94e15ca344f9099022043248df9fae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 16:49:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77260fb021bb455c8e6f57f3557b0da6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5bae70c712534cecab4700d3227e1079 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b089cf128c864c7c9f9275a9f8076aeb : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01a80402b60e40c5956c7732a0f69807 : SUCCESS in 6m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/f0ca93e369a04b7eaa5479670de9cb54 : SUCCESS in 1h 05m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c0873d71fd7942cd8541c3e263af0e97 : SUCCESS in 1h 03m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9c7ce926d580412db595c1db9fe74069 : SUCCESS in 52m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f1b9b3f83fb4314ba5a38b17b0d26d9 : SUCCESS in 1h 07m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9b18b47e93df4c32b8a1f854e0e2493e : POST_FAILURE in 6m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6bab41c7d3044fbdb820c49ca4b4d068 : SUCCESS in 19m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9c586ca9ef34ab3a95f1635ba3136d4 : SUCCESS in 18m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/523ec9489ef844dd8d981e5312cee13f : SUCCESS in 23m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7d1df607c86a452eb64604d648d39ebb : SUCCESS in 28m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8f1a0cb12cb245af8b73d8e447a91c11 : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/63ab1cce025543d0a5ce2cedfa4bda85 : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/464fd5fd060241638aad0a93229f4d29 : SUCCESS in 12m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/14ed45551ef64ec49b04ea676255f1af : SUCCESS in 9m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8785c3c5b5dc4dcf9b755c7c6ea9b4b2 : SUCCESS in 10m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/413cae9572e24e8b8db9cb4f6beb1ed2 : SUCCESS in 13m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/85f48701dcfe45529df629991871c617 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6042c6d561794565bea2b3e0e876e115 : SUCCESS in 5m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e51489dda3d44d0fae685fd8eed14467 : SUCCESS in 48m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/921a94bfeea04fd5b37c9e28cd22856f : SUCCESS in 28m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c5caee1fde064b4ab91c9b9a486b1b77 : SUCCESS in 47m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f64a7d445b834f279bfba74e366fcdf8 : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc649eebcaae48b0bede9d46fed6feb1 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/110d89ee6f2c4ec39e8e4fa8bc17f8fd : SUCCESS in 17m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/924adf98a0d74136aba32e731b15707a : SUCCESS in 37m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca0931c96a3643868ff4b43ab60056b6 : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":2},{"id":"62aabab4f99a318aa7295df517e680f0b96390ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 16:49:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"259d3a236ce975ee9cbc746b9c188901fc6d76b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 16:51:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3e51e395e1748e3940651efb973b9f8\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0aa41eab220e4ab399f993c0996d464a : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c028f272ee04bccb675a6de61b46049 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41841cb2ebcb41ea9ea0cb5a8eb445dd : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f7efaa96178e466698fe8a7bf1c7b190 : SUCCESS in 58s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"13197af6e30518da0f796a95b4552844feaa8edb","revisions":{"d17a67b41083ccc9076471a0df8841c70f5f894b":{"kind":"REWORK","_number":1,"created":"2024-12-04 17:47:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/45/937045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/937045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/937045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/937045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/937045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/937045/1"}}},"commit":{"parents":[{"commit":"54ec8a124a37d170166f10cb13fae48b5c29a0a0","subject":"CI: drop pip --upgrade flag in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54ec8a124a37d170166f10cb13fae48b5c29a0a0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-04 17:45:20.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-04 17:45:20.000000000","tz":-480},"subject":"CI: Use constraints for api-ref builds","message":"CI: Use constraints for api-ref builds\n\nChange-Id: If6cc8702e89f5110ad89ba933f55641de02550e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d17a67b41083ccc9076471a0df8841c70f5f894b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d17a67b41083ccc9076471a0df8841c70f5f894b"}]},"branch":"refs/heads/master"},"13197af6e30518da0f796a95b4552844feaa8edb":{"kind":"REWORK","_number":2,"created":"2024-12-12 23:04:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/45/937045/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/937045/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/937045/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/937045/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/937045/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/937045/2"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-04 17:45:20.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-12 23:02:27.000000000","tz":-480},"subject":"CI: Clean up deps for various doc builds","message":"CI: Clean up deps for various doc builds\n\n- Define a single location for doc-build deps\n  - As a side-effect, use constraints for api-ref builds\n- Remove test-requirements.txt from those deps\n\nChange-Id: If6cc8702e89f5110ad89ba933f55641de02550e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13197af6e30518da0f796a95b4552844feaa8edb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13197af6e30518da0f796a95b4552844feaa8edb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
