)]}'
{"id":"openstack%2Fswift~937072","triplet_id":"openstack%2Fswift~master~I1cbd26e6a18e01abcfb2f0a28649cfcaa6358252","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-18 21:26:13.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-06 23:30:04.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I1cbd26e6a18e01abcfb2f0a28649cfcaa6358252","subject":"Bring py3-constraints.txt more in line with global u-c","status":"MERGED","created":"2024-12-04 22:20:04.000000000","updated":"2024-12-18 21:27:44.000000000","submitted":"2024-12-18 21:26:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937072","meta_rev_id":"ce83d3ca25e386bfce071b3c51dcfd80e7bc4a08","_number":937072,"virtual_id_number":937072,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-18 21:26:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-18 19:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-18 19:16:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-04 23:30:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-05 11:04:29.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"f5bcabcf08df903b3cc974bcc204d78e99dd321d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-04 22:20:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bab9c19086f8f823a17bd4379b04794307467b99","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-04 22:35:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"632965220774b550e7020ec83dcdb511848b3e30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 23:30:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe026a404507480e878a6d63fdf3d66e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a5bb963116f47debf5d2ad8b4c37e57 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/fccacd306d544c7997922369794bac70 : SUCCESS in 1h 04m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/450a589188b5476fa0fbc09abbe3849f : SUCCESS in 45m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c3e7a2f720e2412c9f035d64f689d516 : SUCCESS in 1h 05m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8a6a9c7588743c7b53676ab02500661 : SUCCESS in 52m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df89f71ce7164ff2928f14507e771349 : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd93538c38a14af6b36d687beb794e06 : SUCCESS in 21m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc164b6714834477b09a9273840a448b : SUCCESS in 24m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb812cad8b34436790111a8ccfa5aab4 : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/337954603c9c48d89c7ae4980fe29f97 : SUCCESS in 19m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a43c640945c4ceb8dc4b28855d16e53 : FAILURE in 3m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/139b870f6db44c889d300e9f276ba423 : SUCCESS in 21m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c39aa652ddb14b2db74de722e7157f8e : SUCCESS in 20m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bd2d1deaef6b4c59bf54045b947529ad : SUCCESS in 20m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d76d6705091743738bcbf8a34d243b44 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b2da10981a149dca5185e961e4af27a : SUCCESS in 17m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bfd2e5158f354f908ea6653fc47a89e3 : SUCCESS in 8m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/85ee40bc308242faba0e54c1c46911e4 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/14307900e92c4899ba341a78657daf1f : SUCCESS in 14m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58f53c0f76274f70893e8c3bdee15839 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7daa5bdfb09b453ab1c7d5d6dd9f0ed5 : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f6d33a4036454ce79794fdc3258f7b5d : SUCCESS in 8m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfc129a1cb5a40a9954a1d1e72c15cb9 : SUCCESS in 55m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95f3d789d5d64ffdac64887e8cc6ade0 : SUCCESS in 34m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e519e6e5dfb945d6b816029a50543faa : SUCCESS in 25m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a483802bc644266a848e9ab9c299d5c : SUCCESS in 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc4cafe0dc864a1e829dfb462d04caea : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/939d95c0edbb465eb2ec78452a7fad8e : SUCCESS in 18m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c3e6d3ad282497a8ff4bab2647ca4e4 : SUCCESS in 34m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76fe3092bc004e64a4c7dac352844bcc : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b0ec052b5237f150a8ec528d2d42e5c9f59b2593","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 00:05:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f5160f317664c3f88ce5235fa604929\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f748220ef46409480e6a807486b7da3 : SUCCESS in 1h 44m 00s","accounts_in_message":[],"_revision_number":1},{"id":"674a47563e3a50fcc3a7411a67376c242360c81c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 00:18:57.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa3dd847345b4369a3ad7e2e5dcb61b9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32922635ebce4451b8c8e1a8550b8b2a : SUCCESS in 1h 22m 52s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/62ae32f61a3345e7bb973e20fb33d93f : SUCCESS in 1h 22m 38s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/50d695dea0374c90b006e2f5bc7c3da3 : SUCCESS in 1h 41m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/163ce18f4c7342a6971e78512e90a452 : FAILURE in 4m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6d1b87830961443296bb4d0309bf2cf8 : SUCCESS in 10m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/41bcf6f935b942e98c25ac2bbb23a682 : SUCCESS in 23m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/682ae9218bc640e6912f6f2ef8a0a29e : SUCCESS in 22m 02s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3827af987784ef5b46f03791c6eab0d : SUCCESS in 19m 26s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8956dbba32244b56b0f70df866f4932a : SUCCESS in 10m 49s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b087d03a406e4b00b30e59f2ca432347 : SUCCESS in 19m 15s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/b07b74945f8441068132bce0474a5d2a : SUCCESS in 18m 28s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/84578ae170b14fdfb2f2544c7e23955a : SUCCESS in 22m 25s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/2264e7d5257d493891623c0d37a34294 : SUCCESS in 19m 42s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/e13248263fb74362bcf84a6f5eb0a00a : SUCCESS in 19m 00s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/e942a1335b09417ab98645e6f94a59f0 : RETRY_LIMIT in 5m 45s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/ea197901ef0748d091d02add6d6b3e45 : SUCCESS in 10m 08s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/a69ca14be06e4ef796e868f4f6d0e9eb : SUCCESS in 18m 07s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/53319723db2a4969aa49d68681e5f53b : SUCCESS in 18m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8319cc29f657f627a970821d22532aa2621317a8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-05 11:04:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"550b2f1f7d2ceea1b087add2a40de6890a6dfe95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-06 22:59:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"77f4e1bb29ceb4464c1819248ffa9287afee9ef0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-06 23:30:04.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eb785119a2429f7f38d56a3e14824cfaee5f3ce5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-06 23:49:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3eca1ece2d48df90937036fe8a30ffd0b6122d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 00:17:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c04abb1c19874b9c86323c7890541f5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d24cdfd6dea42d9948f6db8fcc39dcf : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/3ea28b218585467aa02dfd24a40d08b5 : SUCCESS in 52m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6400c3156db4cdb9534b8e72bc28cc5 : SUCCESS in 59m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65184a71f299477b806532f57e1038d9 : SUCCESS in 49m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a16582bb0cde4779add328e13f4293c0 : SUCCESS in 1h 16m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43fd7c0e17c34c6aa3a789924e2bef07 : SUCCESS in 11m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b628b28a7b94ac38f012aa5cf4adfa9 : SUCCESS in 25m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/acc00e40f5bc49d69ad187b3e708c10f : SUCCESS in 22m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b58e8fb78b749f79ba49cdeeaa6c0b6 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a25693c9e5564f18917e2720d873d4cf : SUCCESS in 10m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/44cb5d9ac4d8415c977be79bcd2eb293 : SUCCESS in 19m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/38bce3ce8e8444bc8b0f746acaf13ba6 : SUCCESS in 16m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/00939376583c45cfbdc9eb4e830c0a53 : SUCCESS in 23m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e85d62e42f045e09569eae0af4c775a : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e264b7f2f4634e439c216a54874ece3c : SUCCESS in 20m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b9a182e4053c489da3740283712930b2 : SUCCESS in 9m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0de7399c37bf42218db65156d6a5a16e : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0df94078f1f74860869617bdfe98ee01 : SUCCESS in 8m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7880ae1dbb8b4157bc312c49a7057501 : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7188fe2cae71469383cd59ded00e6079 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c94d4a0ef1dd431b842443ddd650a71f : SUCCESS in 15m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2948846e0cd3447ab13f31b78b477acc : SUCCESS in 6m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/22b1833fec1f41d194373903bd320794 : SUCCESS in 54m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00a331926b9240f4a5c5bf8e6a0d240b : SUCCESS in 34m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c8e18be77714442480c9d5a62078bfc5 : SUCCESS in 42m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aed965265e984eb49ee99b7c62776ea4 : SUCCESS in 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae28be2cdcbb48b28c73520e9f6d0eb1 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2576ccb187e74b02ada68891cd24bd15 : SUCCESS in 19m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e181aaef5b047e99426c208bf683eaf : SUCCESS in 53m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fcbbc928dfb431f999daf72131c5b9e : SUCCESS in 33m 39s","accounts_in_message":[],"_revision_number":2},{"id":"0b3697800fb1294420c6e4a8c11c38cbba55d432","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-07 00:47:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13057f394edad822c0517858cb85b9e951702198","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 01:00:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb8aa27c29274fd7949aabaac5aa6fb9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a87ec2c9a87e4bf8999d3cc98bbabdb1 : SUCCESS in 1h 53m 12s","accounts_in_message":[],"_revision_number":2},{"id":"8d05f922fd2c1ff95a93720b3f260c589c414a50","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 01:09:51.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb4568f78c040a5b387cb93d28945d3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4dfb6cbae9b7454d8d1b7638827d7c33 : SUCCESS in 1h 18m 05s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/343cacbd8bce416c8f68eee7cc09247b : FAILURE in 1h 11m 10s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/099c24c44d884de3b1f2d7cff4ba94c2 : SUCCESS in 1h 15m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/877f07b3a2a948bf876c67be8db906d5 : SUCCESS in 22m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c408dfabf45648938c23748909062de4 : SUCCESS in 27m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7a876d55b7194776aad580f7981dfb0b : SUCCESS in 27m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e748839985204294a5b2265f08de7100 : SUCCESS in 9m 14s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9c1d3eb00e84c42b2cd7383080b158a : SUCCESS in 21m 59s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d704a72bbf1544b19d6ac3336fb164c5 : SUCCESS in 25m 56s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/cad392b22f8c45b8b603da367ec50513 : SUCCESS in 18m 14s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/234c520c4ae544d69ecf1eb37c6de4df : SUCCESS in 18m 22s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/b678a8874a6d4ea18d061c44c8e79db0 : SUCCESS in 18m 15s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/5ba40019184c47d8acc38decf90413b0 : SUCCESS in 10m 41s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/0d64cb63c9ad40adad91826c5342c6d4 : SUCCESS in 17m 08s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/42b77ea7a2904db3865b0f5133edae74 : RETRY_LIMIT in 5m 38s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/19421e25ee4a492caef8bae7e6d09e10 : SUCCESS in 17m 24s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/5cb0a4004553435da241f3d4670d55f2 : SUCCESS in 16m 24s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/1491870672c741c4adcc5363b6264102 : SUCCESS in 17m 56s","accounts_in_message":[],"_revision_number":2},{"id":"3300d28c174f384e99b4bd9c70e8f051a6f46ce4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-18 19:16:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"271c58dbc920e1e6d0d0d1801c5fece3f6ae48ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 19:17:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e9e8cc56c769372cc7167a2c672790b614565f99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 21:26:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16362deb3d8e4fe0804714f442f0fa97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7aba6910d9b4948b77463a9308608a4 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/0f028a22eca44aae9657f1f6cb02234c : SUCCESS in 40m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2e720bd0ed814429976f7364ed824f94 : SUCCESS in 49m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b732cfcb7d4844bfb59f08412b352cfc : SUCCESS in 38m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3de2fb35af514ee3a271f6ccfc9ce3dd : SUCCESS in 40m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/23b1ab2a865340ff924d088a06f95e38 : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a1c9911cfa448caba0f06da22feeff2 : SUCCESS in 18m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1261914aee294f40beafc7ef4b8fefd8 : SUCCESS in 19m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/62144c4a30c141798a7b3aa95a986f46 : SUCCESS in 16m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c44906459e1146a5a9661cb078ad4a4b : SUCCESS in 13m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/675cf8d6af8b45dd88b4d761c5ba59da : SUCCESS in 18m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/82a1d0a2388d4785b3977aedb3e3e598 : SUCCESS in 19m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/14fcb970332643b9b312195ca381b451 : SUCCESS in 19m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/373443a6f5884036a07e1863ced8aad7 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9353e30c1a984556a61035a5a64eab71 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a5e44cbb7cbe45e2b64b5147f27b4146 : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f8c4f938e37f4f329027ef8ed67325b1 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/db3485fa7ac9427496e1de83dc684b46 : SUCCESS in 9m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd048119dc244159bf1e996efafb2d33 : SUCCESS in 54m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2b5326c832d943c1a05b73dbc4938010 : SUCCESS in 53m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3eee0a16a4c14ccbae13c429ca0bb491 : SUCCESS in 34m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92f53b01579040ee86c9895df2619b2a : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9472dd7686924e18940fd7060ce76161 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60897526d561449884a52b82e28f5fb0 : SUCCESS in 18m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0080eac2747041bd964d54d4248a7635 : SUCCESS in 39m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e5b814104de418dadfa5fe9b77f3f9e : SUCCESS in 42m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c8004a423ca77a8ba00ff1f3e0f95900490ea1a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 21:26:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ce83d3ca25e386bfce071b3c51dcfd80e7bc4a08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 21:27:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dce2476050c248b99aa457e540587dea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/664a2e06f537436b8a21d17ccc371c7b : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/428928b93ccd4b8bbbd75a3daf4f8308 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0513b8ceaaff591fb6abcbb66d4f5490d0c15f96","revisions":{"0938a90152ab2a6d90efee673ab3231ae6c33bae":{"kind":"REWORK","_number":1,"created":"2024-12-04 22:20:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/72/937072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/937072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/937072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/937072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/937072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/937072/1"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-04 19:03:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-04 22:19:49.000000000","tz":-480},"subject":"Bring py3-constraints.txt more in line with global u-c","message":"Bring py3-constraints.txt more in line with global u-c\n\nPurely version bumps, and dropping the swift pin (which was added\nerroneously anyway).\n\nChange-Id: I1cbd26e6a18e01abcfb2f0a28649cfcaa6358252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0938a90152ab2a6d90efee673ab3231ae6c33bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0938a90152ab2a6d90efee673ab3231ae6c33bae"}]},"branch":"refs/heads/master"},"0513b8ceaaff591fb6abcbb66d4f5490d0c15f96":{"kind":"REWORK","_number":2,"created":"2024-12-06 22:59:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/72/937072/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/937072/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/937072/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/937072/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/937072/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/937072/2"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-04 19:03:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-06 22:59:50.000000000","tz":-480},"subject":"Bring py3-constraints.txt more in line with global u-c","message":"Bring py3-constraints.txt more in line with global u-c\n\nPurely version bumps, and dropping the swift pin (which was added\nerroneously anyway).\n\nChange-Id: I1cbd26e6a18e01abcfb2f0a28649cfcaa6358252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0513b8ceaaff591fb6abcbb66d4f5490d0c15f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0513b8ceaaff591fb6abcbb66d4f5490d0c15f96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
