)]}'
{"id":"openstack%2Fswift~937075","triplet_id":"openstack%2Fswift~master~I0fe476c56c120f8a6dab6afa2cf03e8278ceefc5","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I0fe476c56c120f8a6dab6afa2cf03e8278ceefc5","subject":"Up-rev hacking","status":"MERGED","created":"2024-12-04 22:26:31.000000000","updated":"2024-12-18 02:21:02.000000000","submitted":"2024-12-18 02:19:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937075","meta_rev_id":"d14b3d7fffb9c00edf79d39871a503ce0e5ea85d","_number":937075,"virtual_id_number":937075,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-18 02:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-18 01:13:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-18 01:13:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-04 23:48:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13bc08c882d42c2116478f7f9d74324b2e1bd835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-04 22:26:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1cc990cf76da9ca3445be4e16a663a67b54cfff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-04 22:34:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5f5d2ae0b0cbf10c1a0a24877a5aee91a8cf536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 23:48:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff7290580e26404d84a64053dab2e776\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/76f6f3f9e2964d628a0c888e86b0494c : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f14b9c96c9e4b18bfa0c5fe029ac968 : SUCCESS in 8m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/96bfe1ef56714c0cb81dfa23bfe9c839 : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5cd705710594330b796da9729bb5be4 : SUCCESS in 4m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/5070d7c093c2497e89c86bc6f6d8d217 : SUCCESS in 35m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3bf97405c9a548409e3e3f21259ed1cd : SUCCESS in 35m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3e6e621bb24e441ebc6b5afbe5e43fad : SUCCESS in 35m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5f6ab581a6c48a99a92b2e75c760657 : SUCCESS in 1h 06m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e23642be5d2d4ab9aab13de3234fd258 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9c810158aec4d4e89cefd001f2c0f6e : SUCCESS in 12m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94f9764f96934bbca27a553ba0e82672 : SUCCESS in 19m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7103f8f5ced4480cbef40f1e6829d47c : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/862661f298a74a1997d50f8e57f8fe0a : SUCCESS in 9m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/42b4b7bdb07b48ba8ff9bc1ea07d0c87 : SUCCESS in 17m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c4208356f0864afe853c31766b6a60d4 : SUCCESS in 19m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0b83db6de37a4c5fbb6519c713fc5fdd : SUCCESS in 25m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c2f4cdf83b64bb78555889fbdca827e : SUCCESS in 10m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a4cfd22b99004bf19a6e5802dc71685f : SUCCESS in 16m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/157dc8d86ada4d33be51bf8d8ca9bc7f : SUCCESS in 17m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f187928302bd444f810b0c2e19b3433b : SUCCESS in 9m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/841cc996ca674aa195d348b13e1b3fc7 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bfa875acc325438aae8392b728463df4 : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/de78ea45494748b8899a995d7a309a4d : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ac1fe4787c44c2e87adbea621632b03 : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7322b07611f744dc82510a55ef80ce34 : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3b080c04ad34ba2ab841a1854a12c30 : SUCCESS in 56m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f41ccc56b4264f008eb154c7213d83c0 : SUCCESS in 49m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9d0524dd8a2e4970a103af09ac069d37 : SUCCESS in 39m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba331f20932d40e5bb72eaee3ab9889c : SUCCESS in 23m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a09366b2f2248c89f928fd9e5bd58ee : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/991694601ba34f27a118941718b1c123 : SUCCESS in 17m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2a4576a53a44ebe982a929829ce7a52 : SUCCESS in 51m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f89b069f839446aa0eb1f386c0b54fd : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":2},{"id":"ce21eaece4299726ad29e529f92ad9a3867262c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 00:26:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f023d3a620841fcaca7808024a02f55\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f91221d25b3c45d284fd1b5e7b13e999 : SUCCESS in 1h 50m 10s","accounts_in_message":[],"_revision_number":2},{"id":"16992319a8434048cbffe64f9853d416de08a3c6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-18 01:13:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e7633c4a5cfee881b053f12c36fc1ee31223f72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 01:14:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ddc76d8469efd8fd76dee33cbfe5b48fc2331059","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 02:19:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/670ab93d5848473ab1418c57f5424314\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a67642acbf8445369e21f5193c0233df : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbfe28bae67f47be96e7e2757db05104 : SUCCESS in 8m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/30cca5a420534928bd249862eca9eabc : SUCCESS in 2m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf7acce91d8140689f3bd80dc560709c : SUCCESS in 5m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/906d7eba3c8c4718b826a9941d5e704a : SUCCESS in 1h 03m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ecd34c2964cc437ea96e979b0053afd9 : SUCCESS in 48m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8dc4e959c604e85a9a13413c7c2f331 : SUCCESS in 36m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c3871b90b714833931296c2b5b2ec86 : SUCCESS in 38m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3f1bc54d5b54422b8658a5ea375bff9b : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42681f3468844a13822b772222a1e958 : SUCCESS in 9m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c15581cee42142c8b826bf191ca09889 : SUCCESS in 14m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2d36b74e34e457db72e4a92838d4e1d : SUCCESS in 13m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3346e6619a6245d4b02631b68e995f4d : SUCCESS in 33m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/77e8de48d60849fb8d41d9749e339e8e : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a6a3f21d96b54dae8c7f1956ec7df6b8 : SUCCESS in 16m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da49f414836e4f01847be4f83d4890d3 : SUCCESS in 17m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/da862a8d7f6c4ed6a044f062d01a394b : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fc92dc02729a42ed91cf5fea7b217312 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9e1e8be88fb74e30a080c29e1f878028 : SUCCESS in 9m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a8256109e32c4788ba11fe5e837cc39d : SUCCESS in 7m 58s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/99a04286018a49bf9a3f9a2e0ba540a7 : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/485f21992dca4dddb85469ca84e9af0c : SUCCESS in 58m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b7bdd7c4108a450ca01fe7f5f4d939c8 : SUCCESS in 31m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0684f4d497d3410dab01803e9ac0d01c : SUCCESS in 50m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3852323ef894a14b65280682e6e5f2c : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5a4f68a15ae44a8820639344c5a20a5 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aee8acb5d2d94aadb9a812c61f72db19 : SUCCESS in 20m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b0d93a2bc7646008ee65de9241088ea : SUCCESS in 40m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b81facdb831442194155ddffaeacfe8 : SUCCESS in 48m 47s","accounts_in_message":[],"_revision_number":2},{"id":"719eff3d07f07cff7aedffee2edd147b510fffce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 02:19:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d14b3d7fffb9c00edf79d39871a503ce0e5ea85d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 02:21:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ffee15279274700813f016356894265\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6151e41a33d347e38a6e73a4af1df93c : SUCCESS in 50s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1e768c30d384906a6f01a4167cc7e91 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17c59ecc7c9444069ff71eef47c6aca3 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0cf6cdbc2b714f90884010340a640195 : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"791d38962bfbe98053ef475db875c654a637cb97","revisions":{"f15a13e3ba1610584977cc29009021544369cf11":{"kind":"REWORK","_number":1,"created":"2024-12-04 22:26:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/937075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/937075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/937075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/937075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/937075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/937075/1"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-04 22:25:54.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-04 22:25:54.000000000","tz":-480},"subject":"Up-rev hacking","message":"Up-rev hacking\n\nChange-Id: I0fe476c56c120f8a6dab6afa2cf03e8278ceefc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f15a13e3ba1610584977cc29009021544369cf11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f15a13e3ba1610584977cc29009021544369cf11"}]},"branch":"refs/heads/master"},"791d38962bfbe98053ef475db875c654a637cb97":{"kind":"REWORK","_number":2,"created":"2024-12-04 22:34:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/937075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/937075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/937075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/937075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/937075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/937075/2"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-04 22:25:54.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-04 22:33:50.000000000","tz":-480},"subject":"Up-rev hacking","message":"Up-rev hacking\n\nChange-Id: I0fe476c56c120f8a6dab6afa2cf03e8278ceefc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/791d38962bfbe98053ef475db875c654a637cb97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/791d38962bfbe98053ef475db875c654a637cb97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
