)]}'
{"id":"openstack%2Fswift~937082","triplet_id":"openstack%2Fswift~master~Ie6db0eb9ffc6e271f53ec25d88d791988ad60bce","project":"openstack/swift","branch":"master","topic":"heartbeat-resp-on-COPY","attention_set":{},"removed_from_attention_set":{"27623":{"account":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"last_update":"2025-01-31 03:13:58.000000000","reason":"Change was abandoned"},"37337":{"account":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"last_update":"2025-01-31 03:13:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie6db0eb9ffc6e271f53ec25d88d791988ad60bce","subject":"add heartbeat responses on COPY requests","status":"ABANDONED","created":"2024-12-05 04:22:24.000000000","updated":"2025-01-31 03:13:58.000000000","total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"52411fbc9b7efdc9d8d2784b81c98d5cdf0c977a","_number":937082,"virtual_id_number":937082,"owner":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-05 05:38:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-12-09 20:36:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-05 04:26:40.000000000","updated_by":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REVIEWER"},{"updated":"2024-12-05 05:38:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-09 20:36:34.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"425ec47a45119fd0c64bb0156d6077e8f094dfab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2024-12-05 04:22:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"242a681d9fd7d6048d8af86ab64169ea3825ec04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 05:38:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d174c0c5eb446f69c1641cd6ec54617\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5ac42f07a8a4d898cb27514b40e513f : SUCCESS in 7m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/11915d5703cd4091aa4f54ddfa17a52f : SUCCESS in 55m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5a03e6f92b82407585875b6f0908cf78 : SUCCESS in 1h 10m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/473ad992aac84959b372dac654fb60ba : SUCCESS in 57m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6ed01c950bf41299dcf27945c34377e : SUCCESS in 45m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8078e2bbb95b41cd9feb2058fee06f4e : SUCCESS in 21m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec2c3fa121384332a0866a47684574e6 : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec267a2b53b74897bb5776502c4cc6c1 : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/88267cadc1f84cba9b4aff74bd68ee63 : SUCCESS in 4m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/82209d9c487542f498e1ced988ca7674 : SUCCESS in 9m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8fc3944b4b584f13a9c93db77cff7fc2 : SUCCESS in 17m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d2fd69d0e2634ebb83b516d009d15fe1 : SUCCESS in 9m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fbb3e9511b7944a895246fe228688afe : SUCCESS in 27m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fe144358617d4736a052919c799ce03b : SUCCESS in 9m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e1de35267d364781bebcd14c21e6559f : SUCCESS in 17m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/99f392cc4b5d43c7b5f51a3698fddfec : SUCCESS in 9m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d889795a9891433891ad492ef643c6f6 : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/647215bbc79043a4bea65fcdae5af73c : SUCCESS in 8m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/786607d9b8a04ca6acbfd4d8db515da1 : SUCCESS in 16m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2cafda4b666f416186915d2da57625cd : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef06bc410abf490198f5e34a338a0b25 : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4d42aa8b59014352b5728a50e37a6da3 : SUCCESS in 9m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1c2143f806a493d9e88aa9ff2fa6b5f : SUCCESS in 48m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08001886ca0b47e49377ca5eae00473f : SUCCESS in 1h 03m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cca848666130409d9a2f271c7a0b20d3 : SUCCESS in 48m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a29418c21584c77ab41f164035acfe3 : SUCCESS in 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd73346f9414fb4b75fd05da98380a2 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f05b72312ba24c369359ef894fa8e359 : SUCCESS in 18m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/496c1dc125a243debcd103562440c4b4 : SUCCESS in 55m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1b04531b19e415397254f10d09498cf : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":1},{"id":"685af98b0d59e8c0e8d07178b0bba57bb6cc7469","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 06:00:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5a5b940d82d45288c37e8c9319ff697\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/06183553af82423898d05a3eda73ed51 : SUCCESS in 1h 35m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b1444561aa71bacd7a168a4961b0e2dde5bbe593","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-09 20:36:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"52411fbc9b7efdc9d8d2784b81c98d5cdf0c977a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-01-31 03:13:58.000000000","message":"Abandoned\n\nI had to reset the entire ubuntu environment and for some reason I cannot fetch this changeID patch. I\u0027ll create a new one for review once I finished up the code","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dff00b4b4afdc6fe80cf97e2360ead66897aeaf8","revisions":{"dff00b4b4afdc6fe80cf97e2360ead66897aeaf8":{"kind":"REWORK","_number":1,"created":"2024-12-05 04:22:24.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/82/937082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/937082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/937082/1"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"nathang15","email":"nguyennathan1502@gmail.com","date":"2024-12-05 04:22:09.000000000","tz":-300},"committer":{"name":"nathang15","email":"nguyennathan1502@gmail.com","date":"2024-12-05 04:22:09.000000000","tz":-300},"subject":"add heartbeat responses on COPY requests","message":"add heartbeat responses on COPY requests\n\nChange-Id: Ie6db0eb9ffc6e271f53ec25d88d791988ad60bce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dff00b4b4afdc6fe80cf97e2360ead66897aeaf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dff00b4b4afdc6fe80cf97e2360ead66897aeaf8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
