)]}'
{"id":"openstack%2Fswift~937189","triplet_id":"openstack%2Fswift~master~I3a677bb67c5700da48f89c847f652b4610ba47c2","project":"openstack/swift","branch":"master","topic":"sqStatsdClientSource-patch2","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-01-07 23:46:14.000000000","reason":"Change was submitted"},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2025-01-07 23:46:14.000000000","reason":"Change was submitted"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2025-01-07 23:46:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a677bb67c5700da48f89c847f652b4610ba47c2","subject":"tests: relocate some logging related unit tests","status":"MERGED","created":"2024-12-05 15:49:42.000000000","updated":"2025-01-08 09:37:01.000000000","submitted":"2025-01-07 23:46:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":59,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"937189-sqStatsdClientSource-patch2","meta_rev_id":"77e1e571e904320b6b3732f5c3ecd6cdfda8e27b","_number":937189,"virtual_id_number":937189,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-07 23:46:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2025-01-08 03:20:26.000000000","post_submit":true,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2025-01-07 22:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2025-01-07 22:41:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"CC":[{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-05 16:54:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-12 22:26:58.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"CC"},{"updated":"2024-12-16 16:22:58.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"REVIEWER"},{"updated":"2025-01-06 17:04:24.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"CC"},{"updated":"2025-01-07 22:41:24.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-01-08 03:20:26.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"dfe7eca17060a83a4f16079d6dffbe3744b2a7b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-05 15:49:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e960a7ec2d67f62ae2d00296be54e199d7f755ae","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-05 15:52:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f022bf53dd9245ba54b288db56fd7f5edc36ee6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-05 15:57:02.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f07c21b9014c162520f17239c834b885273094fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 16:54:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d964487609848a59658905c32c1777c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3736e114a3594632992207a861bdef40 : SUCCESS in 9m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2b203b75c0b44410a6914d5825549e20 : SUCCESS in 1h 01m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d897e8145cb4e3da94d7664cd420a78 : RETRY_LIMIT in 3m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c064456d18eb4d8d87debe01af5c2c10 : RETRY_LIMIT in 4m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e37db1b8a80b4787a432a471feb435d6 : RETRY_LIMIT in 3m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0085cb493c33429fbe59b2b415eb79ef : SUCCESS in 10m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/475c7aa53b724cdea524f87d8eade3cd : SUCCESS in 20m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/775858ad8f834f32952a720ded727808 : SUCCESS in 19m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a4dcfb0d5a9e4119bcd8fbe0f79105ac : SUCCESS in 9m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/00d11817f0ff42088006b82901fc26f1 : SUCCESS in 32m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/27bd0b1bea0544b28c513f0d7a25a0d9 : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/27aec62ba7554b70887ffe7b39db18cf : SUCCESS in 18m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9cf44077bc71402cb6bc686102d7ba8f : SUCCESS in 20m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d426830c096d434a99ccfb35f6f0078e : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5c3d0aee372c4f458247e11f9a9e4e1f : SUCCESS in 16m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fe9e1209e3bc4582923766b97945d1f5 : SUCCESS in 13m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/be3c48dd99d2449f97e14f9952d1b6c8 : SUCCESS in 54m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cae2e23187f444f4ac9f55e14fdf7eab : SUCCESS in 45m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72eff60cff5f432da050fcc3d1a114d2 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d394a84c34e4639a989f67d6d53b9f3 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8426e46944bf4dc3a2d87d88484c3d2d : RETRY_LIMIT in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d844a9e7ad44c7be2f51cc626bfb5ef5ea5855d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 17:30:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a28f0e71bab49beaa31a423395696b1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3030b06d6cc949c7bc80bbeb0d606bbe : SUCCESS in 1h 33m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c65094f1a858c8f35ee1aaac8fea1a75eae1936e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-06 13:28:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fcc5dc2302ade47062d958575d23aa87d7eea771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 14:42:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99c9147b4c644684bc9e43fcdadfce1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce89c2abf874018ad1c0fbbc7e86e85 : SUCCESS in 7m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/073074626cc449dfa3e0410b69714f6e : SUCCESS in 48m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0af805f3ff7f454fa46a09b7f3112a64 : SUCCESS in 16m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/242ebd5192554d20b2f2b941eec98374 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/181a7fa01c5143f6bcb64c67dbd975df : SUCCESS in 14m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d6602f48a65d4059b97a6bca84e79e73 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/39de84395107450696cd73b2e0921d56 : SUCCESS in 20m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac0a4eb88e9e4b06a7aefa5d76803749 : SUCCESS in 18m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7747fc524f0c4db7a6426e67aabf87c6 : SUCCESS in 22m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/023d3b1de5e1443b9f250ca0877537d8 : SUCCESS in 27m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/730433fd6de6418db29f77b7d0d5167a : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/36ba1e7f90324f8491d7bb24a312cc6a : SUCCESS in 15m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/36182633d2284e42bc7aff3fc87de12c : SUCCESS in 18m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/39ff79292a9f4dc081c05681ae2d8f2a : SUCCESS in 13m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/68b27a67caab462b87463e086bd6dc67 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9d704f7ef96f408287e111344dcb59f1 : SUCCESS in 21m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7e3cff93a2a84ab0827dcd99400b9acb : SUCCESS in 35m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a02f8b37741949cf81b7cc213450e5dd : SUCCESS in 54m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8249e87282f7436782521cf983ec5398 : SUCCESS in 22m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af0963b999374b0ca30eed0132f0ff98 : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18f427d459cd4d799837a25256221e8a : SUCCESS in 17m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5f659609f7dab6169ca34d96b6d46cadfccba8b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 15:10:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/808ed4dc369e414890133639362e2f1d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0f47c7d8ad3492c82fe26f37078dcbe : SUCCESS in 1h 22m 46s","accounts_in_message":[],"_revision_number":1},{"id":"67bf493ae565607b9bbaf0cd1fa930db13073bf8","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-12-12 22:26:58.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f119372514018b6303182defc8f6b83e0f8f3a59","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-16 14:41:50.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3b2a18932e25b8d7e4c251f9cbaa414cca29fc34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-16 14:42:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0ced607ee458a069993b0d1819e9cad0856460b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:12:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/456b4bc8815c4660913ebc607352ceed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b37d6fd9c28b4a769b347c68db632659 : SUCCESS in 9m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59cb89f4a31d4cf6926e73d0960f1c32 : SUCCESS in 1h 00m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/923ce11ddab9439996da33d38075cba7 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/768c1fd3d6c74e34bbf361e87d429ef5 : SUCCESS in 22m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9adaccb20df24662a65f9f040c39b9f0 : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f751486d18824a28b8acf235c5fc4da7 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e8a049ded0f44185aac51292d676d887 : SUCCESS in 21m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2ae97f50b290480096b6070b9d9123f5 : SUCCESS in 19m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fcdf74cf1369400a8303a21f36200ea1 : SUCCESS in 21m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/26bc0a90d3ed454b80f49c6f7a2d488a : SUCCESS in 27m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b4322dab5f494b86bf96c9aedaae25d3 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/831d4e040c4a4e82889a200591fff828 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/28f1bbfe78da4bd4b78e98f61c9fd5fd : SUCCESS in 17m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2e173d05729a4e328ab349c86796ca47 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6823adf9c40343d98e244afc0e622595 : SUCCESS in 10m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/270ce32f34ed477bacd5b5d4af4f03ee : SUCCESS in 15m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f32fed70d7e4c2daa6b6ecd81957140 : SUCCESS in 59m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c1b79fd76404b22a1ec4bc985467953 : SUCCESS in 43m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61ca8ed8e8e44bb182aa112ff6510aad : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d172fe13f9bc41c1bc7b65e57ce17629 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a69a7a46938a4f9fb9cf163652d337e7 : SUCCESS in 17m 24s","accounts_in_message":[],"_revision_number":2},{"id":"77495cf0c23238dadce657d62e07b7da9eafe44a","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-12-16 16:22:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3c5ebfe8a0c00153397236b7875747cf5d53d48d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 17:34:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f89ac3f647244aeac0467dad9f0e43e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/67db5ead123b4cc9986a4b6d01428079 : SUCCESS in 1h 21m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f4918e378594de93c481ee1d697b97062662666b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-01-06 15:12:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d6849e0305ceeb1d34faf9b664382023f3cfebf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-01-06 15:28:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2029f7e3aeadeda7df27bcbcad288095d211828c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-01-06 15:47:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9204a851b150dd729f2f434e4dd46532c0389754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 16:58:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8be27d957aeb4edda788af1e682cc4b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/996d9774495a428b9f588904aca5f610 : SUCCESS in 5m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/2655369cbcd24da6b778673b2631174a : SUCCESS in 36m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e8131192f7dc4be79146fa115cf1f7b4 : SUCCESS in 35m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d2646f4167724215a5639a6e38650216 : SUCCESS in 1h 04m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4cc9e0882a04b6eb86a4c067ede7439 : SUCCESS in 1h 05m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a76de527b73e46e98a17c109d88ef82c : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0d2ee0d418a4d8c94656f276a607e47 : SUCCESS in 12m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d8aa4ee5275a438882379909fcfd0bf7 : SUCCESS in 12m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d02a3689b7b1428ba0914f0802577293 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/42a990581b5f499780cac2063a7d9371 : SUCCESS in 8m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/790c00b0aefd48aca02af540b3284a8a : SUCCESS in 20m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d17088be915e41e8ade9b190ba3a32bb : SUCCESS in 28m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ee91faa209cd4896bfc64fa3fe771cc0 : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2da2b32122b940199cdd6fd6cf846402 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6ca0b5c570ee4625b5fb63a5ddd8bb1b : SUCCESS in 14m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/66d0915bbf954f7ca8e56e3db395b703 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d84236c2874461dbdcce44ec12c8750 : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e356e3d032b468484694cf212d8814b : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c96e9781307e46fd944fdc1d0ef65d5b : SUCCESS in 53m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/441eaad55686463fb9e629a7c3256b8d : SUCCESS in 49m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7d9b889e8a144579993ab6e9b497a63b : SUCCESS in 35m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a6f0cdbb23c4a02929be20677a44d15 : SUCCESS in 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d0de9eaf1a1490399fa2ebd1a3a2c01 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8103f2c4c9e04763b10d2ddb7e0d52fc : SUCCESS in 21m 12s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80882e4bcad3481687ba46f94d29b159 : SUCCESS in 54m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48ca160945b34144bc30c32b775262ab : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":5},{"id":"5c92e673f7d98f628af6e90c7abd3da24d3d318e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-06 16:59:46.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"60db9a29f1844a8d251b854d20a0330b80970aeb","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-01-06 17:04:24.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1c3de277859b6ce7f73a66f859454ff21aa311b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 17:09:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/421c4726d2964585a54fbe95782c1785\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9c36f39470b14920b28daf65c64eeb76 : SUCCESS in 1h 20m 00s","accounts_in_message":[],"_revision_number":5},{"id":"5bd93fd042c2539e9f1c939091e673e8091e1751","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-01-07 14:31:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a4166625edcea75d6cd8a12ba20670211967fee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-01-07 15:23:01.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"7b3493d88759e708345ea06a2a8e24ff2497a248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 16:32:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9816acdda7c43c3b13a48aa34d51fc2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1562e5194b0c4704aa4d0618bc16b894 : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/f2c5634b0aa24993b12f5c38f7a08e77 : SUCCESS in 55m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ecdcfe74f484c1bb076a9dda7938fc9 : SUCCESS in 58m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f7085f62bbf747238db1e11bdf955434 : SUCCESS in 53m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f6216d94341442d9556e94ef2db5879 : SUCCESS in 1h 08m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/839b0b221bd749919a838f4a44f207d8 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2ba981c13ca418aacb96269bef048cd : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3633f9c4c204c3088f3e2d769b742b3 : SUCCESS in 18m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28fc7acd48a5415b9b74f7cea7bcbe32 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/adc8946fcd2748c4903ae112ce983e7f : SUCCESS in 17m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f3c46452ea8c4632a14e93737f3f8820 : SUCCESS in 16m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1c8e898962ca47e1a3b4d3bbd9739900 : SUCCESS in 34m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e4e1bfddce48489f8866535b84130bba : SUCCESS in 13m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7bc82570b7154f0c8fd208b073a2b775 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b821b2797dd442d095e31e67b3931c32 : SUCCESS in 18m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/56c6d7e1295c4b908f80842200d7bea3 : SUCCESS in 13m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a03d5be0464440dcbb293bfe5de12ba2 : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ad320a4a5bed4067ac13453d5a092475 : SUCCESS in 8m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d391e182cb8746cbbd240903b1fe6a6f : SUCCESS in 55m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5d6609094c44f86b1e3b92501e6746b : SUCCESS in 49m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c12b4e87ba3b47df8630ee52e5534cb8 : SUCCESS in 48m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e592009dc9e840fbb504895e984c8f22 : SUCCESS in 26m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/971fcaa3db6a45a1bdbbac3b5973ebac : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f4de4d3507a5483b9598268e746f1d2a : SUCCESS in 25m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21ba2b4611784a3499c950dd1c613296 : SUCCESS in 1h 02m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d70d1f768e014c05a357e5a95c47f81b : SUCCESS in 52m 43s","accounts_in_message":[],"_revision_number":6},{"id":"d365da50ca6ba492f4bf2b2375112677d5a3fc72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 17:02:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d84ca71ffbed4b7bbaa174b3281720c6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dd9ae539610141aaa92c194d3692dbac : SUCCESS in 1h 26m 21s","accounts_in_message":[],"_revision_number":6},{"id":"f9d14a1906186061d52af0a6b4f616aecb9a0d5e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-07 22:41:24.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(22 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a63c7b3289fba9d92c6ced6a5b8e250905d5e43a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 22:41:59.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3e010bcf7d09606a7838139b9fc070fc878de192","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 23:46:14.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02b11ca1024b470e9c32a36acffc6aad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30edb952c4204ffc81e394dd85811481 : SUCCESS in 6m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/17d79b9aa3d342679c342ceec2976eeb : SUCCESS in 54m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc2b0dbcbca046be9a755b50cd8831d5 : SUCCESS in 40m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62ded240fab840328a26f29f8386d71f : SUCCESS in 1h 02m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8af43e82e984ea19682284e10b41580 : SUCCESS in 1h 04m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/263514bfb5ac444b914d97cc1556c654 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ab5d408d0ba4f16817709c3afcdfa1b : SUCCESS in 20m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d14911f2ab294acf9efeb15b652b3212 : SUCCESS in 19m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0bdbf81384664bcf99f1dce0dd7a264d : SUCCESS in 28m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/96fb2b3267b24432a2665e8e3d5b4272 : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d1edfb89d173473d8b3937cb96db957e : SUCCESS in 15m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7ffe1d6008dd44c0b018d3ca04c51d27 : SUCCESS in 14m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2481bf737c7541af90a98fd6e04e2930 : SUCCESS in 3m 58s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/863147c0887a487ea698656c25394b47 : SUCCESS in 6m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e5fa8339de34786adb4c890bd002652 : SUCCESS in 52m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f3c49bd54241438ba92f49be0670679e : SUCCESS in 24m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ede1523bf835430294b84892ab04211a : SUCCESS in 49m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a00e61417714352b933b049b0adace6 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45abf44afe35485db95431ad9a63839e : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a1ef48d54b3a48c289a53a15cc979e92 : SUCCESS in 10m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb1fab95c648434580d5ffc94f7c602d : SUCCESS in 48m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6a6f5e67bb247b5a7ca95065733f9d2 : SUCCESS in 40m 59s","accounts_in_message":[],"_revision_number":6},{"id":"65a5aab2cd43da22bf4c5cfa02f89cd91acb7771","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 23:46:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"b9660d91ac98f32791641e63f47c659e6c2b059f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 23:47:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69a13db49558492da00034c5355b82fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a98b0199f654efe9e5b98178401e652 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/209a407ac938408c972ce07a05a6f0d3 : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7f8fe153c0959203267b4e2f665dc7d30ddd44f9","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-01-08 03:20:26.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"77e1e571e904320b6b3732f5c3ecd6cdfda8e27b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-08 09:37:01.000000000","message":"Patch Set 6:\n\n(16 comments)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7d3a32d107997c7cf1b6df590d57c0ccc4d13a42","revisions":{"f6f0c8ef963c739a05e3ac4f9d3ef77e562513b1":{"kind":"REWORK","_number":1,"created":"2024-12-05 15:49:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/89/937189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/937189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/937189/1"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-05 15:40:33.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-05 15:40:33.000000000","tz":0},"subject":"tests: relocate some logging related unit tests","message":"tests: relocate some logging related unit tests\n\nThe Related-Changes moved logging and statsd components into new\nmodules and decoupled statsd from the logs module. This patch attempts\nto re-locate the related unit tests to appropriate modules.\n\n* tests for utils functions such as get_logger should be in\n  test_utils.py.\n\n* tests for log functions such as get_swift_logger should be in\n  test_logs.py.\n\n* tests for statsd client functions should be in\n  test_statsd_client.py.\n\n* tests related to patching a SwiftLogAdapter with a StatsdClient\n  interface should be in test_utils.py.\n\nChange-Id: I3a677bb67c5700da48f89c847f652b4610ba47c2\nRelated-Change: I44694b92264066ca427bb96456d6f944e09b31c0\nRelated-Change: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\nRelated-Change: Ie73988edf6be0e38d9004bee04ff46c906a759ff\nRelated-Change: I5ae2cc5c257fb8d7eab885977d9d9cf602224ec7\nRelated-Change: I4b5b12a3b0288b696a39903264741bc862a94ad7\nRelated-Change: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6f0c8ef963c739a05e3ac4f9d3ef77e562513b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6f0c8ef963c739a05e3ac4f9d3ef77e562513b1"}]},"branch":"refs/heads/master"},"288eb4bbcde5e18144a8f7ec574f8b38190304fa":{"kind":"REWORK","_number":2,"created":"2024-12-16 14:42:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/89/937189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/937189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/937189/2"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-05 15:40:33.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-16 14:41:28.000000000","tz":0},"subject":"tests: relocate some logging related unit tests","message":"tests: relocate some logging related unit tests\n\nThe Related-Changes moved logging and statsd components into new\nmodules and decoupled statsd from the logs module. This patch attempts\nto re-locate the related unit tests to appropriate modules.\n\n* tests for utils functions such as get_logger should be in\n  test_utils.py.\n\n* tests for log functions such as get_swift_logger should be in\n  test_logs.py.\n\n* tests for statsd client functions should be in\n  test_statsd_client.py.\n\n* tests related to patching a SwiftLogAdapter with a StatsdClient\n  interface should be in test_utils.py.\n\nChange-Id: I3a677bb67c5700da48f89c847f652b4610ba47c2\nRelated-Change: I44694b92264066ca427bb96456d6f944e09b31c0\nRelated-Change: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\nRelated-Change: Ie73988edf6be0e38d9004bee04ff46c906a759ff\nRelated-Change: I5ae2cc5c257fb8d7eab885977d9d9cf602224ec7\nRelated-Change: I4b5b12a3b0288b696a39903264741bc862a94ad7\nRelated-Change: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288eb4bbcde5e18144a8f7ec574f8b38190304fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/288eb4bbcde5e18144a8f7ec574f8b38190304fa"}]},"branch":"refs/heads/master"},"d3830da477edee0794b98f5d1bd5a5c67ff55040":{"kind":"REWORK","_number":3,"created":"2025-01-06 15:12:40.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/89/937189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/937189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/937189/3"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-05 15:40:33.000000000","tz":0},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-01-06 15:11:33.000000000","tz":-300},"subject":"tests: relocate some logging related unit tests","message":"tests: relocate some logging related unit tests\n\nThe Related-Changes moved logging and statsd components into new\nmodules and decoupled statsd from the logs module. This patch attempts\nto re-locate the related unit tests to appropriate modules.\n\n* tests for utils functions such as get_logger should be in\n  test_utils.py.\n\n* tests for log functions such as get_swift_logger should be in\n  test_logs.py.\n\n* tests for statsd client functions should be in\n  test_statsd_client.py.\n\n* tests related to patching a SwiftLogAdapter with a StatsdClient\n  interface should be in test_utils.py.\n\n* Improve get_logger tests re. statsd prefix\n\nChange-Id: I3a677bb67c5700da48f89c847f652b4610ba47c2\nRelated-Change: I44694b92264066ca427bb96456d6f944e09b31c0\nRelated-Change: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\nRelated-Change: Ie73988edf6be0e38d9004bee04ff46c906a759ff\nRelated-Change: I5ae2cc5c257fb8d7eab885977d9d9cf602224ec7\nRelated-Change: I4b5b12a3b0288b696a39903264741bc862a94ad7\nRelated-Change: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3830da477edee0794b98f5d1bd5a5c67ff55040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3830da477edee0794b98f5d1bd5a5c67ff55040"}]},"branch":"refs/heads/master"},"b81a7fa1a07de2a5ce3249cb0c05d5d0b88d6c07":{"kind":"REWORK","_number":4,"created":"2025-01-06 15:28:25.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/89/937189/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/937189/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/937189/4"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-05 15:40:33.000000000","tz":0},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-01-06 15:27:20.000000000","tz":-300},"subject":"tests: relocate some logging related unit tests","message":"tests: relocate some logging related unit tests\n\nThe Related-Changes moved logging and statsd components into new\nmodules and decoupled statsd from the logs module. This patch attempts\nto re-locate the related unit tests to appropriate modules.\n\n* tests for utils functions such as get_logger should be in\n  test_utils.py.\n\n* tests for log functions such as get_swift_logger should be in\n  test_logs.py.\n\n* tests for statsd client functions should be in\n  test_statsd_client.py.\n\n* tests related to patching a SwiftLogAdapter with a StatsdClient\n  interface should be in test_utils.py.\n\nImprove get_logger tests re. statsd prefix\n\nAdd a test for statsd_client_source robustness\n\nChange-Id: I3a677bb67c5700da48f89c847f652b4610ba47c2\nRelated-Change: I44694b92264066ca427bb96456d6f944e09b31c0\nRelated-Change: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\nRelated-Change: Ie73988edf6be0e38d9004bee04ff46c906a759ff\nRelated-Change: I5ae2cc5c257fb8d7eab885977d9d9cf602224ec7\nRelated-Change: I4b5b12a3b0288b696a39903264741bc862a94ad7\nRelated-Change: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b81a7fa1a07de2a5ce3249cb0c05d5d0b88d6c07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b81a7fa1a07de2a5ce3249cb0c05d5d0b88d6c07"}]},"branch":"refs/heads/master"},"5299610d03b5468e614691a717a3b400215b9f14":{"kind":"REWORK","_number":5,"created":"2025-01-06 15:47:15.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/89/937189/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/937189/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/937189/5"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-05 15:40:33.000000000","tz":0},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-01-06 15:46:57.000000000","tz":-300},"subject":"tests: relocate some logging related unit tests","message":"tests: relocate some logging related unit tests\n\nThe Related-Changes moved logging and statsd components into new\nmodules and decoupled statsd from the logs module. This patch attempts\nto re-locate the related unit tests to appropriate modules.\n\n* tests for utils functions such as get_logger should be in\n  test_utils.py.\n\n* tests for log functions such as get_swift_logger should be in\n  test_logs.py.\n\n* tests for statsd client functions should be in\n  test_statsd_client.py.\n\n* tests related to patching a SwiftLogAdapter with a StatsdClient\n  interface should be in test_utils.py.\n\nImprove get_logger tests re. statsd prefix\n\nAdd a test for statsd_client_source robustness\n\nChange-Id: I3a677bb67c5700da48f89c847f652b4610ba47c2\nRelated-Change: I44694b92264066ca427bb96456d6f944e09b31c0\nRelated-Change: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\nRelated-Change: Ie73988edf6be0e38d9004bee04ff46c906a759ff\nRelated-Change: I5ae2cc5c257fb8d7eab885977d9d9cf602224ec7\nRelated-Change: I4b5b12a3b0288b696a39903264741bc862a94ad7\nRelated-Change: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5299610d03b5468e614691a717a3b400215b9f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5299610d03b5468e614691a717a3b400215b9f14"}]},"branch":"refs/heads/master"},"7d3a32d107997c7cf1b6df590d57c0ccc4d13a42":{"kind":"REWORK","_number":6,"created":"2025-01-07 15:23:01.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/89/937189/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/937189/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937189/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/937189/6"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-05 15:40:33.000000000","tz":0},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-01-07 15:22:44.000000000","tz":-300},"subject":"tests: relocate some logging related unit tests","message":"tests: relocate some logging related unit tests\n\nThe Related-Changes moved logging and statsd components into new\nmodules and decoupled statsd from the logs module. This patch attempts\nto re-locate the related unit tests to appropriate modules.\n\n* tests for utils functions such as get_logger should be in\n  test_utils.py.\n\n* tests for log functions such as get_swift_logger should be in\n  test_logs.py.\n\n* tests for statsd client functions should be in\n  test_statsd_client.py.\n\n* tests related to patching a SwiftLogAdapter with a StatsdClient\n  interface should be in test_utils.py.\n\nChange-Id: I3a677bb67c5700da48f89c847f652b4610ba47c2\nRelated-Change: I44694b92264066ca427bb96456d6f944e09b31c0\nRelated-Change: I8988c0add6bb4a65cc8be38f0bf527f141aac48a\nRelated-Change: Ie73988edf6be0e38d9004bee04ff46c906a759ff\nRelated-Change: I5ae2cc5c257fb8d7eab885977d9d9cf602224ec7\nRelated-Change: I4b5b12a3b0288b696a39903264741bc862a94ad7\nRelated-Change: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d3a32d107997c7cf1b6df590d57c0ccc4d13a42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d3a32d107997c7cf1b6df590d57c0ccc4d13a42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
