)]}'
{"id":"openstack%2Fswift~937248","triplet_id":"openstack%2Fswift~master~Ic802d87d5539c36a3fb888fbc3962cc6cc3b2448","project":"openstack/swift","branch":"master","topic":"p-update-rhel9-os-release","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-10 09:59:27.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic802d87d5539c36a3fb888fbc3962cc6cc3b2448","subject":"ci: update the rhel9 os release","status":"ABANDONED","created":"2024-12-06 12:04:35.000000000","updated":"2024-12-10 09:59:27.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"5fc980fe91e192fb070f48c7f1fc55a87cf6885d","_number":937248,"virtual_id_number":937248,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-06 13:32:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 13:19:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-06 13:32:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-09 21:56:54.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"8f58f6d0dfe4672c3e1caa0981e97676fd025bf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-06 12:04:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60e15f68f7f52ad8153785852bfac2abae8da011","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-06 12:09:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d6076e62780172bd9d20fe104be49f8a77aa39f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 13:19:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a38fcf52c14b09832a37f6db3ff89a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0430a81012894a2399796830605038ae : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"39c553c52a9a3c7891658637d7e37ac8e9ea1933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 13:32:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b561d80ed0024df9be06dfef7a168997\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bfc0e9d1bc249849500423e0a77e6d6 : SUCCESS in 6m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/872df131ee894b239a9d80b515dca56c : SUCCESS in 38m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/95c8b1e5d5854636a9d0265fbcbe24cf : SUCCESS in 36m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad0dbeb2b9704c6ea34894d2cf471449 : SUCCESS in 1h 02m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc4bb475ad594f4caf3060dc9b37ddc2 : SUCCESS in 39m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce61ed076d3342dda4004d4696ce6a45 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd1a2fa5126b4a56b9f9a2f4e8f76748 : SUCCESS in 13m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/417baa3de068434e8d85d16902c723ba : SUCCESS in 20m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f06da88425e548a3a2f785bb187c8c41 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/08437e6dbf1b4050add274d7e4d3f7ad : SUCCESS in 20m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/82b146a6325042c78dfd5227805c483e : SUCCESS in 18m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4abb52e5260447c28b17119add2b3f9b : SUCCESS in 26m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/52c8139160414a26876ce65384b5d0cb : SUCCESS in 23m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c13c24d931b94849804f55b5fc1b2df1 : SUCCESS in 24m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad96db01093941e9a8b5f34f4fad1491 : SUCCESS in 24m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d2a57344a1c4a31965da9312e0e85f9 : SUCCESS in 26m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/96a262b3bdbc43d6b8d3862e98721bfc : SUCCESS in 19m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f8cb92783e014003a4580b00dcaefd0e : SUCCESS in 20m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6f318c0f2f1d43149bfb59986e60a501 : SUCCESS in 20m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/07e0dac3e14a482a95cd6449bd8fb61c : SUCCESS in 8m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b3ca9803b964e4f91a0e2b15345b495 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bf3c6f5b8b0541c5915ba0c77d03754d : SUCCESS in 6m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bcceb2d5d7743b4a3bdc30d81ec1f97 : SUCCESS in 1h 22m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e22fb9c702b54013bfdaaee4c1e85b50 : SUCCESS in 1h 00m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5ba443b08dc9482db90ec9ca7e5c4cdb : SUCCESS in 38m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19ea69580a23496ca8824dfcf1bbfe25 : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb9bcefb9b24ba182c68e6c94377ece : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/31182e2c52a74661adc9b2b15c8ff734 : SUCCESS in 21m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/728a95a220334002a7453b5fc503245f : SUCCESS in 56m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8536f047307e40f8b3e46fc2dade08f3 : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":1},{"id":"db45896c1e04f93b07a2c869c27fab613a8cf05e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-09 21:56:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fc980fe91e192fb070f48c7f1fc55a87cf6885d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-10 09:59:27.000000000","message":"Abandoned\n\ngate fixed itself and Tim says this soon won\u0027t be necessary anyway","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9007bd3e4c91e09c8683b40231a4db400f6202c","revisions":{"b9007bd3e4c91e09c8683b40231a4db400f6202c":{"kind":"REWORK","_number":1,"created":"2024-12-06 12:04:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/937248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/937248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/937248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/937248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/937248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/937248/1"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-06 12:03:40.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-06 12:03:40.000000000","tz":0},"subject":"ci: update the rhel9 os release","message":"ci: update the rhel9 os release\n\nChange-Id: Ic802d87d5539c36a3fb888fbc3962cc6cc3b2448\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9007bd3e4c91e09c8683b40231a4db400f6202c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9007bd3e4c91e09c8683b40231a4db400f6202c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
