)]}'
{"id":"openstack%2Fswift~937252","triplet_id":"openstack%2Fswift~master~I1e096dab9a97956bf786ccbcd37c20f9a3a5429e","project":"openstack/swift","branch":"master","topic":"make-async-tracker-timestamp-float","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-11 21:22:32.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-11 19:40:14.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I1e096dab9a97956bf786ccbcd37c20f9a3a5429e","subject":"Add unit test for object-updater recon dump","status":"MERGED","created":"2024-12-06 13:14:00.000000000","updated":"2024-12-11 21:24:01.000000000","submitted":"2024-12-11 21:22:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"937252-make-async-tracker-timestamp-float","meta_rev_id":"60b6a491bfb11c3407a4dd72a8c929c1aea25354","_number":937252,"virtual_id_number":937252,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-11 21:22:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-11 19:40:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-11 19:40:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 14:22:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-09 21:54:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"ce6379a96017c4a865e74d44ef7edfb9b3b1f040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-06 13:14:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5cb3aa4cbee64c8eb49f8528e3d0eba233c8f43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 14:22:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/225fc31c2bd642498c657341d3af0a98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3992a9b6202a4547acd9a6f61c3d4ac8 : SUCCESS in 8m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b911e33dfae142e59fadce4840f4c6ab : SUCCESS in 48m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/52f39648f5a44bf294dee9a7bea2dc49 : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b52f9371f174b3c839fc3dfe9351df6 : SUCCESS in 23m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95d5746065434ef5ba25bd0f99ad11ba : SUCCESS in 23m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bccefffd2e4944c1b521e7dbe5a75cd9 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/23df1a46faed4bd49b3929fe26220875 : SUCCESS in 22m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f63b35352db4082aed00ca8227e113c : SUCCESS in 17m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ea0e1e88573c4cefb24de7a06a0a64ff : SUCCESS in 20m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a6141eaf583a4c76b4eac963e0b770a1 : SUCCESS in 35m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1f141a85f1f54ad3bda9fc8fbaeb8d46 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4218d171c3654a9cadc019c9afffadfd : SUCCESS in 16m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eef3320ba1cf4c8883fc5b0b5608fa5e : SUCCESS in 16m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c615f9bce72c4b61bb8639d5c199a0ca : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/18c08352c10243c2923bbf33857a1a21 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1547cd368be249a6941678c73292db15 : SUCCESS in 14m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e20c6c6f91a24414b0094c6ef2c7d8f3 : SUCCESS in 34m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/baf211293e2e439ab704828981452431 : SUCCESS in 45m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cedcb4aa8f2644a1b1bcb1b87f5b4811 : SUCCESS in 19m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d88f9dcbb876474d82f5ef5ee300c557 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76d645cb062543eaab130672396b04e7 : SUCCESS in 19m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d2639e4000379e76bc23e910b8158f41a79eb456","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 14:53:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a92aebdf0da64b8fb99f434af55ea8c8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf55f0c068d74b6dae24a748874eec57 : SUCCESS in 1h 07m 44s","accounts_in_message":[],"_revision_number":1},{"id":"def0e10b16f73cb70e7dc1d2c3b03b279c8b2a7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-09 21:28:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"660065939b468574cd3ecd093c3bd7a04d2b4b38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-09 21:54:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c7579a896eaa214e481d51cf0fc983530dccadc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 22:26:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74e84e2e2e2a4f2c8470b9605f4484bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3caf2e59f2f14357aeb86f2e02fc2e92 : SUCCESS in 9m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7cb790ff5bc4014b7c5ad9fc80d3f79 : SUCCESS in 53m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e8a93d9e4334a1a9a1486aacb19680d : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64c7189b5c9647b59cb1fdc872866bc5 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/171abcd6b2e54d74b40fdc8b92672def : SUCCESS in 21m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/845d50213c4945e5b0cd379998a37ccd : SUCCESS in 4m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/900ffa60290b4109b0acd08f37b0bb69 : SUCCESS in 9m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/380ccb4e9f8b49699dd6437b8cde0499 : SUCCESS in 8m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d7525899c6ad4f58915c6c717a58b954 : SUCCESS in 9m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/40621fda78d94bb181aeda72cc5cef66 : SUCCESS in 13m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/328272ed76e04d0a9ff78831810c9e12 : SUCCESS in 9m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a1fd787ff4da42dfa125485f63361c72 : SUCCESS in 23m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/be53cfd1a1da45cb8028ba439a12be6c : SUCCESS in 10m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8efc19faba994bfdaf43664435212282 : SUCCESS in 9m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/881880593d81476cafffe0542ee43217 : SUCCESS in 9m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/331c586afee64be183c4bd8413fc0cbd : SUCCESS in 9m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30b3c577135f4ac2a24afdf29b97e8cb : SUCCESS in 41m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a774279bee5a4644bf09b71810dba05c : SUCCESS in 25m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29e8cf524bad4921b0a14e482d9c2cd6 : SUCCESS in 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9415b5c34dfc4975a477a0238a703d29 : SUCCESS in 3m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ce47d842c4346faad0fc2c59ffd3a2c : SUCCESS in 17m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3ec5f5f5bb34e00772bc57f245d4f8145f652cab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 23:23:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce4adc6200041ab859abfe7d3859708\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ab31f6adc8c4d798a6bbe72ab166788 : SUCCESS in 1h 23m 34s","accounts_in_message":[],"_revision_number":2},{"id":"37d45ce36a8a2684ba7764202fa63f24b6833804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-10 10:27:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c171283c66c4ec3e8db8e8249e77fa152502120c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-10 10:27:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5454e389a97f06a7474e8eca9f5b09f2c0abce0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 11:23:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cda8cb4b555f4c62a09c1a131a9ec85d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a2a85ca3874eacb41fe1307176c5c5 : SUCCESS in 5m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c038c59eb5c4025b9e73bd1a212a4fa : SUCCESS in 43m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9dcc7b1f78549c380c54fcc6c1fba91 : SUCCESS in 12m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/918c00d261424db8934ae4167bfe07d0 : SUCCESS in 22m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bcbc1abdf174ae3a22f5ba820d497b1 : SUCCESS in 24m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a95ddf1b0084d4a8c70b42cb3c7a314 : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0e670e0b3da8499eabe019b8d07d4649 : SUCCESS in 20m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4fdfa46ca9d4ff0858d06d9f3573c8a : SUCCESS in 22m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ac8eb51b70714b7182f84f755b4293da : SUCCESS in 17m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/29cb9cd0f24347f98a776b2544d9cf4a : SUCCESS in 28m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a89878dbef14bf4b988b504115924bb : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7e60423ed9c14179a887d7ff20a8131e : SUCCESS in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8f4e7c0969c64340b751758b1e64227d : SUCCESS in 17m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/055e73ac7c4f4f0f923458536cc6d762 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b286b7ba41de4366b33b02ad6ab9dac3 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/137be23f732b4441bba427bd049d05dc : SUCCESS in 14m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c63a7484fe9a47a7817530ae3b67d17e : SUCCESS in 50m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/43b9988f907f4b3bbbc63b2bf7e18189 : SUCCESS in 47m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80a8612be4554918ac77316334bdadd4 : SUCCESS in 18m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be1d68ba68d744ab90f03003a9f953e8 : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2af8ee6961db469595f4de0038f8b9a4 : SUCCESS in 24m 59s","accounts_in_message":[],"_revision_number":3},{"id":"4d871d9b83157ace4f3779c432b73c4cbb6d4ddd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 12:08:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9461d994cd44258aaa509fad9075bed\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c280a64b29846bf8725dd21bb7db6f1 : SUCCESS in 1h 39m 15s","accounts_in_message":[],"_revision_number":3},{"id":"abf81400e238dba1ae6ba3650f6d40eea9a2d737","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-11 19:40:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"52de8c2276ab69ac18e25e75611b2866a62faa33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 19:40:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ebe9c3ade85aba669175ea8c88f6716bed1590ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 21:22:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b32e924515ff4b4e83484c8c24d385a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28dada7b65914b76a770671846c7ce89 : SUCCESS in 9m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a723c7176cf484b862b87cce64f0d6e : SUCCESS in 54m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/306fdacfc61b45e3adba2f2b3e2106bb : SUCCESS in 7m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d9aaf919f3c140858f6f11eee71bb20e : SUCCESS in 18m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a84bb4ada2bc40d59b93fb5c3c314b25 : SUCCESS in 16m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2ee1739990c940ec822729a9d1c7fed9 : SUCCESS in 17m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9abbe1ba0d414302a158474abd2922c1 : SUCCESS in 24m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/292dfaccfa1a46f593050382531ca153 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/14087de053c54518a8e1580ab9c8150f : SUCCESS in 16m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/70d66ca5402148f880cd2cb9e5564952 : SUCCESS in 16m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e7321f1414934743baeedd5a879146c0 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b1795f6bc43e4210bb9bd9a4ad183f77 : SUCCESS in 12m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fe3d4af6e3384bef8a12e106c53ef2a4 : SUCCESS in 9m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d2f4a9339b744ffaa66fa40e5e319dd : SUCCESS in 6m 42s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4c31a87a71a945d09fdf9fee864fa073 : SUCCESS in 7m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55c1a17498ee4617a92791ebafaf242d : SUCCESS in 29m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92fc93b87a824914a54fb3f7a710745a : SUCCESS in 25m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e88acff229345e18d1f96eeed630594 : SUCCESS in 21m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9eb7b96a53a4dd0b1acf5ddffb91969 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd3b8807f4764e2fbfeb95af5c5f2c37 : SUCCESS in 11m 28s","accounts_in_message":[],"_revision_number":3},{"id":"82a2e3b5adf89e181c58d93fa1da3942388de55f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 21:22:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"60b6a491bfb11c3407a4dd72a8c929c1aea25354","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 21:24:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca3646eac364f4887cb3145951e9f7b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343bc1ad3c9e4f128e579da946affe98 : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a57c02b714854b99b4a5bbba533f774a : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cde99ff660b8211c154de348160f177bcd426f0b","revisions":{"3571b9a77f958629ea1136c9568cd63e9cd04c92":{"kind":"REWORK","_number":1,"created":"2024-12-06 13:14:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/937252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/937252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/937252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/937252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/937252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/937252/1"}}},"commit":{"parents":[{"commit":"e2fe24bf9386543630aff737a0fac3e7b720c863","subject":"Make OldestAsyncPendingTracker timestamp float","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2fe24bf9386543630aff737a0fac3e7b720c863"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-06 13:08:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-06 13:09:39.000000000","tz":0},"subject":"Add unit test for object-updater recon dump","message":"Add unit test for object-updater recon dump\n\nRelated-Change: Iba43783e880e0860357ba8b9f0a11f28abf87555\nChange-Id: I1e096dab9a97956bf786ccbcd37c20f9a3a5429e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3571b9a77f958629ea1136c9568cd63e9cd04c92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3571b9a77f958629ea1136c9568cd63e9cd04c92"}]},"branch":"refs/heads/master"},"bbe46ab88f57f3afa2f3092f7eaa9e5d54ce8490":{"kind":"NO_CHANGE","_number":2,"created":"2024-12-09 21:28:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/937252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/937252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/937252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/937252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/937252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/937252/2"}}},"commit":{"parents":[{"commit":"83528de7432cca17b3a03367422f9342c4a336b5","subject":"Make OldestAsyncPendingTracker timestamp float","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83528de7432cca17b3a03367422f9342c4a336b5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-06 13:08:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-09 21:28:33.000000000","tz":0},"subject":"Add unit test for object-updater recon dump","message":"Add unit test for object-updater recon dump\n\nRelated-Change: Iba43783e880e0860357ba8b9f0a11f28abf87555\nChange-Id: I1e096dab9a97956bf786ccbcd37c20f9a3a5429e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbe46ab88f57f3afa2f3092f7eaa9e5d54ce8490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbe46ab88f57f3afa2f3092f7eaa9e5d54ce8490"}]},"branch":"refs/heads/master","description":"Rebase"},"cde99ff660b8211c154de348160f177bcd426f0b":{"kind":"REWORK","_number":3,"created":"2024-12-10 10:27:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/937252/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/937252/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/937252/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/937252/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/937252/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/937252/3"}}},"commit":{"parents":[{"commit":"83528de7432cca17b3a03367422f9342c4a336b5","subject":"Make OldestAsyncPendingTracker timestamp float","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83528de7432cca17b3a03367422f9342c4a336b5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-06 13:08:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-10 10:26:48.000000000","tz":0},"subject":"Add unit test for object-updater recon dump","message":"Add unit test for object-updater recon dump\n\nRelated-Change: Iba43783e880e0860357ba8b9f0a11f28abf87555\nChange-Id: I1e096dab9a97956bf786ccbcd37c20f9a3a5429e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cde99ff660b8211c154de348160f177bcd426f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cde99ff660b8211c154de348160f177bcd426f0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
