)]}'
{"id":"openstack%2Fswift~937267","triplet_id":"openstack%2Fswift~master~I70a60ae280e4fccc04ff5e7df9e62b18d916421e","project":"openstack/swift","branch":"master","topic":"fix-oldest-async-manager-run-forever-recon","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-12-19 18:35:24.000000000","reason":"Change was submitted"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-12-19 18:35:24.000000000","reason":"Change was submitted"},"37271":{"account":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"last_update":"2024-12-19 18:35:24.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-12-19 17:17:36.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-19 18:35:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I70a60ae280e4fccc04ff5e7df9e62b18d916421e","subject":"Aggregate per-disk recon stats","status":"MERGED","created":"2024-12-06 17:01:54.000000000","updated":"2024-12-19 18:36:45.000000000","submitted":"2024-12-19 18:35:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":164,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937267-fix-oldest-async-manager-run-forever-recon","meta_rev_id":"90083e41d39c63c1f8d2841740dec14d358147ff","_number":937267,"virtual_id_number":937267,"owner":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-19 18:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2024-12-19 17:17:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2024-12-19 17:17:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 18:54:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-06 23:34:59.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-12-12 22:08:22.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-12-18 17:54:57.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2024-12-19 06:06:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"9a509488f6395ac7831e1d185f3b7053bf9789cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-06 17:01:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09c5ab8bc6678df617b994beef48e79c2a3b5c12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-06 17:30:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6805554c463d0da4f80a98e44e6d505cae77061f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 18:54:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a99e46889f0432ea98fec972213f78e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d64d82b59c29495e86db4451e121d061 : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/7f2e5fdd192e4c88b378d97db7464e80 : SUCCESS in 50m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc8c584040284ffda8d186036fbdf9dc : SUCCESS in 54m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/af0a073ff8774912989d5ba941f7016e : SUCCESS in 57m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6672ae4c98794a1f9eaf2e2e999df741 : SUCCESS in 1h 06m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d0328a86a314c7ab150a95f76fe8905 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95a4b19948b74d5eac1d912f320e9a2c : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7c511e5e28d42b98ebd07b932cbaebc : SUCCESS in 21m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b971431816434a0babee2f877f274fdd : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b65c840dba7142ff865902657284b00a : SUCCESS in 9m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4162859fe277416386c192676d068ddc : SUCCESS in 19m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e1c33ab97da2477bbd7714f6504e74b4 : SUCCESS in 18m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/283c6b8197974e7bb2a03e5e6a46c3ac : SUCCESS in 25m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/23520956a9a846bfbb99661844d227b0 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/18150a4426654faba13d4d724f421e02 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/58224400c2304e9b96b3327cc375bddf : SUCCESS in 18m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7a8cbf50465741aca34a5c0e3434bf65 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ae355490748043f680add7da04de830c : SUCCESS in 16m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/08921dc0d803488394d230089fc1afc9 : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1e426649dfe471db1d3fb3a2e84127e : SUCCESS in 4m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c0cd62e2fe12483b8e1e9caf570c3d99 : SUCCESS in 15m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0ba55246f3c0436dafdcc64401acd97c : SUCCESS in 6m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9250d3d924ed4634adb9e861deec265f : FAILURE in 49m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8cc81a6d576a4586899b60749f688579 : SUCCESS in 43m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2385180dc9344b61a5986563d777278c : SUCCESS in 1h 00m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eb731740f2d422a9ec47fd02bfe1ea0 : SUCCESS in 22m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/660609a5e2544f73b41b9562a1652f1c : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74110e0e66a04a75bc9dbcb2fe8c2344 : SUCCESS in 28m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01f4beb5f6ce4ca9be71d32b97e50121 : SUCCESS in 43m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ab065e51fd54f468cc1a8d908dfae36 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":2},{"id":"5997329db73f6e6c67d7c7202768d4832b45ce48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 19:11:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f9f071f5cf749939f9f967d9716854a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9a26e23c7a6b41358728b29f330f8318 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f12bd2082f7983bc2c714bcb26f59b87512ec473","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-06 19:51:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7f75e37ff780e72e751bf9ae72d8a8d15c2bcb7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-06 20:26:36.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"acb2221c5684dd0fa227520cdf13076ae1b040d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 21:35:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68b9dd7f35334e678ae957c8d96f9738\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0b5eaf9827a4f5e92e7add976eb034c : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/bb6f493948e045e4a36fb98bb1b00795 : SUCCESS in 1h 05m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/147859177bfa4a76be34fae8c938ab32 : SUCCESS in 1h 01m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0fafe3a85a3d41c2b3288484b63b253e : SUCCESS in 57m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e14772a2ad34b3a8961b74033de3c4b : SUCCESS in 40m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb9be5a531e34223839bb4ce06aa0233 : SUCCESS in 23m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d74ae73833fb4a14b04600343fc705dc : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bccb0d68e7304bf4b4d67f36d5bf4458 : SUCCESS in 12m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3df18d769f9d4ef1b59bf0cf35481248 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e9f375c99c24c5ca45441d49cf5902b : SUCCESS in 19m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c26d1e928d24e469c24504a54b9345b : SUCCESS in 19m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a1ce51eee7a1483c98c4e815591dde94 : SUCCESS in 22m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/25374c2687fb4dcf9dc348881e465d3e : SUCCESS in 19m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bb3b8052a1bc40e2bc22d11403c21d2c : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6de1baa139c147969e764ca597f7def0 : SUCCESS in 21m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b286f65e5e204d5784d1e6463f19d5c4 : SUCCESS in 14m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3a42f8dc04b84706a55f89e8c0acf0e9 : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/35a62f8f5ea14cb681bb6874a62d237a : SUCCESS in 12m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3c773c33a08f4d5cbf4ba593b3dbd5f8 : SUCCESS in 18m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f6959a84fad246c1b9376c3fe4fe1262 : SUCCESS in 9m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a25cb72008a94b9bab179b7aaa1959de : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/21181af9106545d19fc4261c1c5fae4c : SUCCESS in 6m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/81db5c0484094d0582b41b02769c6fce : SUCCESS in 55m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce08699a826e417fbbec8da1c41c7d1c : SUCCESS in 53m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a513da8dd459454ba7bbb098189ed098 : SUCCESS in 46m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c65e07ba40f41ce836db5c1edecf0f0 : SUCCESS in 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f85e91b56b1a4743a61598979aa26c1b : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d128ba558d744d25a28c6b8ddff8c3b6 : SUCCESS in 18m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/814371b5e4a3466db94f29a4c82d85db : SUCCESS in 50m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5872b615a93f47dd85c1dee77afbdb45 : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":4},{"id":"9727ce15cf91224beacecc46b28a1ade7d5800ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 21:46:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5163d2f7a2dd47dab295b61444f1566a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e7586e6ac7440d195715701e89a6c7e : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":4},{"id":"8ec7e18b0bbebc1a2d3ca2b15afdae2e48d9d30d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-06 23:34:59.000000000","message":"Patch Set 4: Code-Review+1\n\n(18 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5fa7b798aa25fa65de0e15e3181cb2de03b80b98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-09 19:32:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f461eb9a1634e28f1a97acba9fdf7114a0621b25","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-09 19:32:42.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"cd19e3624b6a5c5cf6c948c037f2b32c6dea0272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 20:33:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dc327919184436886f6dd7d926814e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a1e1fc852c4b6294bec02f25a9c32c : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/6f5ab27f684844e497a40df5684a2dcf : SUCCESS in 45m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c581ab796a8140d3ac31c44066bae825 : SUCCESS in 34m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/011107e3f9f347cb8d65dcbdd33d5cc6 : SUCCESS in 41m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9930d1c512a945739d7d5c3334698412 : SUCCESS in 51m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6fa43f06c9a4b208c4535f34038ad1b : SUCCESS in 16m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42e8eb978fd848cb9d3f1318cc29b56e : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9fe85922bc542f583685942412f8c9b : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a3afeabfc1764ca68b52e812fbe62849 : SUCCESS in 4m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/57189edec5754ee093eb0027d7280aa9 : SUCCESS in 19m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f626ef869fb4ed7ab04232eae855847 : SUCCESS in 8m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3bafb052525544f0aaa3428c61585691 : SUCCESS in 14m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3a6856f4855a4a829dd697b86418b57b : SUCCESS in 19m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e224cb603e34e5d9aa69841b9fc2a08 : SUCCESS in 13m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a7363e9e3a96484da72781891a0fffab : SUCCESS in 20m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7d386f12bc154e76aa73bebf13415e49 : SUCCESS in 18m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/559890d361444abb8a700a79393a060a : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/77a0269c84004570bdf18e912afb9e3b : SUCCESS in 8m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/106a32983eb84e21af82f382917a0736 : SUCCESS in 17m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5de414be4ff24309acf2bf863e45522d : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/07e58f3d97b84899a5b6ed4adb79fa9d : SUCCESS in 8m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a2d5dd9d546437fac58bd5afddfa553 : SUCCESS in 7m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9d29b755fde49c1912a8c53503a1c22 : FAILURE in 58m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/42fcaa8cfaea42138606f57c38465376 : SUCCESS in 35m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d63220a76a254998908f37c7e2a01a29 : SUCCESS in 24m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/102591de0adf44f7b3e701f5b28a3f1b : SUCCESS in 23m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/943748bedaab4389ab8f9fcf6313646a : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ecf7a4f181ee45cdb399c9cf9bc45d5e : SUCCESS in 10m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ae2a87f5eed4d8183447a26089c17b9 : SUCCESS in 55m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8936087d395d41ec9e15c71ce1933d28 : SUCCESS in 45m 13s","accounts_in_message":[],"_revision_number":5},{"id":"550b34881a8c3ee8cf6310d0770b90a0761e4c09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 21:37:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7203b9cdc2f843f6af55495219883d36\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/35ce6fa7e84b43f8a40082cb6d79718a : SUCCESS in 1h 45m 46s","accounts_in_message":[],"_revision_number":5},{"id":"30f84662bbed48ea32dda434d27fdc3e3d30120f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-10 01:55:46.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cb8f9fedcbef0a633a4f48ef77c13449c9117344","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 03:50:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22d745c8b79a4dc2a1c44527a008c51a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa52b3a7089648388647336c422c82d2 : SUCCESS in 1h 52m 10s","accounts_in_message":[],"_revision_number":6},{"id":"46c76735f063f9b08d8fd68d8f2b9cfcf65a5ea5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 04:13:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/123f1150000c458fbee3f1b17f9080bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6feafacf45a40858f6408ad8298e01e : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/f2bbe085e17a40c3a40e1cbe38ed05b6 : SUCCESS in 56m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0fd0b0cd70f0448fac132536b0339e80 : SUCCESS in 1h 13m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7dfcb4507af64c31a1189050d8c096ef : SUCCESS in 1h 08m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6335fc694557426c96fd4fd16afa45af : SUCCESS in 1h 07m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/296799b9807d4d07985ec104e74d1919 : SUCCESS in 23m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0f75cea75ea4cedb49579d16beeca97 : SUCCESS in 24m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c607c9a2e82849dba456919588c7f8f6 : SUCCESS in 26m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a84171381d854b7db43ae8b2fb6362f7 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4fe6cc66ab774201bdddbaf98bc9d250 : SUCCESS in 23m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c7587b3294cb4b7e9c20c7d15fd2a328 : SUCCESS in 22m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6f883e7bf3064e1e8204fd442e6ea51d : SUCCESS in 22m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/12a6bcac26364aa788d6b2716858da49 : SUCCESS in 28m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2d8b4c277930426799458133cd42a70a : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bdfab645ca0c4c618578086f542043b6 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/36a456b3e1ff411f85183478b27b21c9 : SUCCESS in 18m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/38210e96762e4998b2a3ef87ee8565d6 : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a86b18c7381543a4aad52d8feda0d51f : SUCCESS in 14m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ce4f96d21c2c4fe187bc7ae58f7b9d46 : SUCCESS in 17m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7dedaf90ecf24e40b811c1df2138aace : SUCCESS in 7m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/816ca6875d074bb1802ac4f26fd952f6 : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9623c19e7db74712bfe4ab5b69e9ca4e : SUCCESS in 6m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d3413f1f466494ab2e345203e8ce94b : SUCCESS in 53m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/550b7320aae947a48f09f9489ccb3d6f : SUCCESS in 31m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4711c226be834ac9a87eee989b699e5b : SUCCESS in 36m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4173df27c0a64b91bf9f5ec80cdf8f76 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5409b9f69c704f6e9b5e39b5c237e0f9 : SUCCESS in 2m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fcc8b4e697c7458d99fab631f100f63e : SUCCESS in 17m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/010fa3d404a5408b957074e896c3e879 : SUCCESS in 37m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c60630e0b7df406280d711a2573a07d2 : SUCCESS in 1h 13m 38s","accounts_in_message":[],"_revision_number":6},{"id":"a60d1ccfefc6e092716c819353f641b9809492fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-11 17:05:37.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"58ff11b705a3e1fcebfffd4be7c19cdec399e519","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-11 17:17:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cfb0ceaebbc185acc7f7bf47a91d6ada67216608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 18:31:03.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a64956ee24cd42bdb18703ee76954a8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45cc04ba0f7047d79d0f1bac37e15e36 : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/40ee6ba9df194549a368c33510170e7a : SUCCESS in 33m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85da0fc0e6e84558a2b99333cda0e7d8 : SUCCESS in 58m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0cfdbb887d1c47c08aaa912be484b786 : SUCCESS in 58m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b304388214fa49a7b8333dccc386b475 : SUCCESS in 1h 01m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/040dcaf298b3482aae739d65f457f915 : SUCCESS in 21m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b087da2934074830a15d376088f71920 : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6673dbeaa584687a029998d152998fd : SUCCESS in 22m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/784f99a1dfc847679a0be93e033b05cb : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1cccfc7d0de7455197b0c2d7c8ad8b5d : FAILURE in 16m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf60be6bda78405ba3cd262dbe409d97 : SUCCESS in 14m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5cec95d017f94983a27f404448fe48a8 : SUCCESS in 16m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b5a9f4956d70447b870828cb8138c7be : SUCCESS in 22m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2f1e258f43ce4532970f7b3f49b1bc40 : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/29f0ee10c4a54de68446f013e70780a5 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2979ca37cb5a4316a95ba0c340f1ef97 : SUCCESS in 19m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/abebe04b1dde41c99bc7367e63758471 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d067873ea9d940c9b90cc294d5851b9c : SUCCESS in 15m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7282faad577343319b419a48b20f1114 : SUCCESS in 9m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b9131bc38c84d528ec24757ea39e83f : SUCCESS in 7m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0c3bb2a4c6154064b21d14816d5306c8 : SUCCESS in 14m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/32a398bd62db423bb735eeb774c876a1 : SUCCESS in 6m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9023eee875d945d5adfaeaca646b87db : FAILURE in 1h 01m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/14725b99b1bc406fbc64ebbcf9a3a87a : SUCCESS in 25m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe6b7079fc3a4a6f864005b3e1807b8e : SUCCESS in 43m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fa4cac744c24a0aa4e6fc3eae38b283 : FAILURE in 22m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3154bccad3742e7887eb42a282470b5 : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/edc80ce282494a2aa65a0c6fea68a874 : SUCCESS in 10m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e42323d3fa6e4cb8873f00cee782ea3f : SUCCESS in 48m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ea2f51719104986ab53f5379bfce932 : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":8},{"id":"d21afa8c9869268828744d4ad4462a7d127b4823","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-11 18:31:19.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"ed20852e555cd372bcd9171ba207c65085c81de1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-11 18:42:33.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1e050d04b59800ca9497dadf69dd645cb8638a99","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-11 18:57:20.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f555b8deebf6d56c6628fb4976e0536136447bb8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-11 18:59:39.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f9e31d5de5366d57fc2e2458697355bec71a24c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 20:23:43.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b989b668a0f49e2bde4f1389d11ae3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec5e81c7bb884fdc8fa42c9214ceb401 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/d34ebb88fe6349ffa3086885f5d51311 : SUCCESS in 1h 05m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/71b34048892543bba562f68fff65c85a : SUCCESS in 1h 01m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/065fa0118e504cf7beefd7ab58c5453e : SUCCESS in 50m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2728f00c6c584abea498fb8854fa079b : SUCCESS in 52m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09734a2352cf4908b39d46bd352cdb82 : SUCCESS in 21m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ca376c4689b4137b256fbee7f3f4849 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4016cb979cf477182761e6118578add : SUCCESS in 17m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c85f33e7ecb4c33b8189ee6e5ffbba5 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e815aed05cfa46f8abdb7774d8f7f14f : FAILURE in 22m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/014372b7a7644e6c82e1902b151de307 : FAILURE in 21m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e91259f634094ef281438ed84591cae5 : FAILURE in 19m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ae661b4712b3419a9ce82623181ebbba : FAILURE in 28m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0a0535fda77424ca439de0cfbf5864b : SUCCESS in 12m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c8c59be0f3604efaaa1f5c60c1beeef3 : SUCCESS in 19m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab70928b796a4a5488a7fce1ed4d0e8f : SUCCESS in 19m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f6ea332ff80b4cb1a766113b30476400 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7e5a2e3d9be649f0b3672eefb325bd90 : SUCCESS in 18m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4cf07acded96435e835a5338d71465ef : SUCCESS in 13m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/da89e81237a54a31b160ddc8e1e182ab : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/52fb49162fcc4153afe0f1feac0a34ce : SUCCESS in 15m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/38a72d0f409641febc5d3043ae14fa76 : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7dc4aae194540799016f9f832790245 : SUCCESS in 58m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d6d62ceac2ae4203ab6d73bf9f6bd9b4 : SUCCESS in 47m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5505207c5048490ab573d6dfbe279078 : SUCCESS in 45m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1e0853b7ed8429fb8f0e36ffac593d0 : FAILURE in 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b228d98524d54be592cf56ccf74c4f0f : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/793dc17033454edda09ab963c17c51c6 : SUCCESS in 19m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af6449abe9d04e83a73bdc625cecd51a : SUCCESS in 44m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5141ecb9c7149768d421627744bef3a : SUCCESS in 44m 42s","accounts_in_message":[],"_revision_number":12},{"id":"646664a1a91dd40cced120a8c7cf4a8ed63b65f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 22:39:16.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bd52022c9c44805b34561a97cdbdd41\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c1e68178a184425abe226a5324d4019 : FAILURE in 2h 03m 18s","accounts_in_message":[],"_revision_number":12},{"id":"5ce8d50f6a51e855a21b505edb2a2a5ab7100b47","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-12 19:09:28.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"eda8b83257174a422394b35cd7dc2534c28831bf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-12 19:15:44.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9bb2d236d75fb32aed24fb82f069c5fef2a42c04","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-12 19:51:17.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"580c03e576fd53b7a2e76772abf50e03124d7271","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-12 19:53:26.000000000","message":"Patch Set 15:\n\n(14 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":15},{"id":"3de2d37b872030a9c8aeae09bd501ea081b90a7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 21:05:42.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97d7d6ab43c24062bf257512cd7b041b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c06e541a8e84684abc87aa9604aa557 : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/6ca8593786244080a902cfab1f3cda9e : SUCCESS in 34m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66fe1e7917dd476f8ff10a4439846a6f : SUCCESS in 1h 08m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/603b6703902d4ac0adc5d4df8f5a1a42 : SUCCESS in 46m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed2cdf4af8c84cc18eea314345e3e62b : SUCCESS in 1h 04m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c6df62b789448b88895fa34a70f8581 : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/897909f2959e448a9d73d580e53b0808 : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c288fd8dbca4fc28e098896f5337a56 : SUCCESS in 18m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e433934e8d64c49888c5cf2858297a5 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/768b782edfb748e385a7188a8f7c0efe : FAILURE in 20m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b0d8b2c58eba40b1a00debc8b752056b : SUCCESS in 21m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0ce453da3a544089b60219ef9a73aa6a : SUCCESS in 22m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5d7f7f5180124dac9111dc0e719ca51b : SUCCESS in 18m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06e33b612d3944f3a2a17cae9e653cf7 : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f69ee9ab0f04524b4d387ed734ef5d5 : SUCCESS in 16m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/512ecaeaf0134086bb749075206390d6 : SUCCESS in 22m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0ad5631d7c39409aaf716f8e014a0daf : SUCCESS in 12m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8c00c8cd52f84ce7985df5739c5b442b : SUCCESS in 18m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/08e4a83f32fb412baf81231a58f58d60 : SUCCESS in 13m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6af40925574b44f0b519d1c4db3fb64b : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1a218a972b74e52bcc67eaa210ec204 : SUCCESS in 16m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/73996ad601304fa6b8407ceb48c83372 : SUCCESS in 7m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c76729ae576a41b0bf2fab5f1c5f11ef : SUCCESS in 52m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df3addc8ff614e9b9c379c9e8223febd : SUCCESS in 31m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d0388ecd5fe4e02be5e8f111fdd4d61 : SUCCESS in 53m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4976e466cb24456b80e866c1b26d3a9 : FAILURE in 21m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6f5d4e20bdd4176a0661874e8df5a84 : SUCCESS in 3m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/092730211a7b46858095bfb4f31e1761 : SUCCESS in 17m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0119a6afcbf54404beef09cbd60acd22 : SUCCESS in 45m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b44fab660f64971a076743df4ace7a3 : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":15},{"id":"88a9bccda9c73626bd4a7843d23c9050f2b26d0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 21:20:26.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5b113d1278f4dcb98db2d4f51d7a772\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/16e0d011272d473aa232b89b6836db97 : SUCCESS in 1h 21m 27s","accounts_in_message":[],"_revision_number":15},{"id":"4c8727ccc5e99d740c4c2cb1b1bece906be3488c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-12 22:08:22.000000000","message":"Patch Set 15: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"79b59e25f8a0fefb0287908d15ddf296e85573e2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-13 00:46:25.000000000","message":"Patch Set 15: Code-Review-1\n\n(46 comments)","accounts_in_message":[],"_revision_number":15},{"id":"30a11a09a13beb6114d4bce71d51c237fae537fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-13 19:41:13.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"8966c8138cdd7c7b97f7ba6456f018584885036e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-13 20:08:02.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"11c8c3da7063aac03db0234be28fdbd62efa4cb6","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-13 20:29:29.000000000","message":"Patch Set 17:\n\n(10 comments)","accounts_in_message":[],"_revision_number":17},{"id":"a8b3bee0663db26c023751bfd3d0b668e2de389d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-13 20:29:34.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":17},{"id":"d7839769346232596c47128f25eafc45aee3c383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 21:32:20.000000000","message":"Patch Set 17: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/684d5024c29c4387a5639f76a2053fc6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a5d040652149d59318bc2dec123648 : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/145b5093e0984c019287126be914089c : SUCCESS in 1h 04m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5eb22f1a7d4e465d94aa9dd5fcf96dd7 : SUCCESS in 1h 04m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6651040e88244f58ae10966c9eee42ba : SUCCESS in 58m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac7fc0d32967408f9d2a1fbb55cf992c : SUCCESS in 50m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/923a0c84906b402f881bcbd79e2c6ae9 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a70adeab15f9411394f47fd1c81e7675 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12ed8449ba3e4985b634d99f505e01c3 : SUCCESS in 17m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2ce6f3d1dcc94192999bc6ddb4aa5c5a : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5029a978845849e0b0ea0cd88553c899 : FAILURE in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/adf22f519b124808bc5a61d2a49858d3 : SUCCESS in 17m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6a5bf6285f77410e9c526555bcdfc7d2 : SUCCESS in 9m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c4513aefd3954de9b7befd2bd100eca5 : SUCCESS in 28m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/375591c7a1674743881724c29017bbc3 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d1050dfeb33e4d12996486a474ca0a9d : SUCCESS in 19m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8525b49cc8e8475ca32f055a5a0ee301 : SUCCESS in 16m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f62764e9b5094cd683ed001742673a55 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/367d51ae175047148f48ae00ed24b270 : SUCCESS in 8m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b5b90193a3ab49e99a86aa549b1ca862 : SUCCESS in 14m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/74423271d14945c7b3480e2d3fdc8ccc : SUCCESS in 4m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/484a5e405fd64ce9907d45778e10ba6d : SUCCESS in 16m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/92fa71d0e3224b39ad883e2c9fd8386c : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/44294d81a33f45bfbdebac44c511ab0f : FAILURE in 53m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57396067c3cd41198da7c09027031e6c : SUCCESS in 24m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bc08e892920d42beba04c3f0cd62ef3c : SUCCESS in 37m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3a41d56c403493fad1a5d4263582d79 : FAILURE in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce943f24506418a9e7796e18530da49 : FAILURE in 3m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e8b00cc9f4b47cc9b8b0560766ce25c : SUCCESS in 18m 17s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be9fabcd87174b6cb9d0aadc542e8937 : SUCCESS in 31m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1011f9c1723f46dab2997164175ecb18 : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":17},{"id":"9d0c27d57be74e9424338e0dc529aa1fdcf2fb9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 21:38:28.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d059d2c9e9064318b7cdcfde60b95176\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/52e4e33ddde24162a14054c7494c63ca : SUCCESS in 1h 28m 32s","accounts_in_message":[],"_revision_number":17},{"id":"0447d6032d35986207de9d87d6722dfe8c2289c4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-13 23:44:51.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"de74242a5335d52ff81ed89ee5c21bb63bffcb0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 01:00:23.000000000","message":"Patch Set 18: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89b0043125fd48f9a8ed8e5867612ab5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f38d472fdc5449038b648ba7fc3684d1 : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/fe13751db31e49439aaa7b1eddfd96e6 : SUCCESS in 42m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e14c0b5992a7463b8032141c1c56d3ea : SUCCESS in 33m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a25ce90c2e0948fdb93eaed0eab968d6 : SUCCESS in 39m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54fa8b49037b4eea8a7f6e179f56ec71 : SUCCESS in 51m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0ec97fe50db4036a979e7de8527acdc : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fddc51bdb65e488f9fbe50c76ae866b4 : SUCCESS in 11m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6929db2817f245069128422919249c4a : SUCCESS in 21m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d94f50cd4dd644f08047ce19e32a0fd3 : FAILURE in 3m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7cc7238e24b7430e8f9a395d702a13ad : FAILURE in 19m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8269ccabb0de49bfa2c9222109d107d9 : SUCCESS in 18m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ffbc2e0e5a8349f9bba7f0e75625fb43 : SUCCESS in 24m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dbc876e773fe45d8a5f3ac0a35f1380f : SUCCESS in 12m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/243659b7927d418cb2f3be5a89055c59 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b83779fd45d944d286d43228c2c829b8 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cadc45d6fdf74a489627e2b5c99ec38d : SUCCESS in 15m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c3f0661570574e87b275e672bf6065d9 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/872a014735744317971f4aa3bfcdc5da : SUCCESS in 8m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0afd9d06f0174f20b117b749ab449a19 : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d834f46236e440086e488fb7353da42 : SUCCESS in 5m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/018044b9fa954003a0edd3d31941a1a9 : SUCCESS in 14m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b00ed2912a244f7481001a034faf0cd7 : SUCCESS in 7m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba4fc37fecea4ebd8032835f3f4d1015 : SUCCESS in 34m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/353432ad12734377979fafb0484a1348 : SUCCESS in 38m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9c205024dd949e486909b670cafd4e7 : SUCCESS in 34m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03c43795737944598fa00c1893922430 : FAILURE in 19m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f17e9b6c0044f13bc5f720d0fbc7179 : FAILURE in 2m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad32eff79eb0453da0d388ebb99484e0 : SUCCESS in 20m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a4d407095b24996bdf48370e3ed3970 : SUCCESS in 56m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f420c361092e4a5cb747762b26e687e9 : SUCCESS in 1h 09m 09s","accounts_in_message":[],"_revision_number":18},{"id":"7b4178684ba7966e389b3413f4d73f8eb8d97193","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 01:00:49.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b133794de7d4436b5b1124f0777f411\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ef9bdbd4a9d43e6854e1a971d879f9f : SUCCESS in 1h 14m 16s","accounts_in_message":[],"_revision_number":18},{"id":"373658b64d90fdbf62d8613906612dd2ca9242ac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-16 16:59:40.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"fd96b0454abc7e51a6e5052e8c90b931be105325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 18:17:27.000000000","message":"Patch Set 19: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22f86f267455415a866c50c5aa776139\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76a15f92e61149bebae2f5f91050e54e : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/3af4b79307ed4fc38ea9f7cb3771d3c7 : SUCCESS in 1h 00m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59bd5aec57d14724b655de297a56a6e9 : SUCCESS in 43m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48274a89b5ec4f2fab3c43d430ad0548 : SUCCESS in 54m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dc54bc6517a484baa2acbfda60edfe2 : SUCCESS in 1h 10m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/852d84757680481697a329f3084bc8f5 : SUCCESS in 15m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cdb259ef912467596a5db4d1b4ec1ad : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7aff8357849a45138440e3e7b792a36f : SUCCESS in 23m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/79eeda26a7244940b79fbb5cbcae02ae : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28b8b0752a034124bb21b287a5472dde : FAILURE in 19m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a365c61eb1884376b80cc3573b435598 : FAILURE in 18m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2a504aeb9a9e41faa536d971629c94c3 : SUCCESS in 14m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/38ac141f7dec4bdca89885e8217cda59 : SUCCESS in 19m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/49a54f3792f14170b9e082c6688b6c11 : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05bb4786be784451be20a540488a4fe8 : SUCCESS in 12m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e485a57837cf43c39a36c25d22726a98 : SUCCESS in 18m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/11c01d57e168441bb67635f997c99f6a : SUCCESS in 13m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8f4d7c87a1e24685a494389e6a262769 : SUCCESS in 8m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9c166533de6f4b6996fa3f01d81c06ee : SUCCESS in 8m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7d0c2fcd54d4c48941bdb85214b4634 : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/19ded9626be641b5a267e634fd3f3b37 : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/26a700f654e64dacb69dab2e45c04fd9 : SUCCESS in 4m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/678bcfcf2ced44658554993d2dc87b49 : FAILURE in 57m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/abd496fccbc74b57a7592a7b169e697c : SUCCESS in 23m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d55872fce0c48b88dfbbd681347ec5a : SUCCESS in 36m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2359b32fbf9a450183010ea44ddb7178 : FAILURE in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af80079fbbb443798fe148702d37dd3 : FAILURE in 3m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3a441f9dbeae4453870eecf7b2f4baab : SUCCESS in 14m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97a4b1a1f75e47beb217686c8dba6b3f : SUCCESS in 33m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ee9b942a19d46ac807716126ef01445 : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":19},{"id":"0c08d6bf30aefc9e8256fecc900062bb5e24f9ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 20:39:36.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a24dac83d254517a805f4581a39e1f7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d695d029c6184ea6bbd3ca8b66e2ee5c : SUCCESS in 2h 00m 48s","accounts_in_message":[],"_revision_number":19},{"id":"b8772a76803de83a4d6155bcd948f6d6d7f7bf14","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-17 01:58:50.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"6ed31b49d81f72e6d77adf9c41f39d2bea74a438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 03:17:24.000000000","message":"Patch Set 20: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/551b87fead2649b1a78073b26ce8dd3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af852ea6551f4ff09e58818e4d5a79b5 : SUCCESS in 7m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/b7ec19e8c5ae4b7a9cdc1b28db4e109a : SUCCESS in 53m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/420f717eb3c349d2896a28b44c086a6e : SUCCESS in 40m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0f2da0b50854dfbac9c4e491d7810bb : SUCCESS in 38m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a5b64146f5049ddb6d960c0f66d65b6 : SUCCESS in 51m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e12694c45d1438fb333c6a2bdf5bd75 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6cae810b10941169d366c60f9b30070 : SUCCESS in 25m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e324e174f0a4a8c88ba0f97a8975148 : SUCCESS in 26m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/df83151e24334094b469aa8d279198c6 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c6cb028860f94cc7800313271867be21 : FAILURE in 4m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4c9eb61e7bc740d791234c415fb0362e : FAILURE in 9m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b9992f575ede4d0cbbe7f56e78269fec : SUCCESS in 21m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e499f3a26456420e9d0127960c5769fc : SUCCESS in 18m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6b9e2311b7f44e48b0ad2918a9bfa42d : SUCCESS in 9m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8143bda657db4aca9a201f4027808c98 : SUCCESS in 20m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cd4be347d74a420db90b7e15918f81d5 : SUCCESS in 21m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/646ea96fed0547a9a7af96057e9163d5 : SUCCESS in 8m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e4acfa5f98064dd8996cb8cdb2e3d7c5 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eaa74be166bf43eea1d928bb27c2ebf1 : SUCCESS in 17m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/68130cf26d234295b159d8bc900c2721 : SUCCESS in 5m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5bbfd7c58322400c95f743f991f764df : SUCCESS in 15m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f5a748fde5fa40b8bde97d1df8d9636b : SUCCESS in 5m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff89505501b647908b27fff9714fec6b : SUCCESS in 54m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f227682bdac84c8bb4de121a7928b04a : SUCCESS in 34m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e0f28504f9114775a866cc0f1e491932 : SUCCESS in 52m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b0da192c1c943d4bc764212a51888fc : FAILURE in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29e63c78f47d49a898826cdef1837cb4 : FAILURE in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8f5ca56ece20408caa520deb4a295fd1 : SUCCESS in 11m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0523387b36d4ba69b7e57139e83e722 : SUCCESS in 39m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0935c6f4141148499afd01ff833aa9ba : SUCCESS in 45m 36s","accounts_in_message":[],"_revision_number":20},{"id":"2ef32f5cc531b8696130ac00f32b022bc5ee277a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 04:40:55.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7033b6bc54f405f88b23f8664822ea7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e9f34c2782ac431b860c382c85f6e114 : SUCCESS in 2h 39m 59s","accounts_in_message":[],"_revision_number":20},{"id":"2c0c5cfafa7325a054ba2e771bcbd9db7384dd67","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-17 16:25:55.000000000","message":"Patch Set 20:\n\n(33 comments)","accounts_in_message":[],"_revision_number":20},{"id":"3730376d52ed3a24130759d891c55b4ebb829bb8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-17 19:26:53.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"b69623873a526c9227c8beb139e3df09dff38aee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-17 19:42:50.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"418a725f3661b959d8b595733265635edb35076c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-17 19:45:02.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"2d52cdf17f58515b3c5eb7854adb59508a2401e7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-17 20:09:15.000000000","message":"Patch Set 23: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":23},{"id":"f0d1c72b24874e9e254e83c8923d7615870e2225","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-17 20:10:28.000000000","message":"Patch Set 23:\n\n(4 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":23},{"id":"4e3681ac38ff0da2526cfd71f782f208a0d70bbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-17 20:11:33.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":24},{"id":"cd6494e6d606f5df2dcc043dfac196dd91320d51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-17 22:05:45.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"172acdcd6761e96a692bf6ed7da96d787c4a7f50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 23:17:07.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac230722ecb745e2af9a5c1405914467\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ca77087149456cafb703629d739811 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/08e5a3bee36f4e679b529f63c2b153b9 : SUCCESS in 58m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3bed43ab0f3247e2aaedb9cb979d1c46 : SUCCESS in 41m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd710c50dc544bf88bffa01b0e2d8eff : SUCCESS in 35m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/189f5b2351d241e6ae70f160fb7ee9e2 : SUCCESS in 1h 05m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d76a0d7a374421e9af712db6d1c525a : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbbbd250326445cb94c2c7863bc33ab9 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2706f3eea0e64644a039cfe7c1694c45 : SUCCESS in 15m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8faf4a445d6d47f5889cdcbb2cabfef2 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e596041956347a6b7912930c7442d5d : FAILURE in 6m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/434817a329be4997a401a33bb84037d6 : FAILURE in 9m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dcf942aafd234652877d1d6141668401 : SUCCESS in 21m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2015f88d8ba5411e84f2e6683c78687e : SUCCESS in 25m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b52c8ced46ec41309941068e949c20ef : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4947e72f3f664a40a806c20c1bedb79b : SUCCESS in 17m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/82f8a8afd81a41cba6672ccd7e302817 : SUCCESS in 20m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/73aa663d24654dabb2e5775c2d0dcf23 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f9ab1e0cc8af4780bc7d2e42f625387f : SUCCESS in 16m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9adcb138f94346abaa6177edd518cb5f : SUCCESS in 17m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0f0765ba0c4c4a9685988fd1199e2067 : SUCCESS in 7m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65559d24c7054bf4a33f568b904942bd : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/97ff896735344c16ba9cc4c8ac4eba82 : SUCCESS in 6m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a09483bc410548b088455f530659e5b9 : SUCCESS in 45m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b54cad6ff489478896b133761b448926 : SUCCESS in 26m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a488cbbd71d14830b77961b64177d11b : SUCCESS in 51m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1dc3ae75cd848c0961c85c1b25b5f81 : FAILURE in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b225e67239f433899f5ae0d89a493cc : SUCCESS in 3m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/200c3cb28fdc49e69e80a0014a999aac : SUCCESS in 19m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34473409f2a74e99b2c3c715700b6352 : SUCCESS in 57m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce221560d6e844a0ab6ef178580cf1e8 : SUCCESS in 38m 13s","accounts_in_message":[],"_revision_number":25},{"id":"0d7329a4ceaf70e8883ca9c8b722191b107eed13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-17 23:37:39.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"9e8c0f55e1122ce7c66532ef10b15205f5310657","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 00:48:48.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46bfd8c807624679a849a58611dcca83\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab69e978568405f94db9f14a64d8689 : SUCCESS in 6m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/a179b1888d884617a5dac76d6efedf24 : SUCCESS in 45m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2c82325ff95d40c191fb058382c3e570 : SUCCESS in 45m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2444b5bef84c4273b9a47b7230873aff : SUCCESS in 42m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10f22b984506488cbbde775bda27b6d0 : SUCCESS in 50m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/376733f385fc403fa25dc324668bb033 : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fee551fd60164b5983cd68af0a152212 : SUCCESS in 12m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cf4782b5b434c7898d5f133c1018786 : SUCCESS in 22m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/edeaa79b5e404606aae7701d3353a846 : SUCCESS in 4m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7b5b6ad3135d4dbaa340a6a07f147e80 : FAILURE in 6m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3f3b912b832e4fab9daa4d0910d6d2f3 : FAILURE in 9m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9687b2266ba046fe90f116cb605319ba : SUCCESS in 17m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/86eca9d839584f32a256cd2414462890 : SUCCESS in 18m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0061ab02cc784966b8c4bea13b6b7b74 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f2c82f1672924451ba4abf6dc53d014c : SUCCESS in 18m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/20fc4ef0bd434a568fef0d5d346945e9 : SUCCESS in 17m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3713dedea7a743969655a6c1a4796975 : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d7d60e9b8c104d9d9cc3c920c5548e94 : SUCCESS in 14m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f50def19414e4ccdb9707cf4ffc0322a : SUCCESS in 14m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e49370e3f78446d8a4afd48bcdd62b0f : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1d8b5c8a56249608977a02897f2334f : SUCCESS in 14m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0406bae48fb34b0f9c31939508e9cd84 : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/23493efb4c90497b80d1dc2b53e7e29c : SUCCESS in 52m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/79414b502bbd4478a7df29eed45fc6d7 : SUCCESS in 53m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0aaff67106cc43cc9165e61da66108b9 : SUCCESS in 1h 00m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d24b773b23744f2a3a6d950f80f5f77 : FAILURE in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a292551199c4a058b7fe82541cc1e09 : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b0ebcc384af5440b8fe5db444da5f092 : SUCCESS in 17m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81ea71fce7ac4d01947ef7b720fdbd39 : SUCCESS in 47m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62721c7c854a4a329e34828dfaa0bc23 : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":26},{"id":"106ae542aa7fca4d29d37d212b7f9bd6541954e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 01:30:55.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f2e0e9ae0a2464c9f8f4c31b2f5112f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/acfc5071a62c428187d794ac9e6f99ff : SUCCESS in 1h 43m 45s","accounts_in_message":[],"_revision_number":26},{"id":"8b9449a821bb277f3cd671c7b3608a16bbda44bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-18 16:39:00.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"43ef5511af7279dd651dd9d59c86c37db79c1ef2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-18 17:03:22.000000000","message":"Patch Set 27: Code-Review+2 Workflow+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":27},{"id":"f125762e30ae4bd565f9e30e72e5e697d8b7b166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 17:47:23.000000000","message":"Patch Set 27: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62be82f8d2ef4f208df58506476919fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf6f3c982034016b9cb4a485f775805 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/2ddaf9a46ec947c190fe4d25b0202c63 : SUCCESS in 1h 00m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74fb071c24524d4e864b7a773bff3dc2 : SUCCESS in 47m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f7504a6d751492aac14cd632cfd6480 : SUCCESS in 37m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9bd335d04174f5982032a1347194857 : SUCCESS in 40m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7b8c4bb98fc4b18b129f861a1b77e2e : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc9ed61cae56431d955d0bde724c7432 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0447f0c5df884b01b0a0c5a66ced041e : SUCCESS in 13m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0cc07fcae714a8e9c78f7c167d535f0 : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa383858296e4953a8747582b6e4c41f : FAILURE in 23m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7404ad7310e246c09c6904e8a7af0c11 : SUCCESS in 17m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2dc372681fd64421b06a93aaaba0332b : SUCCESS in 21m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/89f57d1a2eca4546855b5cb2620f7bd0 : SUCCESS in 18m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5c7755d96830447e81500eb406febacb : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8bab650c04794550877c5850aef206c6 : SUCCESS in 16m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da5c5f4cc0404012a4d82e99954fa415 : SUCCESS in 20m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/404566a9a52e46aabe8d04a817a444f9 : SUCCESS in 13m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d2a7081a1ee54cb1a0acac9bda892efa : SUCCESS in 18m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5b044a6f5eb74798935c62497df4038f : SUCCESS in 14m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/44e9ae84b2714ad0b97ff8ed6c6d33ca : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/92120855dede43d8b6b7b3837381f351 : SUCCESS in 13m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1fe2b3815f1444b498c8de7fd4f63f4d : SUCCESS in 6m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/344c7cdc1c02418a8f5713f3807c30cf : SUCCESS in 56m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b78f04da7e654a1fa3addbf8c629d5b6 : SUCCESS in 50m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/76c7e006d7d94d19b92ed1728db22437 : SUCCESS in 52m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/598902e6ad30482c8ba696c032f4e660 : FAILURE in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05aba5f4b7fa4d06bab84cb7aef1ad09 : FAILURE in 3m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd249ebf346b49919f038761139c0e9d : SUCCESS in 17m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff887a71e6a34e36baf12223db466722 : SUCCESS in 51m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b594102932de4720b2bf094858db1812 : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":27},{"id":"b36b3cea2711acebe3d31b2ca3d4e1a94b265c67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-18 17:51:19.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":28},{"id":"26728f09eafbb5ba626e079eee682cf220f1a367","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-18 17:51:35.000000000","message":"Patch Set 27:\n\n(9 comments)","accounts_in_message":[],"_revision_number":27},{"id":"5d789c4be89658438a874fd76209707660f488cf","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-12-18 17:54:57.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"3012bfe43b19afc10ac6e2603af5edcaea56ee8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-18 17:58:06.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"19f315968a766fdcec01309ec34142ff7ea3b8a1","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-18 17:58:13.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"4ac03d252c1d1468db5f56f68083ccac09a0959b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 19:09:45.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5439d76c5fd6485a845b843fa3adc3be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23278e3129ee4e779f91fe3f908860b4 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/e5a532391e254e08b3dfb282a3ea5175 : SUCCESS in 1h 05m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d8ff169843da4a2e9ae15ded6f9f04d9 : SUCCESS in 57m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92928aa347ec465cb7b61a2718c65ebd : SUCCESS in 37m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a96134dacd6a4a17ad7f7ab248f6fcf2 : FAILURE in 40m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9aa4108837654353b5d2987d17cd68cb : SUCCESS in 12m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3c38e0206af4c86a6a18851d823ff10 : SUCCESS in 12m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb2fe772d314436990eb9a1aa5bb50a2 : SUCCESS in 13m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19ad13e05b274b7d9209c00ab77c3b11 : SUCCESS in 4m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8f8065aa6ad64eeda98aa7915caca42a : FAILURE in 11m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14dd80acddd34229b65a2312049526ea : SUCCESS in 10m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb671a1539ae467fa5e05e15c490ea87 : SUCCESS in 16m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0833905b1efc41a1af393058e3cc1c5b : SUCCESS in 17m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c4d38450a934de581783c71b9ac49b3 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b1c7914ee8d4143bec8df6c25be9840 : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f9849eadf6d942eb917a71cabce53dff : SUCCESS in 16m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/15ae381437b44a1197e79f075837010d : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/31c7bb24fa6d4d8581e524b7e4cf3ff0 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eafba4737baf43e1b844470150223693 : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1482c7c0f60f4c79ae13b68e8ca5efe7 : SUCCESS in 5m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da677740f7a54acdba95226d02c1b40f : SUCCESS in 14m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e3d3a8a927d64ea8aa0d703bd7dfba75 : SUCCESS in 5m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e79a73c98c14c1bb09f7527cb2ca560 : SUCCESS in 48m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e3bcb9beecfe4e248aa8d1b7b0db9f7f : SUCCESS in 51m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/07315a5dac4f43d0a3df6cf2ef301127 : SUCCESS in 44m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90e49ce965dc48309ecd172a610fb8be : FAILURE in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69817176bac3430985598a61bd84cfe0 : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/47bef4a41fd2437193f27653a9d26934 : SUCCESS in 15m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d3eeb62e5734e94ab70781d7fb75c08 : SUCCESS in 55m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecfb34a3279b4f5f9db0437a9a8c0920 : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":29},{"id":"112b0f1840585e8a7e9646ef9bc5ae1bbefcf073","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-18 20:21:58.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"3b44b8aa97d1598e161c315f14dd96fe4114c5c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 21:39:26.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6ecb60cfd1246be843c21b130c1eb98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71674450f04a40eba69652ece6352ff1 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/11b8724b6204402794769a4062322633 : SUCCESS in 1h 03m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/83f5d9943a25414ba388eabf3c01780f : SUCCESS in 57m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3fdab337851a421ba68f49300dc3440b : SUCCESS in 52m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1cd2a780de14b58bd1bb167e5dcc726 : SUCCESS in 1h 05m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77fa8074558940e0b34a77c59c695b8e : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb6007e0b0c34f47a56b39c0374d724d : SUCCESS in 21m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a01bb5cf94a94e359b1efc33a55d33cc : SUCCESS in 22m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/751e2892d747460cbc93f46bbe5d9f55 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0fccf16fb8004ee4bf50f06118b35074 : FAILURE in 19m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6c31b0cc908941309f565cb448cf8813 : SUCCESS in 19m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4dbd7b72e19245ce8ae6df87c5bf1c4e : SUCCESS in 16m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/96a58e7c69a7405ba8c8d47841239521 : SUCCESS in 21m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b9d65065fc93468f8bc03b657f9cd7b4 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6405438b207f41ecb89d7e6481d576b8 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9af3beb624d04232bbdb702a70523017 : SUCCESS in 15m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7af5f45cf19d412d9839de2570db6c58 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fbfa2f6ccf794115be8d0684b78e08c3 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/54468fb2bf064a20a711f5df079bab82 : SUCCESS in 14m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5da3e898df7c4629aacf8b592f7dad14 : SUCCESS in 8m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/57a590c3ab2c4af1ad185b000cd945bf : SUCCESS in 14m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0e77c81d0c1e4f919ac1eecfc03ada16 : SUCCESS in 4m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8385ff719c444135a5032de48d9c7f8f : SUCCESS in 57m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/507b8a99a6b045858c76d3ca5e20e5ba : SUCCESS in 38m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ae7a2d883f2450c9826aec6b1994c80 : SUCCESS in 49m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8409d11d642a48a19868444e46020bb6 : FAILURE in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d2196f1ae94ed08c3710700ba5cbab : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cae0263fab95428a9c7b8fff70202193 : SUCCESS in 25m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b725de9a4634908bb348954c677af53 : SUCCESS in 37m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bf82389bf404d18b38af857088f7e47 : SUCCESS in 50m 25s","accounts_in_message":[],"_revision_number":30},{"id":"799193d8cce5dbedeee11a165bce9581e2abd796","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-18 21:49:44.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"70db2b4033c9a6d2c1ae002d25c0c9830b2718c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 22:50:31.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eaa0bc81c94f4b3f9f5239300350e702\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f471744facf843bbb47a51ff79718017 : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/0a1596ba238645b4bc34318784c95208 : SUCCESS in 34m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d90fbf8f2f474304beb7ea7334313d64 : SUCCESS in 42m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0066289290849a08e9790f9678f10ec : SUCCESS in 36m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82027e77473c42a395222d8861e61ef7 : SUCCESS in 46m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/800923f335b74ee788651b96fb3c9c2d : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2059043fc02498b9d504c24deb75855 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06a7d86255744943a0acc5c7f6d6a0c8 : SUCCESS in 19m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3481029547044696984ae8809608612d : FAILURE in 6m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06cb84b444924008904f41c33c0a1538 : FAILURE in 20m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/388bc168d9794a35912144229022fb50 : SUCCESS in 9m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4cb5810a3455464ba7068a148c1cca73 : SUCCESS in 20m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8cb00ee8aaeb4a48be2a476732c39f17 : SUCCESS in 16m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/37fcfcb1245647128db1fca89a4351b2 : SUCCESS in 13m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b3199e6a1144416900e0a4eb47a6662 : SUCCESS in 14m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4571c25b5b2d43dba7bcad6c55351008 : SUCCESS in 13m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/92122812c3594ae1a56bc6b4a1b5493b : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a9df8da7ba4c423b8786ad4672139240 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/208778d162ee4bb6a404d7e1dea3e581 : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b70455a8eb2d408e93e2abdfffe5012e : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bc0a139a7d63444ab3d04ce794e8aeab : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2259473e7942482e9217d267b55b5fe7 : SUCCESS in 6m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b274f94629f4b6ba3751818d27c0065 : FAILURE in 55m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00926b1f1ff74602acd19c6ea8022681 : SUCCESS in 49m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cb056863e79640d09d83136d2d17547f : SUCCESS in 49m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c05f01ad92ac419893e9c23db89e72f7 : FAILURE in 19m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ae533a7d3946a699942f34f072b13a : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18c2d0b2bb844fbfa902340d0a93dbcf : SUCCESS in 17m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fc08c6359564ff3b96ae0606cc88753 : SUCCESS in 44m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd199005a2724b58b858e972301e4b0b : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":31},{"id":"8aaf44e58f951791d693ee6f0b38316f2de8b97b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-18 23:45:01.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"a1e30aebfee0edf5726444279a2c26a0bd8dd88a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 01:03:33.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62c22261166e4131b1659a2595b39e20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a6b382c4dc453bbb21bff942e2d893 : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/342d22d79231434c941a63f2e2d2a578 : SUCCESS in 55m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2040aee3770c4b5fa77269b985368b7b : SUCCESS in 1h 07m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b1150e83eca94dd38e48f96224f8aea4 : SUCCESS in 1h 08m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c8ed750f2384cd7bc2e4b172fe6752a : SUCCESS in 1h 12m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6dab1cbe016f4e93998f1f1f25238730 : SUCCESS in 12m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45fe26b146724139bd193451d16e4c30 : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/816f2c178b3346b5b9b9a543187b594e : SUCCESS in 20m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c569c7bf387e4ba2adedadddfa525dcb : SUCCESS in 4m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/253840430b4a451c80a41b0e21a5dea0 : FAILURE in 16m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/183e2bfa5f144b7d8980c2f997a52130 : SUCCESS in 13m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/95339feab705459981743b1afc77086e : SUCCESS in 20m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fd5b47e41b2547e698f8e40e78ccdfea : SUCCESS in 13m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b3f241286c94409dacada1795e0c785f : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb389ec6c6b5416db17ef9e2ea36d1c3 : SUCCESS in 12m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/57092d3f26cb42d8a9eecc44017a85d4 : SUCCESS in 19m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/067f812ecd3b420199803f965d3330b6 : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/832ca6c636e84820bef6226d842b62d8 : SUCCESS in 13m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/48a7223d5fda478fa86ef65cd97922fc : SUCCESS in 9m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/25583b1bf38c4705b28ecffd04ba5dd1 : SUCCESS in 6m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a9797cb34ab46eea853fad40e7a42cb : SUCCESS in 14m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5928684f7c7647cbbd5a4d6d0694c5ec : SUCCESS in 8m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b14dd8d2eee409685409db1cc5870ba : SUCCESS in 57m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db06ddc6425f499eaaccee50d8049a06 : SUCCESS in 35m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c1ec36c50cc4b89b6d203b333b7e779 : SUCCESS in 25m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e74dc969d7748c489527f4e106bc09e : FAILURE in 19m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47aafa6f26e84281a0f257d3a276a790 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/367ae22b2969404eb75872f59f0558df : SUCCESS in 18m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e54f927fe81448da76bbe8969d7a616 : SUCCESS in 1h 02m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02ebe6027a8a468fab1bdb8a31451893 : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":32},{"id":"9f049815aa788e36156d6d70bf00382ab4d9f981","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 01:44:26.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b786f0eb350e43988ded60b9dd179dbf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/52c3b46ce3074ddfb5d4c9a9799827cb : SUCCESS in 1h 57m 30s","accounts_in_message":[],"_revision_number":32},{"id":"527dcf80154b8e86bfe76d474387ad4bab48cf02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-19 01:52:06.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"94165d00b56797d256c68120b4b211379832e62c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 03:04:29.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1ac0febc5444fe0997fefe981c9a5b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c09183cb77545b5aba9b1b024b05f63 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/47f49796aa7940f88660c05ef5fe07d5 : SUCCESS in 55m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/537b353b8f4d414dadd39712d8e3a364 : SUCCESS in 53m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/725f28933962481994ae26defea14894 : SUCCESS in 57m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/201a7b106652444dbb9b139acae094ff : SUCCESS in 1h 04m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6c958fa2d1b4bd687a025d3e2d1522f : SUCCESS in 12m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/635c8fe0b1e74d3fa955ef4bd6b965f4 : SUCCESS in 27m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/299a0dd654b74784b9460cfdaa01a5eb : SUCCESS in 16m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/adf2e15708ad4d268502f7761a162d1b : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8b3804f7dbfa4889a09fc5c285753b21 : SUCCESS in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/32d801bb9ef44cca9d9d40b235ef3608 : SUCCESS in 21m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/54baf10b59624de0a3ff205addf5e7dd : SUCCESS in 14m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fe9691b482f14204b19ea95713b9807a : SUCCESS in 18m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f96ae51f41c1437f847cf6aaa25b3354 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/670010bb9d644da8b2f05fc7a3fc9374 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d028b5e6e54245c498bc17174da2738e : SUCCESS in 18m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/79301975f4a344d9998a0d8554d3fd7a : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1a5d6fd1f662484bbb9e25123b153ad4 : SUCCESS in 15m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3669d23fb60f4d9496321df941f9e219 : SUCCESS in 17m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca7f6ad2c4de4810947a5b238fb4ef21 : SUCCESS in 7m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/994a68d3e69a44ee8a46b57f1f0d30a7 : SUCCESS in 15m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/da002d1623264c589055a71d63a6d3e7 : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/77a50ff2e13c45cc9501dfdf65b13bf1 : FAILURE in 1h 03m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f9d44cce3bad4ef99ee69f9124c29d54 : SUCCESS in 51m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6de5a4746e94d848a031ff68051d8f7 : SUCCESS in 51m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/370a5be87c2b4df2999d31217c393dcd : SUCCESS in 23m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6f03cc6bba43659b12398d789ca079 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55d08f1c88034a54b078f29f672920d6 : SUCCESS in 18m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2448a181b624e47b0ed810f46b6ebf0 : SUCCESS in 49m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/041bc8689f1d479e8003bbf59271de49 : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":33},{"id":"d917173da1b2456362b7443b0fbde4866853a0ca","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-19 03:16:48.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"c442e36d6a2dadda53049f9822262b2ed832db64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 04:17:02.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c22882c823ce4e249b4499a78657b72a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb050506d38f47cdbfd83cc6f925336d : SUCCESS in 2h 23m 08s","accounts_in_message":[],"_revision_number":33},{"id":"21479c421d7a4d24fdef37c03053a19f2eb5d56a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 04:29:59.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09fb2cb5636249eaa8e7dce5a460a85a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b2583b5cdaa40fd9b9f270e0cc0e2dd : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/8431201cca724b59a868f335a7b51e5d : SUCCESS in 1h 00m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b1aedad2e96a4ca7b2cc5d6b5cda92a7 : SUCCESS in 1h 00m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ef1abb7692c545bc988fe9b7683ed193 : SUCCESS in 55m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c9fa50bf52647d8a037054ca8980e50 : SUCCESS in 1h 07m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/290a37bb123842fea0dfd92895696310 : SUCCESS in 15m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40913d2fe91d4996b96d495e78e2d7e4 : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6ee5bdd6e574bdd8e3c86af1f19bdce : SUCCESS in 12m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/219f12aee435480bb473cde4772ac73e : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/65269e4f872f40e99132bb03d6e3c214 : SUCCESS in 11m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9666a0e757b9488dbf7166e44d63985b : SUCCESS in 10m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f56422f81aa54ac393b0e54df4c62ee9 : SUCCESS in 11m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8230ecf70fcf4ffeab16294dbbe45906 : SUCCESS in 13m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e2476650320b4d1085e998c05f14c5be : SUCCESS in 11m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d460c33380a34696ad690a38b8ea191a : SUCCESS in 20m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1350b4a40ce24cc6a087e8dbf6de253b : SUCCESS in 11m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e66e67e9a3c14075b76949165f055b1d : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e74fbd853827410eb1020a0e9f7cc3b9 : SUCCESS in 10m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/85b65b54adce405b82af3291d5efc048 : SUCCESS in 12m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/96c5ad35dd1340beb1f76fcd11f1aa49 : SUCCESS in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa680a4f062f49899585a53b761dab81 : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/29112cb3b5254143807980c86d7504eb : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/147c2b30516040a5adf3efaf8fb175b6 : SUCCESS in 56m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ed62519aa73453196bdd7efe1858252 : SUCCESS in 31m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef28fb6f757144d1b796ce85d83242bf : SUCCESS in 50m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68f9c866bf274bc69b3199621ecc0cac : SUCCESS in 21m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39532cbf86074826b09972134f03e957 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc2901fb60a649ceab3590343a06bb3c : SUCCESS in 24m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6de2b859763b4f8286081828fe544076 : SUCCESS in 1h 06m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c963b29b7e664334b61a8d9db6ac49bf : SUCCESS in 1h 03m 55s","accounts_in_message":[],"_revision_number":33},{"id":"cf99ebe849bf597ff2dc84eeb6930033a44e136d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-19 06:06:02.000000000","message":"Patch Set 33:\n\n(4 comments)","accounts_in_message":[],"_revision_number":33},{"id":"4df588cdeb64439da9de21388b40b72e6d30cc56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-19 10:02:53.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"515b5169bc80ea4e28487ee6ed19649b3f845c8b","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-19 10:03:19.000000000","message":"Patch Set 34:\n\n(2 comments)","accounts_in_message":[],"_revision_number":34},{"id":"7caf6a1e77625dcd8228b61e853725aa63cf4070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 11:15:02.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45a6185a467d45cfbcb66ecfaa61df87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54eb30e4c35441619ea35ef1c5a05cad : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/386f4a29201746a2bd188594ca22a939 : SUCCESS in 52m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0894f518bd2347aebd5b3c55affd8e83 : SUCCESS in 50m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a25df7c45f6e45e6a2bafb044961dd1b : SUCCESS in 47m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21753b9b5b5f4e539e789318e856e7dd : SUCCESS in 1h 05m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a23519c914dc43308c1fe4dd463fc4ed : SUCCESS in 23m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5885d7726114a43972518730a79e395 : SUCCESS in 21m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c498fbb6745d431087343e2da386f750 : SUCCESS in 21m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c1426b37297e4146a0c6d9644beff3b0 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7fb7f1a9ebe64df8ab1a878d21461338 : SUCCESS in 20m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b873b639541436e89de6450f0672575 : SUCCESS in 20m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f006200b6ff14806b5adf743795bd583 : SUCCESS in 14m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/990c85167af34765b1b48234112dae55 : SUCCESS in 28m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af2730c4178f4952834167b6bdcb87df : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/85d3862cf3274153a53c68b788ba85f9 : SUCCESS in 22m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2bca95a02add4aaea24f4e7f546481b8 : SUCCESS in 22m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/509555e01a614c888ff5b934ac3b9569 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/63ed92c6d15641a0927a8c47af7330c7 : SUCCESS in 11m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f00468442d7a4460956f9a9e35df98f7 : SUCCESS in 19m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0fc6094ec9b14e748f43873046812def : SUCCESS in 4m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e20b55a26ab344f499d098df70420a87 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/de506e53ec944e61b73f35cbf42ea813 : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4be31374ee7c42c19e3d62758497d60c : SUCCESS in 52m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6acf611b5e6548cca177e032995c1a5f : SUCCESS in 42m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6e02c86eabdc415c9997ac924a5d1a75 : SUCCESS in 48m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c964f9a2992423eaeac0f733a5ba043 : SUCCESS in 22m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d29e01c331a42208b72282850341768 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b2de1b0c7954b6fb5390591f39bfb7d : SUCCESS in 21m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad119866daa142e7a441c1b977dc92cc : SUCCESS in 51m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ec80d3a3a464994b0f3f3486d7f5746 : SUCCESS in 54m 18s","accounts_in_message":[],"_revision_number":34},{"id":"01a7dae846346eaeaa9c8f095f0646ea41e1ead9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 12:01:17.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4a86a799f74444a9b3c5ae7d7d7bf33\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b298947d335d419daf25a319205ab377 : SUCCESS in 1h 55m 32s","accounts_in_message":[],"_revision_number":34},{"id":"57478226d0455af03700110d9a41dead04cf3928","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-19 17:17:36.000000000","message":"Patch Set 34: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":34},{"id":"a197bb8e32a0f61c5c88468ed78d71b67833e49d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 17:18:12.000000000","message":"Patch Set 34: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":34},{"id":"d8a125f76ad4cbbec58aa62c9f0128f4026bb360","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 18:35:24.000000000","message":"Patch Set 34: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a1191a7bf064f32bf3930ca8f387511\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d862a2ccac9a40caa2cc3f5a5df2fd9e : SUCCESS in 5m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/18500e8d085748f592d14e75043b3de9 : SUCCESS in 49m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee99b3ce1f844d2f94d6c11e31cc3cfa : SUCCESS in 46m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e36faba08957471a98fb3db70cdc8785 : SUCCESS in 35m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22213a3482254591bfe27912af70ade4 : SUCCESS in 55m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/95ac7f03b4134d0c9e3dc1c735ae9109 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/813bd887ab60454399a43b20defafff3 : SUCCESS in 19m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0728a1efdd9a48e18db57bd5dc96ee49 : SUCCESS in 20m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e7f467019d0c4f53a02c363b430d1751 : SUCCESS in 28m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/32cdf13279a246bfa61d46d0ccf37542 : SUCCESS in 28m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de100672b1c54035824360e2be2397e8 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/986d455f500c4ad2970e79b08262b035 : SUCCESS in 23m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7da44fe9c64341bf820c4dfa4d355670 : SUCCESS in 19m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d81bcb29aed546fc8741e879947aaba9 : SUCCESS in 8m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a632a8ecdce648c284398f9c6227b7a2 : SUCCESS in 15m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3f143004f43f4caebe4fed72766fb0c3 : SUCCESS in 12m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05e67f42cdf945f5b0484895eced09c8 : SUCCESS in 4m 22s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/05a3c60e6fc144b096ae400c0dc72f7e : SUCCESS in 8m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/61f5359a10934d7ea0278d8f5ad7e257 : SUCCESS in 53m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f015cbdeca9482285bf6c93a75c18e5 : SUCCESS in 25m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed9dc4037da24b0cae255bf88b7e5a95 : SUCCESS in 25m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b7bca88254a4aeba84d7e2065ac1925 : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/599af0f9d93443c48b212d33d02e27e9 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd64874739f448ecb216c184d7dd61e2 : SUCCESS in 10m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77eb7fc2ba8d46748fd5869c5b88a9ca : SUCCESS in 39m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4624fc6716364dfebd21d92c90290893 : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":34},{"id":"7847c3c2313020c4afa64cfb7d6f8ebb9018d48a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 18:35:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":34},{"id":"90083e41d39c63c1f8d2841740dec14d358147ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 18:36:45.000000000","message":"Patch Set 34:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03845487d23043f7bad99c2dcf4c0a5a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a70759c676654707b24e7b69baa5cb43 : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/764c891a320043f58a02f99e47922b4c : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":34}],"current_revision_number":34,"current_revision":"af57922cd8b30cdb333d536f0506c673b591c069","revisions":{"43975460acd91354b1c4d5b5ab8453c466cdf795":{"kind":"REWORK","_number":1,"created":"2024-12-06 17:01:54.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/1"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"subject":"Fix Oldest Async Manager Run Forever","message":"Fix Oldest Async Manager Run Forever\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43975460acd91354b1c4d5b5ab8453c466cdf795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43975460acd91354b1c4d5b5ab8453c466cdf795"}]},"branch":"refs/heads/master"},"00670ff8d0c6218b019ec59e43dff36e44170740":{"kind":"REWORK","_number":2,"created":"2024-12-06 17:30:37.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/2"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:30:27.000000000","tz":-480},"subject":"Fix Oldest Async Manager Run Forever","message":"Fix Oldest Async Manager Run Forever\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00670ff8d0c6218b019ec59e43dff36e44170740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00670ff8d0c6218b019ec59e43dff36e44170740"}]},"branch":"refs/heads/master"},"dcb26662ba24abb0e9dc0fd4fbf7bb658c0757e3":{"kind":"REWORK","_number":3,"created":"2024-12-06 19:51:46.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/3"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:32:54.000000000","tz":-480},"subject":"Fix Oldest Async Manager Run Forever","message":"Fix Oldest Async Manager Run Forever\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcb26662ba24abb0e9dc0fd4fbf7bb658c0757e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcb26662ba24abb0e9dc0fd4fbf7bb658c0757e3"}]},"branch":"refs/heads/master"},"ef0b4218297cac5a4a3893a615f437664efafc69":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-12-06 20:26:36.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/4"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 20:25:50.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef0b4218297cac5a4a3893a615f437664efafc69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef0b4218297cac5a4a3893a615f437664efafc69"}]},"branch":"refs/heads/master"},"4aa13d0f6d8baa38f5ec4e1764d516760e32b9bb":{"kind":"REWORK","_number":5,"created":"2024-12-09 19:32:25.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/5"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-09 19:32:05.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4aa13d0f6d8baa38f5ec4e1764d516760e32b9bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4aa13d0f6d8baa38f5ec4e1764d516760e32b9bb"}]},"branch":"refs/heads/master"},"d6f069648cbaa42772c232292029106107765340":{"kind":"REWORK","_number":6,"created":"2024-12-10 01:55:46.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/6"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-10 01:54:01.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6f069648cbaa42772c232292029106107765340"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6f069648cbaa42772c232292029106107765340"}]},"branch":"refs/heads/master"},"28c3049071e429f16960dfa424cb1658d73c301a":{"kind":"REWORK","_number":7,"created":"2024-12-11 17:05:37.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/7"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 17:05:05.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28c3049071e429f16960dfa424cb1658d73c301a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28c3049071e429f16960dfa424cb1658d73c301a"}]},"branch":"refs/heads/master"},"fd78334139592d836e655489263a7bb6a71a12a1":{"kind":"REWORK","_number":8,"created":"2024-12-11 17:17:10.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/8"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 17:16:53.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd78334139592d836e655489263a7bb6a71a12a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd78334139592d836e655489263a7bb6a71a12a1"}]},"branch":"refs/heads/master"},"e1a9798b33eb4d7eba8e2bcb65d327b9e10c7f9f":{"kind":"REWORK","_number":9,"created":"2024-12-11 18:31:19.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/9"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 18:31:07.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1a9798b33eb4d7eba8e2bcb65d327b9e10c7f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1a9798b33eb4d7eba8e2bcb65d327b9e10c7f9f"}]},"branch":"refs/heads/master"},"bb2d5663606aeed9b0cde503a1835632920ce136":{"kind":"REWORK","_number":10,"created":"2024-12-11 18:42:33.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/10"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 18:42:23.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb2d5663606aeed9b0cde503a1835632920ce136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb2d5663606aeed9b0cde503a1835632920ce136"}]},"branch":"refs/heads/master"},"13078e24b2b70cea115520c7e8ac7c429d75dcf3":{"kind":"REWORK","_number":11,"created":"2024-12-11 18:57:20.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/11"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 18:45:19.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13078e24b2b70cea115520c7e8ac7c429d75dcf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13078e24b2b70cea115520c7e8ac7c429d75dcf3"}]},"branch":"refs/heads/master"},"2463c4703e5ee6184fb33f86c0015982df97e560":{"kind":"REWORK","_number":12,"created":"2024-12-11 18:59:39.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/12"}}},"commit":{"parents":[{"commit":"c2f25214002a85e8c6b892115bc1819973f33f28","subject":"Merge \"s3api: add more MPU cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2f25214002a85e8c6b892115bc1819973f33f28"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 18:59:31.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2463c4703e5ee6184fb33f86c0015982df97e560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2463c4703e5ee6184fb33f86c0015982df97e560"}]},"branch":"refs/heads/master"},"a3f80e21256bcca287b1e9f32d82d1335cc83956":{"kind":"REWORK","_number":13,"created":"2024-12-12 19:09:28.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/13"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-12 19:08:54.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3f80e21256bcca287b1e9f32d82d1335cc83956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3f80e21256bcca287b1e9f32d82d1335cc83956"}]},"branch":"refs/heads/master"},"17eb13813df6cc0b75a18be1801e53872bcc0b0d":{"kind":"REWORK","_number":14,"created":"2024-12-12 19:15:44.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/14"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-12 19:15:33.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17eb13813df6cc0b75a18be1801e53872bcc0b0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17eb13813df6cc0b75a18be1801e53872bcc0b0d"}]},"branch":"refs/heads/master"},"9f423d1fbd5503caed504715a09800963f9fc611":{"kind":"REWORK","_number":15,"created":"2024-12-12 19:51:17.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/15"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-12 19:50:56.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f423d1fbd5503caed504715a09800963f9fc611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f423d1fbd5503caed504715a09800963f9fc611"}]},"branch":"refs/heads/master"},"49ff6605dc872cadcce6e874fad08cd60439cc7b":{"kind":"REWORK","_number":16,"created":"2024-12-13 19:41:13.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/16"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-13 19:40:52.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ff6605dc872cadcce6e874fad08cd60439cc7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ff6605dc872cadcce6e874fad08cd60439cc7b"}]},"branch":"refs/heads/master"},"363be835cb3861eb18997619836c1af77d1429d0":{"kind":"REWORK","_number":17,"created":"2024-12-13 20:08:02.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/17"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-13 20:07:47.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/363be835cb3861eb18997619836c1af77d1429d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/363be835cb3861eb18997619836c1af77d1429d0"}]},"branch":"refs/heads/master"},"0489af9816b4b20670473deb417dece001c65ec6":{"kind":"REWORK","_number":18,"created":"2024-12-13 23:44:51.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/18"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-13 23:44:42.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0489af9816b4b20670473deb417dece001c65ec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0489af9816b4b20670473deb417dece001c65ec6"}]},"branch":"refs/heads/master"},"062c9139494003667dcfad5fde7761b067ff3212":{"kind":"REWORK","_number":19,"created":"2024-12-16 16:59:40.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/19"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-16 16:59:31.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/062c9139494003667dcfad5fde7761b067ff3212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/062c9139494003667dcfad5fde7761b067ff3212"}]},"branch":"refs/heads/master"},"216b2bd30ba71cbb05d883d2a92e0f5e788c75de":{"kind":"REWORK","_number":20,"created":"2024-12-17 01:58:50.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/20"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-17 01:58:35.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/216b2bd30ba71cbb05d883d2a92e0f5e788c75de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/216b2bd30ba71cbb05d883d2a92e0f5e788c75de"}]},"branch":"refs/heads/master"},"556245afc5252d57c8599b931d3b9330b183e761":{"kind":"REWORK","_number":21,"created":"2024-12-17 19:26:53.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/21"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-17 19:26:41.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/556245afc5252d57c8599b931d3b9330b183e761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/556245afc5252d57c8599b931d3b9330b183e761"}]},"branch":"refs/heads/master"},"b00d8a0d3ee681ed7da5fa6ca4f2eac02228525a":{"kind":"REWORK","_number":22,"created":"2024-12-17 19:42:50.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/22"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-17 19:42:43.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b00d8a0d3ee681ed7da5fa6ca4f2eac02228525a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b00d8a0d3ee681ed7da5fa6ca4f2eac02228525a"}]},"branch":"refs/heads/master"},"409d545e661d4ed44a7f95fae9f40f76defd2f4b":{"kind":"REWORK","_number":23,"created":"2024-12-17 19:45:02.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/23"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-17 19:44:54.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/409d545e661d4ed44a7f95fae9f40f76defd2f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/409d545e661d4ed44a7f95fae9f40f76defd2f4b"}]},"branch":"refs/heads/master"},"3efa78ee8f1b69c67e0392554d1b00283c0eedc8":{"kind":"REWORK","_number":24,"created":"2024-12-17 20:11:33.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/24"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-17 20:10:08.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3efa78ee8f1b69c67e0392554d1b00283c0eedc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3efa78ee8f1b69c67e0392554d1b00283c0eedc8"}]},"branch":"refs/heads/master"},"84a53784278b11441482ae1f0fab3bfd6dd6c1f2":{"kind":"REWORK","_number":25,"created":"2024-12-17 22:05:45.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/25"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-17 22:05:37.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84a53784278b11441482ae1f0fab3bfd6dd6c1f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84a53784278b11441482ae1f0fab3bfd6dd6c1f2"}]},"branch":"refs/heads/master"},"5cceff6de0b60b0e5468fdc26ce026f15980fc29":{"kind":"REWORK","_number":26,"created":"2024-12-17 23:37:39.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/26"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-17 23:37:27.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cceff6de0b60b0e5468fdc26ce026f15980fc29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cceff6de0b60b0e5468fdc26ce026f15980fc29"}]},"branch":"refs/heads/master"},"f2d7fd79d531471924b125977c9367e455deae00":{"kind":"REWORK","_number":27,"created":"2024-12-18 16:39:00.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/27"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-18 16:38:40.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2d7fd79d531471924b125977c9367e455deae00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2d7fd79d531471924b125977c9367e455deae00"}]},"branch":"refs/heads/master"},"d438fff5bb364f2bfe5ed36d80847a3163c2db89":{"kind":"REWORK","_number":28,"created":"2024-12-18 17:51:19.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/28"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-18 17:51:08.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d438fff5bb364f2bfe5ed36d80847a3163c2db89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d438fff5bb364f2bfe5ed36d80847a3163c2db89"}]},"branch":"refs/heads/master"},"738db70cb982230dda801fbdb19dc42d80ef66fa":{"kind":"REWORK","_number":29,"created":"2024-12-18 17:58:06.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/29"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-18 17:57:53.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/738db70cb982230dda801fbdb19dc42d80ef66fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/738db70cb982230dda801fbdb19dc42d80ef66fa"}]},"branch":"refs/heads/master"},"c5ecf38006337051a84f68f2ac2a4c502100cf23":{"kind":"REWORK","_number":30,"created":"2024-12-18 20:21:58.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/30"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-18 20:21:49.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5ecf38006337051a84f68f2ac2a4c502100cf23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5ecf38006337051a84f68f2ac2a4c502100cf23"}]},"branch":"refs/heads/master"},"62a03c0370c12135ae7217600219d07405cbbea0":{"kind":"REWORK","_number":31,"created":"2024-12-18 21:49:44.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/31"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-18 21:47:11.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62a03c0370c12135ae7217600219d07405cbbea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62a03c0370c12135ae7217600219d07405cbbea0"}]},"branch":"refs/heads/master"},"d6373cdf98bcaaa67f851230332893086ffd60c4":{"kind":"REWORK","_number":32,"created":"2024-12-18 23:45:01.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/32"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-18 23:44:48.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6373cdf98bcaaa67f851230332893086ffd60c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6373cdf98bcaaa67f851230332893086ffd60c4"}]},"branch":"refs/heads/master"},"3a6c587671fac06d0672dda23f7b81a5c2518c17":{"kind":"REWORK","_number":33,"created":"2024-12-19 01:52:06.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/33"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-19 01:51:54.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a6c587671fac06d0672dda23f7b81a5c2518c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a6c587671fac06d0672dda23f7b81a5c2518c17"}]},"branch":"refs/heads/master"},"af57922cd8b30cdb333d536f0506c673b591c069":{"kind":"REWORK","_number":34,"created":"2024-12-19 10:02:53.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/67/937267/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937267/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937267/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937267/34"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-06 17:01:34.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-19 10:02:41.000000000","tz":-480},"subject":"Aggregate per-disk recon stats","message":"Aggregate per-disk recon stats\n\nAddress an issue where `OldestAsyncManager` instances created before forking resulted in each child process maintaining its own isolated copy-on-write stats, leaving the parent process with an empty/unused instance. This caused the final `dump_recon` call at the end of `run_forever` to report no meaningful telemetry.\n\nThe fix aggregates per-disk recon stats collected by each child process. This is done by loading recon cache data from all devices, consolidating key metrics, and writing the aggregated stats back to the recon cache.\n\nChange-Id: I70a60ae280e4fccc04ff5e7df9e62b18d916421e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af57922cd8b30cdb333d536f0506c673b591c069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af57922cd8b30cdb333d536f0506c673b591c069"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
