)]}'
{"id":"openstack%2Fswift~937294","triplet_id":"openstack%2Fswift~feature%2Fmpu~I54e47c5471dc3c617858a82eb990522154a5aa90","project":"openstack/swift","branch":"feature/mpu","topic":"test-fun-version","attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-12-17 17:51:12.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I54e47c5471dc3c617858a82eb990522154a5aa90","subject":"mpu-test: mpu func tests w/t versioned container","status":"NEW","created":"2024-12-06 21:30:50.000000000","updated":"2024-12-17 22:32:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"078706c45b3ef71606fa4959c2da31972089bd45","_number":937294,"virtual_id_number":937294,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-17 17:51:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-12-17 17:14:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 22:50:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bafb9f0ffc0f96e3550b2b75df683e8fdd4a4a85","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-06 21:30:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47eec1adc687b4e69dec0e10853d09e055cc46b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-06 22:12:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"190abf26663b2aaf065054c85a0d1998cf159468","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-06 22:12:57.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"5c2a79867360a329f8c1c03098a9090449da0e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 22:50:25.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36e3cf7b9dd444518fd72f65fec4c4f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c84739b4168d4e828836bc94fa5e9a2c : SUCCESS in 8m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ce6fd1eee5f4cc995b4a744161cfe40 : SUCCESS in 22m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16bbb1afaa034a35a70a8a808e4c87fd : SUCCESS in 12m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36afb1742952465eae6070979eb2c021 : SUCCESS in 13m 19s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d56ae6fac1c4281ac672753c1d6ea4d : FAILURE in 10m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e2b944fc07c945b4affed194b924a6bd : FAILURE in 10m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ec52babd54c746f6b63267510feb58e4 : FAILURE in 11m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e0b1f600d29a4edea0a2c28493bd0998 : SUCCESS in 8m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/946bbed3971e4a59883695b4a44f94f7 : SUCCESS in 9m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6deb0c4d47b94bb79ec1cc9185e4a576 : SUCCESS in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8177f3962da34b86a6f39b057e1b15ea : FAILURE in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1fbb16cb6de94250881def4a37d37d39 : SUCCESS in 18m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e2f731be04c249998e27869e73ba9c74e248ec18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 00:55:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60026e204724426f97d9b1c009ed7452\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29aa0b9e242442bcb1c8829efaf29e8e : SUCCESS in 2h 04m 41s","accounts_in_message":[],"_revision_number":2},{"id":"b1b7140ff7814f659de2feee66c7b8f8a71a1ade","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-17 17:14:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f3a0a86bf1092be43f1324ff27adf4419a1c1b6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 17:51:12.000000000","message":"Patch Set 3: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/58a4a5c80e1b463e915121395d29b288\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd462aef4524a6d9fc159d324bec30e : SUCCESS in 6m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9192e14ad6a24e6c9a9c42ab1487f7b6 : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b02e3d2edec4ff7a752d943de8ce4bd : SUCCESS in 13m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e931206c5c6447c9b6fcfbdd69a086a : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/acd1f442f77f49859d65de44296820f6 : FAILURE in 15m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bd74c3b57a754b288a7ec0b8fea2a2e7 : FAILURE in 15m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c2ddc7976a24c9f995edeacdbfc3fb1 : FAILURE in 11m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bd57580d546d465c832cbcbe8de52871 : SUCCESS in 8m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9a9889dc62864095a180303ac15f0201 : SUCCESS in 18m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0e78d07ee0a44292a8fd0b3e2c950eae : SUCCESS in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f6809216cd4a3ba50626954f87c465 : FAILURE in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/567ad5e1a5b2464c941f740e3993364b : SUCCESS in 17m 36s","accounts_in_message":[],"_revision_number":3},{"id":"078706c45b3ef71606fa4959c2da31972089bd45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 22:32:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e17ccc612e04398bf7c2c53da5c43bd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b8c707ec4614bcbafc58148452b6a4e : SUCCESS in 1h 53m 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d453d9fe565981ea0eb5d9f146712cb610184885","revisions":{"54643ef37cb5d7e8bca9ff7720ae2ca118ad5579":{"kind":"REWORK","_number":1,"created":"2024-12-06 21:30:50.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/94/937294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/937294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/937294/1"}}},"commit":{"parents":[{"commit":"65dba8bd5b1a98fab0dcf436be1b62bac5b11168","subject":"Merge \"merge master to feature/mpu\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65dba8bd5b1a98fab0dcf436be1b62bac5b11168"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-05 15:45:40.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-06 21:30:03.000000000","tz":-480},"subject":"mpu-test: mpu func tests w/t versioned container","message":"mpu-test: mpu func tests w/t versioned container\n\nChange-Id: I54e47c5471dc3c617858a82eb990522154a5aa90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54643ef37cb5d7e8bca9ff7720ae2ca118ad5579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54643ef37cb5d7e8bca9ff7720ae2ca118ad5579"}]},"branch":"refs/heads/feature/mpu"},"19dfbef7a7581632f9b32dbdfabd47a82498908d":{"kind":"REWORK","_number":2,"created":"2024-12-06 22:12:21.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/94/937294/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/937294/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937294/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937294/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937294/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/937294/2"}}},"commit":{"parents":[{"commit":"65dba8bd5b1a98fab0dcf436be1b62bac5b11168","subject":"Merge \"merge master to feature/mpu\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65dba8bd5b1a98fab0dcf436be1b62bac5b11168"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-05 15:45:40.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-06 22:12:13.000000000","tz":-480},"subject":"mpu-test: mpu func tests w/t versioned container","message":"mpu-test: mpu func tests w/t versioned container\n\nChange-Id: I54e47c5471dc3c617858a82eb990522154a5aa90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19dfbef7a7581632f9b32dbdfabd47a82498908d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19dfbef7a7581632f9b32dbdfabd47a82498908d"}]},"branch":"refs/heads/feature/mpu"},"d453d9fe565981ea0eb5d9f146712cb610184885":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-17 17:14:15.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/94/937294/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/937294/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937294/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937294/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937294/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/937294/3"}}},"commit":{"parents":[{"commit":"f131432d917c10224e8f5da71a1f37fd1ec78ace","subject":"mpu: support upload-id-marker param for session listings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f131432d917c10224e8f5da71a1f37fd1ec78ace"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-05 15:45:40.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-12-17 17:14:15.000000000","tz":0},"subject":"mpu-test: mpu func tests w/t versioned container","message":"mpu-test: mpu func tests w/t versioned container\n\nChange-Id: I54e47c5471dc3c617858a82eb990522154a5aa90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d453d9fe565981ea0eb5d9f146712cb610184885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d453d9fe565981ea0eb5d9f146712cb610184885"}]},"branch":"refs/heads/feature/mpu","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
