)]}'
{"id":"openstack%2Fswift~937298","triplet_id":"openstack%2Fswift~master~Ib0a62256584a5c7d458ebe6740f0f477f61dd23a","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ib0a62256584a5c7d458ebe6740f0f477f61dd23a","subject":"CI: point antelope job at 2023.1-eom tag","status":"MERGED","created":"2024-12-06 23:06:19.000000000","updated":"2024-12-07 05:38:30.000000000","submitted":"2024-12-07 05:36:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937298","meta_rev_id":"0d3f7db1c2265f48a66b175958e356f84bdfd6f5","_number":937298,"virtual_id_number":937298,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-07 05:36:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-07 04:50:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-07 04:50:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-07 00:16:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ebbf0110a4d679393ddf249302eab302b7ad7c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-06 23:06:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc5360eccc305364a15f07ab77f54ee1826d173d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-06 23:50:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"113a2d024af14a8a06ec9868ebe03a890beaedd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 00:16:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2f027ec807d44a785f4f68990de058a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d157fd97f02c4a0da53474fb5c643343 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/b587c9b9770142b2b719a51f4a5af85f : SUCCESS in 1h 02m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a49d7fa348c9451fbe84ef3a940244d4 : SUCCESS in 34m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3663216874b408dba4df80f8783f03e : SUCCESS in 1h 02m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bf58e9ef76945328bee0b4fd092df8d : SUCCESS in 52m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc55ed32decd4231b2fd3f7a68fc5bb6 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a693429872a240d697bfb896f9c4739f : SUCCESS in 21m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b14554f6c51149c4bf9727fc7b6e86cf : SUCCESS in 16m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38610c3a945d40d0843ec5821c61d2a9 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/884da9322eea48c4a2c88604975d76cb : SUCCESS in 9m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/541e4f48bf0a463cbe032bf77da3005e : SUCCESS in 18m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6891b97458594851a94854e232d3216c : SUCCESS in 21m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1ad1d033d66647bea7b9da1b0ef6dee5 : SUCCESS in 24m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/82f2aeb9b3924fffb2cdf3b73eb4514b : SUCCESS in 9m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f8fd980db4924bda8f13e401a8677a9c : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ed1dc6a9e94f41ddba6473d24dc59c73 : SUCCESS in 18m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e0373096ad3a4facbe41ace95e7025af : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/033143294fee4b0a8ffb3e1b29ad0ae5 : SUCCESS in 13m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cda2025688654ccd8ecc02860754db38 : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8a7463729d141cb9c71a62b6d6a5c34 : SUCCESS in 14m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2daee93fbb1442138825c83d1598f7ae : SUCCESS in 13m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/252088293c544363ad6baf68d9d8d809 : SUCCESS in 7m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/285813b0a93c4f3dad911816c9b0a77a : SUCCESS in 41m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3dc99447f81d49fea463fa539fbe3bf5 : SUCCESS in 44m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/325b73320afe48cd8ab06f6cad83a7e6 : SUCCESS in 59m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/808f4ac38b944dae93f86587047133cd : SUCCESS in 17m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a514011b823a4c93997444a6dfd02b3c : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c82b780b72a48dd8c7d05f4c5f7b113 : SUCCESS in 18m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6def93117ad4ef49f109196e6368e91 : SUCCESS in 34m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95d3f4aa947245dcb34cabdbbabc31fe : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":1},{"id":"703814b3c6f30a3d221cc6fd6218f51f3860956c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 00:45:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc356eb871c44ada80ce75ed7d5a7e95\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/17d4651e152b4abfa4e5ff91e1bd9ff3 : SUCCESS in 1h 23m 05s","accounts_in_message":[],"_revision_number":1},{"id":"49d2c014dfd41f8600b60bc3b3fa771814a0814e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 01:09:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/097bf1ffc7fb423d8c5f002e7e28d0f7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6b3542e43d54da38171400283efa85f : SUCCESS in 1h 04m 11s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f9317f20ab648c2a661080f917d43bc : SUCCESS in 1h 09m 36s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d5e2863f6464bca83da8d533909e4c1 : SUCCESS in 1h 01m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/397a9714233a429fbefe38d426b735cd : SUCCESS in 17m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a99ecd2b4cbe47b0a104ddeb9e337211 : SUCCESS in 21m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/99853404fd514951a22fbb48c808f31e : SUCCESS in 21m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1bab0d6d71714814a6bfa1efc61feb0a : SUCCESS in 22m 28s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/97869ce034ce4db0b36e03edc046f328 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5315dd878574534978b2c0bbe580a7f : SUCCESS in 17m 25s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a10b06ff42124413bb6909cf7587004d : SUCCESS in 16m 48s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/d2a1b8b6a5534765bd4157f2ca952fd9 : SUCCESS in 19m 59s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/491c2b4b1faa4b72980fb73993a330d7 : SUCCESS in 17m 31s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/764c091906094efbb634c9ff0404970d : SUCCESS in 16m 32s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/78c19b59b535427e9f3d9f6c13c15653 : SUCCESS in 22m 45s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/1b5e9dfccd0543c6af604f0d6708a5dc : SUCCESS in 26m 22s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/55c6f49bfbe641d280e658c120364a9e : SUCCESS in 24m 41s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/eb6685f1d9c2440eb67599735fa73a19 : SUCCESS in 15m 55s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/79d2adb2ac554458968733beb7bcbda8 : SUCCESS in 25m 51s","accounts_in_message":[],"_revision_number":1},{"id":"51df31912ef1c75ff3778002f5f06cd9ac3fd0b8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-07 04:50:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fc19625d26ba3ac7da310a8e3d739a481efaf41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 04:50:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9baa3ac0e3b30c73853c7030c69cfb806c9f4ed7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 05:36:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/530217a66be04b5399be0cc133a5b9cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc6797ff0ff84c1f9713facf28fc82d4 : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/c3cd649d13cb4a668fd7690c1e5781ad : SUCCESS in 40m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a3a9b0812de443bd8997c791f7fcb7ee : SUCCESS in 41m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/192dfc49ba264b72bceec3a4e4d24098 : SUCCESS in 37m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9834459587c54f18a177fe0c87871e59 : SUCCESS in 44m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/90187412b1b04814abfbda6b2529d37e : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c0ffc6eaabe5449790d88b300906d1ae : SUCCESS in 14m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d1ffcec7247746b4aa1791a42eeccf20 : SUCCESS in 12m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8408c0fa2f8a4cd7a1e542c25da930d6 : SUCCESS in 8m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/32874e1424b84deebff22a415c7d82c9 : SUCCESS in 28m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ad069731e7a4af0a20d3a28a317b094 : SUCCESS in 12m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/82778b3a938d4c73983c4f37c9ed5e95 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d653283e9fb496ea222b55025bad975 : SUCCESS in 9m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/06232445299f4ea380932841b52a0474 : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/62ecc22e93754e15b9cce628bcdbd9a1 : SUCCESS in 8m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/26227f3129054a22938a4c0ee0f38866 : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4f1c94dd27284bbba4e969b1d43c60fc : SUCCESS in 4m 07s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c22612a06da04d8eb09c018ce798b9f0 : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/94ed8af140f240539d6fa1cca73f8203 : SUCCESS in 34m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c294bdcb6ee405ab91c1f24559e9f87 : SUCCESS in 45m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c651f516cfaf4ea8a372c222101169d7 : SUCCESS in 31m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e59124f6aa3414696c8693372e6270d : SUCCESS in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43a82ec65a1a4dc1b18dbebfdc7f309a : SUCCESS in 3m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52bb894c656e46bca07d19e5eb5d60a2 : SUCCESS in 16m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/960703114e5146cabbc9a1168643f871 : SUCCESS in 36m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e17ee36123ec42a8838e4938f94a6aab : SUCCESS in 38m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a073c5f3ffb22e69a3aad32e1c0b6f39265dabc6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 05:36:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0d3f7db1c2265f48a66b175958e356f84bdfd6f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 05:38:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/407911a32fd24c489b837b2025e79a2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b847bc7abcbd4a36a88a026997507407 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/093b9bd6535f4fa0b530990eaa88943f : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5bd48a6a4b26c399f58de047cf8b7e4306390e6","revisions":{"e5bd48a6a4b26c399f58de047cf8b7e4306390e6":{"kind":"REWORK","_number":1,"created":"2024-12-06 23:06:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/937298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/937298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/937298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/937298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/937298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/937298/1"}}},"commit":{"parents":[{"commit":"1a6958a2ff5fb5886d7749d4fd4a4c11642637c8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a6958a2ff5fb5886d7749d4fd4a4c11642637c8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-06 23:05:54.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-06 23:05:54.000000000","tz":-480},"subject":"CI: point antelope job at 2023.1-eom tag","message":"CI: point antelope job at 2023.1-eom tag\n\nThe stable branch is gone now.\n\nChange-Id: Ib0a62256584a5c7d458ebe6740f0f477f61dd23a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5bd48a6a4b26c399f58de047cf8b7e4306390e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5bd48a6a4b26c399f58de047cf8b7e4306390e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
