)]}'
{"id":"openstack%2Fswift~937378","triplet_id":"openstack%2Fswift~feature%2Fmpu~I76f1a0ea7ab4d6c51db0fcba1d69e15c2a7abda1","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-container-detects-orphans","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-11 12:26:46.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I76f1a0ea7ab4d6c51db0fcba1d69e15c2a7abda1","subject":"mpu-auditor: add statsd metrics","status":"MERGED","created":"2024-12-09 15:44:02.000000000","updated":"2024-12-11 17:21:31.000000000","submitted":"2024-12-11 17:20:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937378-p-mpu-container-detects-orphans","meta_rev_id":"d1ec46e8aa021cb6a1b94e95a79f0efbe20264dc","_number":937378,"virtual_id_number":937378,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-11 17:20:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-12-11 12:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-12-11 12:26:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-09 17:09:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3eda6f195e9dc152cfa37f62b245a19e71fa01dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-09 15:44:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d924a5b0f00cafe4f7c693c7701922eec2ff388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 17:09:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c10f6dfce3a4f268945830c7c2f077b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec16a7be4cb8499fbde46c86587be225 : SUCCESS in 8m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc59f299ff4a4de4973b16aee6a25625 : SUCCESS in 11m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5907a1bfef434bc795aa71d2df6e5e27 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa364715b4bd48b6a4b7a6e77c1b295d : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0f32564dc7b4ef88d886c5fbd03e59a : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7f67363f4b634afab96a10305c95996d : SUCCESS in 19m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d7132868ab949778640d5312f76e945 : SUCCESS in 17m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/21d7f0a0f1f84cc79e2c18e3c239a290 : SUCCESS in 10m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/72010998f8ac48e996aee6ddbbc1ea51 : SUCCESS in 26m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4e2510411f3542e4af6bb9adf3388c4b : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/02f71628365f4a8b99bca1281c1dfd43 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6552a78d55aa49adb4424cc98e8bb589 : SUCCESS in 11m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/97c7a9dcdaea4ffba2f2d94515298edc : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2c1e99e6e9374b5a94fb64071a09f382 : SUCCESS in 18m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3d6155c82e6d42388d190dbb48436116 : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/abfd07ba6e544888acefeae90e015fba : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/253f8545720445c39934620ed261f699 : SUCCESS in 13m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/311c30c7b7e44186b376b67b60806281 : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/67c6e5f73801497792387600a4dd90f7 : SUCCESS in 53m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45f23ea9b03d4661bfd2849901629f9b : SUCCESS in 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795479e941e24423b42b39c5736f0957 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b3730a50a8b04657b65a93b358a5647b : SUCCESS in 22m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0c0ecf32c93c112fad75abcfc0f6c390c10a8588","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 17:29:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaf6ffa38d904686acb2da337e4d49fe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c21a093726144c6afce3410b33a5e06 : SUCCESS in 1h 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b532fcfe343f77f6a1a0b85026e0b15da3d8e92a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-11 10:47:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"755fd17c9915fc4aaadec09a59f1ad21acc2a230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 11:44:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee8c03535ef142a3b2e11fa3f3419bbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7104551b53043e483771071b1341f5e : POST_FAILURE in 7m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b66e6a1f67fe4768930af116abebd21f : SUCCESS in 25m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f17c946b96f54b17b2d9fc845d28c162 : SUCCESS in 26m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81603c52354e4bdea00e93b154417a7d : SUCCESS in 12m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae6902d5125f4bdaaeeaf7fc36ef51be : SUCCESS in 4m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fec052fb374f4095aca3b360057039de : SUCCESS in 9m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fbb26ce6c1ea4ebdaaab78a14c7e14e1 : SUCCESS in 8m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a1a47f35ac9f44bc833c939ed44eaf80 : SUCCESS in 22m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bbc8bcae7f734f54892e20233382aae4 : SUCCESS in 26m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/85b920699f324e81a8590e872c56119a : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/76c3a3729f134b698d35e6c140a53a8c : SUCCESS in 23m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f2f91cd722904ddaa245ee22be46b55b : SUCCESS in 17m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0e35251233784312b158cd270b3f7a3b : SUCCESS in 13m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/61c9bcd03d984aefb0381ed66356c688 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76a513615c484dcfbb59beefdc41183b : SUCCESS in 15m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/309962388d6944b5acea1c73056f26c5 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c660142509fd44e99c834a602e8f4436 : SUCCESS in 15m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/588f3e5a51e2453cabfd91a9178c75a3 : SUCCESS in 6m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ad0c63170ca459295a841ff237c3b64 : SUCCESS in 55m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5acdc1c539d4b6eb082d590793a6c37 : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9b8515f6bfb4320bb2bcb16c3db3248 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/85800a152f3b481cb9e1f2405bd8ec6b : SUCCESS in 17m 17s","accounts_in_message":[],"_revision_number":2},{"id":"25e9871cd9d43eb775a3203e645dc8fd4be9c167","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 12:17:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bab7153c127b49198b1b0f671dec1f64\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ef17a68213844d8ae93c806e900fbbe : SUCCESS in 1h 27m 35s","accounts_in_message":[],"_revision_number":2},{"id":"3bd72e95521caeeb169b4c7b2bd086593e01d833","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-11 12:26:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3555a9756243923c7f321466b665034a8b361e36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 13:26:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c2b76db066240c9a99534f4a2767f75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7114d60c5f57445c8f47310b7dfb3e41 : SUCCESS in 8m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4c2756daa3b4287a4d3dcde7283aec2 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7cb34f49db0495487a8e909f9222d75 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df8787a98dff47a1bce643672f36aa57 : SUCCESS in 19m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/916fe6404f944b4eb2ad372c6247688b : FAILURE in 3m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8bb510e798c44a3ba29c93216db928f5 : SUCCESS in 19m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2be1901ad6084ef68f2c9ac37e035f28 : SUCCESS in 20m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0466bf8362fe4cee903ceafa908e3016 : SUCCESS in 21m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a1654f9ce534418d8b4b381f28d3daca : SUCCESS in 27m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/189f257a0a6b4117a16886b798aa0a0d : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dda33276b4a443a09734c6a84664a405 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a126a08d24d84c2e8fff90e325962fd6 : SUCCESS in 21m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/72ac289134b146018b4783e4e5fb8a45 : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/48cc25189518481291823315e52aaaf4 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9b5873456fb4499fb33724bb0cca4412 : SUCCESS in 9m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4142c267c3b54d7eb2a8a5cfa28335b1 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/08775a161b5848249db1f596d344922f : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/67e67079d5194f659b32cad4f8747ae7 : SUCCESS in 7m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e80e788dc6d44f1b932e897c37881d0 : SUCCESS in 54m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e42fca969834e3b889384dbcc1a1b61 : SUCCESS in 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c303d92cc9449a5ae3e499f49803414 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b00c76d206a74f9fa0b75c342d3f553f : SUCCESS in 19m 14s","accounts_in_message":[],"_revision_number":2},{"id":"8695704cb1eca19da03b97cdfbd5430a53218900","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 13:26:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0c38f6f236f843c4db48162e69ed7889f6c8bb98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 17:20:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b96e6cae88f409dacdba1eb130f9511\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ebe5dd1a364eaba6caa4ce4d55fce8 : SUCCESS in 8m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3bb7200169d84820936b9aeed8361ef1 : SUCCESS in 8m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/74c3491f71df4e43a2dd5b50a43318b0 : SUCCESS in 19m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/18f24467117340a5a29fa7a0b7b31706 : SUCCESS in 20m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8785ef91ee854751afe9c02139276c44 : SUCCESS in 18m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c6f3dbaea1ba42d68b21ec9e8fea51d3 : SUCCESS in 35m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41b06ca397634d5bbda696cc41bfec5a : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e02ea6b97eee4702be554218ebee9065 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7aa1bde318824d7f90af892aa59b9d18 : SUCCESS in 16m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/250e1c8390d24d22b6b4e529495deb40 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a95ff6c523d54b3a900a1742617f6635 : SUCCESS in 13m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1f30fed66c004c9191ab6932d97e22ad : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dc68b16c3f8444608d4fcf3f1d049965 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b305ccc171364388b6d3bfac1ee91267 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/78b4bc730f7b4592a7acc1b85ef6832e : SUCCESS in 43m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8db6d42bc409488bab144d7078e9fc66 : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aacfc4af15040cfb5b329c611ac3f53 : SUCCESS in 7m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26229f78e3c342958b4bf4cfa849992f : SUCCESS in 24m 36s","accounts_in_message":[],"_revision_number":2},{"id":"979891d045ba50d8cbccb7492c97d3058b394bef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 17:20:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d1ec46e8aa021cb6a1b94e95a79f0efbe20264dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 17:21:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90f404ed08bd43baa64aa234b7b64561\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/191d016bc4864e20bd7bf96b768554a5 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b8fe06c338c249a5ad6275475c4b166d : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"545eef7289a30ba295fa044e8cd1e2554bf03a0a","revisions":{"de6bb781b1c0a136885089659b0138cc910b9117":{"kind":"REWORK","_number":1,"created":"2024-12-09 15:44:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/937378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/937378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/937378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/937378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/937378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/937378/1"}}},"commit":{"parents":[{"commit":"65dba8bd5b1a98fab0dcf436be1b62bac5b11168","subject":"Merge \"merge master to feature/mpu\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65dba8bd5b1a98fab0dcf436be1b62bac5b11168"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-09 15:21:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-09 15:43:48.000000000","tz":0},"subject":"mpu-auditor: add statsd metrics","message":"mpu-auditor: add statsd metrics\n\nChange-Id: I76f1a0ea7ab4d6c51db0fcba1d69e15c2a7abda1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de6bb781b1c0a136885089659b0138cc910b9117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de6bb781b1c0a136885089659b0138cc910b9117"}]},"branch":"refs/heads/feature/mpu"},"545eef7289a30ba295fa044e8cd1e2554bf03a0a":{"kind":"REWORK","_number":2,"created":"2024-12-11 10:47:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/937378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/937378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/937378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/937378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/937378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/937378/2"}}},"commit":{"parents":[{"commit":"65dba8bd5b1a98fab0dcf436be1b62bac5b11168","subject":"Merge \"merge master to feature/mpu\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65dba8bd5b1a98fab0dcf436be1b62bac5b11168"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-09 15:21:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-11 10:38:45.000000000","tz":0},"subject":"mpu-auditor: add statsd metrics","message":"mpu-auditor: add statsd metrics\n\nChange-Id: I76f1a0ea7ab4d6c51db0fcba1d69e15c2a7abda1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/545eef7289a30ba295fa044e8cd1e2554bf03a0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/545eef7289a30ba295fa044e8cd1e2554bf03a0a"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
