)]}'
{"id":"openstack%2Fswift~937389","triplet_id":"openstack%2Fswift~master~I9473fad7c46ac03bbc71328c17e988af9d21386c","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I9473fad7c46ac03bbc71328c17e988af9d21386c","subject":"Up-rev hacking","status":"MERGED","created":"2024-12-09 17:42:08.000000000","updated":"2024-12-09 23:06:45.000000000","submitted":"2024-12-09 23:05:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937389","meta_rev_id":"8d9b6ad54590d5ce0ba2fd49928dab30af3c262f","_number":937389,"virtual_id_number":937389,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-09 23:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-09 21:21:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-09 21:21:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-09 18:56:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfc8002f59e4e9a56068aa103b2a0f8b830d8fff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-09 17:42:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddb295a571f3d32766ab990035b336d1e0755188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 18:56:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a735102723db4a2199c07e0471d485cb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d242c1e852564241841107572f630c3b : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd2ead2469542658bdd97be6059db74 : SUCCESS in 10m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3afd9914f5084b8f95f7d74368d776be : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34c252747e9c402e83026ab999c6a4bd : SUCCESS in 7m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/225f28045322451da108d1369a5444b9 : SUCCESS in 59m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9576e8b69554b89b928bac605554167 : SUCCESS in 37m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e7edd61e30e43b7a224b8f987bd2048 : SUCCESS in 47m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74ae77d605664173bbf2843cef8e0540 : SUCCESS in 51m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f89f42bbb0c4da6b057f9a332f6e596 : SUCCESS in 25m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1494d2d515e4aff838f826586c836fa : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/867a12d0eeb84e5aa6bb1d06723211c0 : SUCCESS in 19m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a8ce18c0c7124224b9b72d43decb501b : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5683f5940ef04d2690e8aceab0cb60a8 : SUCCESS in 18m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1553854baeb440979dfb2a60e34ed5cc : SUCCESS in 14m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4cb087b4205249fe8caa92eb3d2422bd : SUCCESS in 19m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d68f5e441fc741519dd36f1e4feeda0e : SUCCESS in 20m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1dd226a666af4dc793bdc7be4172eee8 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/956e36b56f9f46c787a55a24cc0d7b0b : SUCCESS in 15m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c834233ca6134f82b38d9bf168c75728 : SUCCESS in 17m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e139632753ef478eb0e2c0219348c2a4 : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/94ab7ca3d35b494dacbe401cfad53eef : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b4be2c665a844136bafb2e5c3628a349 : SUCCESS in 14m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/14581035ea9d41cfbc3a7c1db5bcfdb9 : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e3ec4f64034241658e4a7564dfad5cbd : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6dd44b1c95f14c1ab36476a36398928f : SUCCESS in 6m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/02f7018f71a84ea3ac62c1f84bccb3cb : SUCCESS in 55m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/04e2837865a246629dedcd43abec1817 : SUCCESS in 55m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bc89046fe7024c6abacde880b6e817e2 : SUCCESS in 1h 05m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04ae4736ccc9433c90ebbcd14f4b9514 : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba65c509c64746fabbb9052919474d6a : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d2701e1f10e42f990453afba4c44919 : SUCCESS in 18m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/395b4a65ad9049b0bf9f28aac9516e2d : SUCCESS in 42m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ed87c27141143a5b8f99870e4aad85f : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a5fab088732f3e20edf756a1e7e955c7b4b3ed49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 19:49:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ca9f20be0514005afe1aadc78c27130\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4492eba2bd7e459cb3b043fcca96917f : SUCCESS in 1h 38m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c7d2c16b445b3575d0bef41f17f51c5c5d8a7d77","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-09 21:21:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca00e20543c00fd60cb4715a82e352a2aaf08fc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 21:22:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a61d9e392f2d4bf5829ed3aa4e4b3331868b0d7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 23:05:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b57387e14ea4f62be6d5d3b6cd507c3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c6d6f21078f14e1ca518aeb80adf4bcf : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb2fba370d54f769f12849dc8c54e0d : SUCCESS in 10m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1d9a7439b7a941f8a8ea77e798b63164 : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49246a667e9e4e91af5030f32e332da7 : SUCCESS in 6m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/212ee1a144324b1c8dadbe78d062ddeb : SUCCESS in 57m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db413e1ded124afda298fd97c0e3ea69 : SUCCESS in 55m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a972e6e4495b45faa63d9143d0a170b3 : SUCCESS in 47m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02e37c0b7b6040d8b1e9ebda4a729ea1 : SUCCESS in 56m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/03510486d5e14a10b0910bfd5c31334a : SUCCESS in 4m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d71cb2794ff146e0adc0d8b5999dab13 : SUCCESS in 22m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd78b913308f4a3bbbfb64cc5bef3dde : SUCCESS in 9m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/60461bda5df846ebb77e778b80e877d1 : SUCCESS in 9m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c7f79416bb81402f842e8f52ef7136e8 : SUCCESS in 13m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/42a50e349af747409744074c92e6d81c : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e2b761507bb74cdeb60eab9dee6704d1 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6aae2fac5de3420a97b990ee216c9271 : SUCCESS in 17m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/329c56c7bfb04f7e8d90622e6b282d81 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d66123ff0beb4e88a5330105c1cc76e1 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/92943c91489b45118af1ba75512b304e : SUCCESS in 14m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d6e5d8c459f4284922eb3b9ce1808d8 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5f7a5c95618f49108f9abd7c066bd4ba : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f1f3c521a1c4e5288a4c8ab61c1294a : SUCCESS in 54m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/38e0fd9537934d79ac47bfc7791dc1ab : SUCCESS in 43m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14198410bfd4438e924bf141c8dc2dae : SUCCESS in 51m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee6f1c8ce557472f86920aa2fc13e36c : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84aa234720c74b2b8e9fe54b19e03c7f : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e0fe4f57ead5459fbd05727ea5dea4b3 : SUCCESS in 19m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2bce6cf3acb4211a8db8d5addf2686a : SUCCESS in 47m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3aa1305f43a94019a419cc9a6d409284 : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3d113aad95904fe56cd58546a20f5e59cd80c1f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 23:05:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8d9b6ad54590d5ce0ba2fd49928dab30af3c262f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 23:06:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bb39c10486347a188a4fbfcee0e5bb0\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/8bd3ec26221a4c20876da5c868cee0a3 : SUCCESS in 56s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f13086224a0845009d46eef7227300fb : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8b1525fdca042da9b96ec95eb8fdfad : SUCCESS in 36s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/19a2ad575db84e78afec4e291173b171 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"992d70198ccf770cf84380f19a35431ffb99f31d","revisions":{"992d70198ccf770cf84380f19a35431ffb99f31d":{"kind":"REWORK","_number":1,"created":"2024-12-09 17:42:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/89/937389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/937389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/937389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/937389/1"}}},"commit":{"parents":[{"commit":"1d69f04d5800567ac2006d3628c17fc7a247fac2","subject":"Merge \"probe tests: Run relinker via subprocess\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d69f04d5800567ac2006d3628c17fc7a247fac2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-09 17:39:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-09 17:41:56.000000000","tz":-480},"subject":"Up-rev hacking","message":"Up-rev hacking\n\nChange-Id: I9473fad7c46ac03bbc71328c17e988af9d21386c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/992d70198ccf770cf84380f19a35431ffb99f31d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/992d70198ccf770cf84380f19a35431ffb99f31d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
